)]}'
{"id":"openstack%2Fpuppet-designate~825691","triplet_id":"openstack%2Fpuppet-designate~master~I9875a2cec185c9ed10f13fbe883c85cdc880358e","project":"openstack/puppet-designate","branch":"master","topic":"oslo-messaging-ut","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-01-24 23:02:29.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"I9875a2cec185c9ed10f13fbe883c85cdc880358e","subject":"Avoid testing details of oslo::messaging(::*)","status":"MERGED","created":"2022-01-20 23:44:48.000000000","updated":"2022-01-25 00:48:50.000000000","submitted":"2022-01-25 00:48:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"825691-oslo-messaging-ut","meta_rev_id":"5846f865bffb9cb742055ea8a8d83de8c7a1b253","_number":825691,"virtual_id_number":825691,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-25 00:48:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-01-21 14:15:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-01-24 17:08:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-01-24 17:08:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-21 00:02:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-21 14:15:03.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-01-24 17:08:38.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"54803ad57182adb040e504e5e907eb0fabb47285","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-20 23:44:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cfe31179b973966efef16db7a9c7034d59919cbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 00:02:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17bd522306104e12bd1987b2bab38776 : SUCCESS in 3m 40s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a2ac000fa314954a723786b837bb42d : SUCCESS in 3m 38s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d9eb14b88104839b23ebd77022ac56b : SUCCESS in 3m 12s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/99a7b7504637433a8de05c7b9a9944dc : FAILURE in 8m 53s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/b00966304b9346cdab751fa914361781 : FAILURE in 9m 20s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7dc09c7087f14c03990d6d6cbc0d1459 : FAILURE in 6m 35s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10f3e1f709d449c4b87e642b78a314fc : FAILURE in 6m 52s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/277937edacd745348ce66d5fc26f81ce : FAILURE in 6m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"03e57d71aa6dc2de029000352b0db76dfc185887","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-21 00:18:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0b93b90225378e639d1bf787ad712415ee67fae6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 00:35:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76844af92dfc4d1eb6918fdd745f1279 : SUCCESS in 3m 44s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/407603fbe5b24d20bb2d969b363bdb1c : SUCCESS in 3m 34s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3d37622ce3641c988a8078f56d44133 : SUCCESS in 3m 40s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0e19fba9641141de8f5229f731e0dadc : FAILURE in 10m 25s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/54877675e76c4bb590081480874a9dd6 : FAILURE in 9m 21s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80c75bec90a842db895101f050a32a0d : FAILURE in 7m 21s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/526b5751077f4168ac11f3bb52e0cd9b : FAILURE in 6m 14s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2757a5f713694e8297615bc57405e32f : FAILURE in 6m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8f8296822254175126cc09a8c0abca7e2821ac22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-21 00:39:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3e3e009f951a0d466ec28f98dc8c8d1071b7413c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 01:10:00.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25eda8f85bbf425fb1ab061abcaa623b : SUCCESS in 3m 07s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a21433fdca043ee8fd77707efc71fde : SUCCESS in 3m 51s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a2a56272bc542c49432c85c48b56830 : SUCCESS in 3m 28s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/7e593398a1bf43d48b6b497f39193d72 : RETRY_LIMIT in 7m 12s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/f724d7564d4d457eb17451df4e1722f3 : RETRY_LIMIT in 6m 06s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1990c16b805f4a8bb07b268ded7a16e7 : FAILURE in 7m 41s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5d24586734144c8a4cb37a680664979 : FAILURE in 7m 20s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/367762e5125346c2882053ce5ae63149 : FAILURE in 7m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"13ab6d62f55d3af04ced679b3e24e889a686d697","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-21 01:46:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c23ab073c3cd8ee5b241b7536ff59d9628cd8279","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 02:14:42.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f38ba2b3a17e4aa888076615820e932a : SUCCESS in 3m 17s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e4155ea0f64416d93bfe5a08f9cf156 : SUCCESS in 3m 10s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05ff76758acc41419b588a2d8110d37c : SUCCESS in 3m 24s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/be22387b369d4cd892d0e7df61e0bbdc : RETRY_LIMIT in 8m 37s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/dc155f1d9ed847bd9acd57bb07027ccd : RETRY_LIMIT in 4m 54s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0107064f0d724e93921437e7878a8f3b : FAILURE in 8m 04s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f6f87157a5848c2be68a49d95b378e4 : FAILURE in 8m 05s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d412ae4101b4e9782d160610efd88a4 : FAILURE in 8m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9b0ba2fb8daa5c874e1c96aafd5d9902c8f9c6da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-21 02:34:32.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6e2797006fac5f9dd1c531b808f3ccb04a54f477","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 03:55:16.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c6595d833f049ffadc3021cbe416a30 : SUCCESS in 6m 03s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad687ad6be2d4d23b61ab1bf2589a9b0 : SUCCESS in 6m 07s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f3b0751bcf7494c97220016fef921cc : SUCCESS in 4m 57s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/1a8417e062804f459279e1d01113fed4 : RETRY_LIMIT in 10m 28s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/295f857e014347c3a0ead44cf4db60fa : RETRY_LIMIT in 7m 42s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75d2220aef0c4f439be20ecad19a27b5 : FAILURE in 12m 54s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/160ca038f2d340ac9f5bdcb779448aa0 : FAILURE in 10m 28s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d82bee0cb81b4291ae1fd0b3aa53c0a8 : FAILURE in 11m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d0611616744fd0a4e48be87cff03a4488d1697e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-21 04:37:47.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a4bc9459e728c6302c71a0393e5fa237bbddd3e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 05:04:05.000000000","message":"Patch Set 6: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"d6abcb1d7bd85c3d2283f99d2a6fd54f20cbf317","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-21 08:29:39.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8e8726e0346606d64e53fd158b6fc3b415c2fb61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 08:47:01.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b08e1ad30f884a9ba0c4efc422792739 : SUCCESS in 3m 28s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/74b494ed6c6c42f18a4bcb284646b90c : SUCCESS in 4m 00s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2b6dab3513244989e57aeb25063469a : SUCCESS in 3m 22s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/57509fef70284903a9a05d4c01f1f81f : FAILURE in 10m 52s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/46555f4691b94bfeb4e6247596ea7508 : FAILURE in 14m 19s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ecc23d7f60c54948b298ecd978289184 : FAILURE in 9m 37s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9dc9c11254548a29a675921c670efca : FAILURE in 13m 54s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b30f7abfe4944c9caa16dab3b8522b4c : FAILURE in 12m 03s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"20ab9eeecf2722e0a503a208a6e393f16dfee0bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-21 12:38:36.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"2d369a315e160578408f31c08e8168ddad4657d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 12:52:56.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04a9e665e3c14c6fb1a8cd08746c1439 : SUCCESS in 3m 17s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/351a7fe004e64704b4582c98cf4c9b7a : SUCCESS in 3m 08s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8cf8f7c0dfb4493f953d3759fc8c8234 : SUCCESS in 3m 24s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/64210d7bd67242a683066c651512483d : FAILURE in 9m 48s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/17d1ca35aad0479da6150dd871aa897e : FAILURE in 10m 19s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4c67648cd674d468296fbfc1b025e2d : FAILURE in 8m 51s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/934354eff04c4bb9bccc8cc7a63d3c46 : FAILURE in 7m 12s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48bbaeac83fa43499ca46cf12902dd11 : FAILURE in 7m 41s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"74a57582204a73d1c92f7dcc53e615de8d4d8007","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-21 13:01:17.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"390d22485b9b4812adfc5ff69f426f74e11a295d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 13:22:47.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/612bf2aa75cd47fd9f7e15ae69bce0f3 : SUCCESS in 3m 24s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90664651c58b46a1a4eed01fd3514bc8 : SUCCESS in 4m 36s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5017b06610104d20bb462d5acc16512c : SUCCESS in 3m 18s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/eace87a0aa0a44d38a329464caba619e : FAILURE in 11m 47s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/9d40a7abebbd4df5813738c67489b397 : FAILURE in 14m 01s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28da65d96f85404487a2706814c38675 : FAILURE in 12m 05s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fdf07b0b047b4197bb0000bc4e4f66cc : FAILURE in 15m 23s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42a3023bb12042c0a96fdfd591dd5731 : FAILURE in 9m 35s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"823d369768b6044f12cdcef3848dec015db09a42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-21 13:49:07.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"78fde69b3005feb54ee98c2ca84d5ed16a7dd06c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 14:11:30.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/537548f63d9d43a4a5ef37f2e91c8274 : SUCCESS in 8m 57s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a0cfff47c694b7897868f093ecb94b8 : FAILURE in 8m 10s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/587e6e9473e5400db3daeb9ecfabe391 : SUCCESS in 4m 11s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/da3fe4ea0ffc4a14b1bee1b95157ab1a : SUCCESS in 11m 25s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/82eb621658354b5580d36ddc1b98aba2 : SUCCESS in 11m 33s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fee46ade4a154ea1b83cc5dfc8ca7d8c : SUCCESS in 8m 48s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e3b41dc02ee45a08abb4211e4ae7f5b : SUCCESS in 14m 15s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8084a1782e94426fb45f27b90599c368 : SUCCESS in 8m 43s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"a3fe3be23afda948f61a7d517a650d3019695b8e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-01-21 14:15:03.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"bceac7cdfb12e7eed7a6b35c13679a7ff9531684","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 14:32:19.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/761e8609e0104569bf1102bf4fd1d567 : SUCCESS in 4m 19s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad334cb551864145b4ec5ecfa650650c : SUCCESS in 8m 07s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b0fb817718240408f626c32b4dcd2f4 : SUCCESS in 4m 10s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6edc1a5f1f9548a0833d9bea2d9601c9 : SUCCESS in 12m 54s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/b42554dd1e2f4932b4a3ae3cced01601 : SUCCESS in 13m 03s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dcd12f2ba2be436fbeab60db31cb5815 : SUCCESS in 11m 06s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/814deb4b6d914fc6864940ac14a01dfc : SUCCESS in 11m 14s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2142ff982ebc48c1bcf856b9359e0612 : SUCCESS in 11m 05s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"949e44abcb4810856eb95d6c75139be7df100d3c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-01-24 17:08:38.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"3bfeb7cb48caeef09b46ec7bfd061adb86504c1e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-24 23:02:29.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"bb69ace7b89ca2a6bddc96cf179d85da3d4b4a65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 23:24:42.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1319354a00a84a50904ff158c18bc680 : SUCCESS in 7m 47s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b781f33891f4fefa2d6272f84654811 : SUCCESS in 3m 39s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb30cfebf5934b669c9c343acfa5f24f : SUCCESS in 3m 52s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/5da67db5c3e04df080d3c331dbbe2b4c : SUCCESS in 10m 52s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/ccb83d59daa945b89780ce6ec6461c5f : SUCCESS in 10m 56s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa1621c08cb445aa8a4e473d20bfda21 : SUCCESS in 9m 45s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/428402143d1346d7ad87b14982c328a4 : SUCCESS in 8m 37s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/155c47c580bd4ec4bb60fed8eba95800 : SUCCESS in 9m 19s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"c5546e461bbf155721fe8b9c07d0b8c95f87bc1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 23:26:18.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"ab38255cc109f0a7469451b880c0e75a7a548dcd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 00:48:50.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6ba91b274ea468ab83588f4a6719688 : SUCCESS in 3m 52s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb7cb957daa64f529d8f12a35231e9dc : SUCCESS in 3m 20s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37b2c5e997c24b5a8dea1e41e5a6d9e0 : SUCCESS in 3m 54s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/715c4ea4bc7e4ef59e0584bc3535296d : SUCCESS in 12m 31s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/c7c54d56c3d1418cb5466863217451b1 : SUCCESS in 14m 29s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e01eabe9661143759a0527faa3dc5af5 : SUCCESS in 11m 30s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80d8956a413d45d2819f94f8c0467802 : SUCCESS in 10m 22s","accounts_in_message":[],"_revision_number":10},{"id":"5846f865bffb9cb742055ea8a8d83de8c7a1b253","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 00:48:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"fe519a63e80d1b0ff51dfde92b0ba3de9635fcc6","revisions":{"0120c6efddc1c99f87f43a35081b68f847b76a8f":{"kind":"REWORK","_number":1,"created":"2022-01-20 23:44:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/825691/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/91/825691/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/1"}}},"commit":{"parents":[{"commit":"da4a50abf44dd7cbe7fcca4df4bc6d24affa0d5d","subject":"Merge \"Fix dependency to purge default vhost config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/da4a50abf44dd7cbe7fcca4df4bc6d24affa0d5d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-20 23:44:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-20 23:44:32.000000000","tz":540},"subject":"Avoid testing details of oslo::messaging(::*)","message":"Avoid testing details of oslo::messaging(::*)\n\n... so that any change in puppet-oslo would not directly break unit\ntests.\n\nChange-Id: I9875a2cec185c9ed10f13fbe883c85cdc880358e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/0120c6efddc1c99f87f43a35081b68f847b76a8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/0120c6efddc1c99f87f43a35081b68f847b76a8f"}]},"branch":"refs/heads/master"},"f560e19ac83d22fa3ada1df5a66c2a3b6ac56205":{"kind":"REWORK","_number":2,"created":"2022-01-21 00:18:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/825691/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/91/825691/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/2"}}},"commit":{"parents":[{"commit":"da4a50abf44dd7cbe7fcca4df4bc6d24affa0d5d","subject":"Merge \"Fix dependency to purge default vhost config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/da4a50abf44dd7cbe7fcca4df4bc6d24affa0d5d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-20 23:44:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-21 00:18:05.000000000","tz":540},"subject":"Avoid testing details of oslo::messaging(::*)","message":"Avoid testing details of oslo::messaging(::*)\n\n... so that any change in puppet-oslo would not directly break unit\ntests.\n\nChange-Id: I9875a2cec185c9ed10f13fbe883c85cdc880358e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/f560e19ac83d22fa3ada1df5a66c2a3b6ac56205"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/f560e19ac83d22fa3ada1df5a66c2a3b6ac56205"}]},"branch":"refs/heads/master"},"74e808e3ab31650f4a17c2d255d10213ff5886be":{"kind":"REWORK","_number":3,"created":"2022-01-21 00:39:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/825691/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/91/825691/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/3"}}},"commit":{"parents":[{"commit":"da4a50abf44dd7cbe7fcca4df4bc6d24affa0d5d","subject":"Merge \"Fix dependency to purge default vhost config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/da4a50abf44dd7cbe7fcca4df4bc6d24affa0d5d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-20 23:44:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-21 00:39:31.000000000","tz":540},"subject":"Avoid testing details of oslo::messaging(::*)","message":"Avoid testing details of oslo::messaging(::*)\n\n... so that any change in puppet-oslo would not directly break unit\ntests.\n\nChange-Id: I9875a2cec185c9ed10f13fbe883c85cdc880358e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/74e808e3ab31650f4a17c2d255d10213ff5886be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/74e808e3ab31650f4a17c2d255d10213ff5886be"}]},"branch":"refs/heads/master"},"69af50a8b46db555665f91f7b7f8c185cadbb5ec":{"kind":"REWORK","_number":4,"created":"2022-01-21 01:46:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/825691/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/91/825691/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/4"}}},"commit":{"parents":[{"commit":"da4a50abf44dd7cbe7fcca4df4bc6d24affa0d5d","subject":"Merge \"Fix dependency to purge default vhost config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/da4a50abf44dd7cbe7fcca4df4bc6d24affa0d5d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-20 23:44:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-21 01:46:44.000000000","tz":540},"subject":"Avoid testing details of oslo::messaging(::*)","message":"Avoid testing details of oslo::messaging(::*)\n\n... so that any change in puppet-oslo would not directly break unit\ntests.\n\nChange-Id: I9875a2cec185c9ed10f13fbe883c85cdc880358e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/69af50a8b46db555665f91f7b7f8c185cadbb5ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/69af50a8b46db555665f91f7b7f8c185cadbb5ec"}]},"branch":"refs/heads/master"},"1bc5a255ae36c69b8ba91bfb68cd77f9d2f5976f":{"kind":"REWORK","_number":5,"created":"2022-01-21 02:34:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/825691/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/91/825691/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/5"}}},"commit":{"parents":[{"commit":"5d7577fa4beb73c23eacb89168f77ed200acde91","subject":"Remove validations of oslo::messaging::rabbit parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/5d7577fa4beb73c23eacb89168f77ed200acde91"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-20 23:44:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-21 02:34:22.000000000","tz":540},"subject":"Avoid testing details of oslo::messaging(::*)","message":"Avoid testing details of oslo::messaging(::*)\n\n... so that any change in puppet-oslo would not directly break unit\ntests.\n\nChange-Id: I9875a2cec185c9ed10f13fbe883c85cdc880358e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/1bc5a255ae36c69b8ba91bfb68cd77f9d2f5976f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/1bc5a255ae36c69b8ba91bfb68cd77f9d2f5976f"}]},"branch":"refs/heads/master"},"54bdafac4b418408256132e083390be9a88f7a36":{"kind":"REWORK","_number":6,"created":"2022-01-21 04:37:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/825691/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/91/825691/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/6"}}},"commit":{"parents":[{"commit":"5d7577fa4beb73c23eacb89168f77ed200acde91","subject":"Remove validations of oslo::messaging::rabbit parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/5d7577fa4beb73c23eacb89168f77ed200acde91"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-20 23:44:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-21 04:37:38.000000000","tz":540},"subject":"Avoid testing details of oslo::messaging(::*)","message":"Avoid testing details of oslo::messaging(::*)\n\n... so that any change in puppet-oslo would not directly break unit\ntests.\n\nChange-Id: I9875a2cec185c9ed10f13fbe883c85cdc880358e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/54bdafac4b418408256132e083390be9a88f7a36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/54bdafac4b418408256132e083390be9a88f7a36"}]},"branch":"refs/heads/master"},"fa56ce9a9239d9b66680fd4a902a31de120af900":{"kind":"REWORK","_number":7,"created":"2022-01-21 08:29:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/825691/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/91/825691/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/7"}}},"commit":{"parents":[{"commit":"5d7577fa4beb73c23eacb89168f77ed200acde91","subject":"Remove validations of oslo::messaging::rabbit parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/5d7577fa4beb73c23eacb89168f77ed200acde91"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-20 23:44:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-21 08:29:30.000000000","tz":540},"subject":"Avoid testing details of oslo::messaging(::*)","message":"Avoid testing details of oslo::messaging(::*)\n\n... so that any change in puppet-oslo would not directly break unit\ntests.\n\nChange-Id: I9875a2cec185c9ed10f13fbe883c85cdc880358e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/fa56ce9a9239d9b66680fd4a902a31de120af900"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/fa56ce9a9239d9b66680fd4a902a31de120af900"}]},"branch":"refs/heads/master"},"02b1a27d66c009086e5a85fd487abfcd86bd5fc7":{"kind":"REWORK","_number":8,"created":"2022-01-21 12:38:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/825691/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/91/825691/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/8"}}},"commit":{"parents":[{"commit":"5d7577fa4beb73c23eacb89168f77ed200acde91","subject":"Remove validations of oslo::messaging::rabbit parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/5d7577fa4beb73c23eacb89168f77ed200acde91"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-20 23:44:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-21 12:38:27.000000000","tz":540},"subject":"Avoid testing details of oslo::messaging(::*)","message":"Avoid testing details of oslo::messaging(::*)\n\n... so that any change in puppet-oslo would not directly break unit\ntests.\n\nChange-Id: I9875a2cec185c9ed10f13fbe883c85cdc880358e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/02b1a27d66c009086e5a85fd487abfcd86bd5fc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/02b1a27d66c009086e5a85fd487abfcd86bd5fc7"}]},"branch":"refs/heads/master"},"0353bf3cf6f90593423c2b559131ebd57389803f":{"kind":"REWORK","_number":9,"created":"2022-01-21 13:01:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/825691/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/91/825691/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/9"}}},"commit":{"parents":[{"commit":"5d7577fa4beb73c23eacb89168f77ed200acde91","subject":"Remove validations of oslo::messaging::rabbit parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/5d7577fa4beb73c23eacb89168f77ed200acde91"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-20 23:44:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-21 13:00:58.000000000","tz":540},"subject":"Avoid testing details of oslo::messaging(::*)","message":"Avoid testing details of oslo::messaging(::*)\n\n... so that any change in puppet-oslo would not directly break unit\ntests.\n\nChange-Id: I9875a2cec185c9ed10f13fbe883c85cdc880358e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/0353bf3cf6f90593423c2b559131ebd57389803f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/0353bf3cf6f90593423c2b559131ebd57389803f"}]},"branch":"refs/heads/master"},"fe519a63e80d1b0ff51dfde92b0ba3de9635fcc6":{"kind":"REWORK","_number":10,"created":"2022-01-21 13:49:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/825691/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/91/825691/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/91/825691/10"}}},"commit":{"parents":[{"commit":"5d7577fa4beb73c23eacb89168f77ed200acde91","subject":"Remove validations of oslo::messaging::rabbit parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/5d7577fa4beb73c23eacb89168f77ed200acde91"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-20 23:44:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-21 13:48:41.000000000","tz":540},"subject":"Avoid testing details of oslo::messaging(::*)","message":"Avoid testing details of oslo::messaging(::*)\n\n... so that any change in puppet-oslo would not directly break unit\ntests.\n\nChange-Id: I9875a2cec185c9ed10f13fbe883c85cdc880358e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/fe519a63e80d1b0ff51dfde92b0ba3de9635fcc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/fe519a63e80d1b0ff51dfde92b0ba3de9635fcc6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
