)]}'
{"id":"openstack%2Fpuppet-designate~833426","triplet_id":"openstack%2Fpuppet-designate~stable%2Fwallaby~I28a91ab99325970db1e96aa904f35e8f9a5e109c","project":"openstack/puppet-designate","branch":"stable/wallaby","topic":"bug/1961516","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-03-16 21:51:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I28a91ab99325970db1e96aa904f35e8f9a5e109c","subject":"Fix setup of bind9 backend","status":"MERGED","created":"2022-03-13 16:18:42.000000000","updated":"2022-03-16 21:52:53.000000000","submitted":"2022-03-16 21:51:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"833426-bug/1961516","cherry_pick_of_change":832763,"cherry_pick_of_patch_set":2,"meta_rev_id":"ecc4b82da0a25252c0dc77e8fb5a3122908e6642","_number":833426,"virtual_id_number":833426,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-16 21:51:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-03-14 09:24:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-03-16 19:54:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-03-16 19:54:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-13 17:46:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-14 09:24:07.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-03-16 19:54:26.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"8a05d44fdba107c0f4dfd71978a3552a31cd24da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-13 16:18:42.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"f804db24e8cdfa0f910a0e06764042dd5829ef5a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-13 16:19:51.000000000","message":"Topic bug/1961516-stable/wallaby removed","accounts_in_message":[],"_revision_number":1},{"id":"53cef4cef135da919e90258d0e9e15849bd50a65","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-13 16:19:53.000000000","message":"Topic set to bug/1961516","accounts_in_message":[],"_revision_number":1},{"id":"e42c92e8decd9b8716176f14d2ed319994dc9354","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-13 17:46:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff3ee7eecbdd4e608d46fb37a29b9d69 : SUCCESS in 4m 36s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24e50a8d26804d54a464de4cace7f987 : SUCCESS in 4m 30s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/4900232c96a348c6a31531a63127cf76 : SUCCESS in 12m 29s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5db5c9c333fe47a4bece6a9f70b6d78c : SUCCESS in 10m 01s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce1e75b55e574cdbaaf65e796ac7ea89 : SUCCESS in 9m 10s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/266a86dd35054d03bc54a40978348bfb : FAILURE in 1h 00m 09s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a29ce845a8b34f9984f955c5047e92dc : SUCCESS in 1h 22m 54s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/29608fe0e87c4978b551ed00d7696fc2 : SUCCESS in 21m 43s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2495537d7624d44af942fda29e682a1 : SUCCESS in 15m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0a74f08aaaf429591d9d923b9f85e8d : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":1},{"id":"b2cfc447dee1e9bee2660a574fb4be08a3fd1f76","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-03-14 09:24:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2cf15859b7b0f68c05852bcee2ede9ecee01a121","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-03-16 19:54:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3af47a2bb0a13639c9c3d410bd3592fdd43c4e4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 19:55:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"de2de49fc6f6b3703a2fed700e5edffc608b9538","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 21:51:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80a371ecfc7b4dc5a017159a9f524617 : SUCCESS in 4m 59s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c1de91cdcc9444289f28dc760710252 : SUCCESS in 4m 54s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/aae78236981148099b62726085e84f26 : SUCCESS in 13m 25s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26f673078b91435cbd1139afb7cc8d2e : SUCCESS in 10m 39s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f863834343e442d199cc19433e170dba : SUCCESS in 1h 50m 17s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/4558ee7fde254657b592f9435dd14dc6 : SUCCESS in 24m 26s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36001c96189d473e960f52c75bb028c4 : SUCCESS in 14m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a0db3042b22430fb730bd0db780060d : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":1},{"id":"2d5cd323a90280bf50556c8ca5471f70829b85a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 21:51:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ecc4b82da0a25252c0dc77e8fb5a3122908e6642","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 21:52:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73774814e66a4a409c07d2f463e97c9e : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"34bffaea0075ca45d1d72ee39d1d3d2b1e6d3d37","revisions":{"34bffaea0075ca45d1d72ee39d1d3d2b1e6d3d37":{"kind":"REWORK","_number":1,"created":"2022-03-13 16:18:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/833426/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/26/833426/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/26/833426/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/26/833426/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/26/833426/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/26/833426/1"}}},"commit":{"parents":[{"commit":"6d06a978263678388268d177cb7e2378b2b2bc71","subject":"Prepare a new Wallaby release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/6d06a978263678388268d177cb7e2378b2b2bc71"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-20 08:41:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-13 16:18:42.000000000","tz":0},"subject":"Fix setup of bind9 backend","message":"Fix setup of bind9 backend\n\nDesignate no longer uses the parameters in the [backend:bind9] section,\nsince ppols.yaml was introduced. This change ensures the parameters in\nthis wrong section are purged.\n\nAlso, this change introduces support for setting up backend using\npools.yaml according to the current installation guide.\n\nBackport note\n- Adjusted some logic to use hard-coded user/group names, because\n  34ff6e6f04a69fec9579accc0bcc5f1c29049abe is not present in\n  stable/xena.\n\n- Squashed the commit 979dd5b2dc69590a457f2c318b2dfc0fd6810dcd which\n  fixed the wrong default dns_port.\n\nCloses-Bug: #1961516\nChange-Id: I28a91ab99325970db1e96aa904f35e8f9a5e109c\n(cherry picked from commit 22019e46e2553e047fd7706c82d29d3cfb978e39)\n(cherry picked from commit 93f0d9b7aa09d1c6e40d5775e3d924845508a6f5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/34bffaea0075ca45d1d72ee39d1d3d2b1e6d3d37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/34bffaea0075ca45d1d72ee39d1d3d2b1e6d3d37"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
