)]}'
{"id":"openstack%2Fpuppet-designate~837470","triplet_id":"openstack%2Fpuppet-designate~stable%2Fwallaby~I3e0682b358fe33b7ae341f3d7a5a1de5943c30e3","project":"openstack/puppet-designate","branch":"stable/wallaby","topic":"mdns-opts-stable/wallaby","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-04-12 16:17:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3e0682b358fe33b7ae341f3d7a5a1de5943c30e3","subject":"mdns: Accept array value for the listen parameter","status":"MERGED","created":"2022-04-12 06:59:09.000000000","updated":"2022-04-12 16:17:49.000000000","submitted":"2022-04-12 16:17:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837470-mdns-opts-stable/wallaby","cherry_pick_of_change":837329,"cherry_pick_of_patch_set":1,"meta_rev_id":"3663bf4a26b9de5351041263b1e1129060f8bbdb","_number":837470,"virtual_id_number":837470,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-12 16:17:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-04-12 08:22:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-04-12 14:20:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-04-12 14:20:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-12 08:13:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-12 08:22:43.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-04-12 14:20:54.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"5c6d72d6b61b0617917c339fa2bfc66a70415f07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-12 06:59:09.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"9a8b5759d6997525d08848bac11dbbf1be64de11","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-12 06:59:16.000000000","message":"Topic mdns-opts-stable/xena-stable/wallaby removed","accounts_in_message":[],"_revision_number":1},{"id":"0aee6ff3eced10e9ae3d449d6bc06337bb45cb35","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-12 06:59:21.000000000","message":"Topic set to mdns-opts-stable/wallaby","accounts_in_message":[],"_revision_number":1},{"id":"38b369fefe1da6e0a6081a048785b7a2c959bcdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 08:13:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36bb303f702d44a48e1c574f731d4130 : SUCCESS in 3m 38s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23cf3eb7a6d44df98fa446ad375f3d37 : SUCCESS in 3m 33s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/5fb1f91661b244a8b0140a97291543e6 : SUCCESS in 11m 43s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ec6972256f742a991b2821cd88b83a9 : SUCCESS in 8m 22s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba91f03a0e234f3b99e96a07074e2929 : SUCCESS in 8m 33s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/488dd696074d448b99bc1ebfbd4545c4 : FAILURE in 31m 15s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8ca0a15ee7df498f9e0c313cae544899 : SUCCESS in 1h 12m 26s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/14d7bc5f945e4e338c9091edd2994d5b : SUCCESS in 20m 51s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5ded93341524c63b34982aee35fd08f : SUCCESS in 15m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"08e4efacd75d0102776a33f16cba68375192d57b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-04-12 08:22:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"26e10fea0cd5a87ebc9a0a6b6bbb22bbf4c97dfb","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-04-12 14:20:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"84d3c38bb716ccee2f20693b4f66ea9f9a2cad35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 14:22:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1322b9600f74ae2714e1aa0cd18da3cd2a5f5b2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 16:17:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b912c66c0eae4e3db39fe43307f54e37 : SUCCESS in 4m 04s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9b7ce83be7d4872bc39803ceee1560a : SUCCESS in 4m 20s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/814bc5b14f804f96a48ec3655d9d9d62 : SUCCESS in 13m 32s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/837b7a6d79854477a7dad17b1da5b0cd : SUCCESS in 9m 24s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/66bddd3850d344a18d661912a291d209 : SUCCESS in 1h 44m 36s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/802777d6d47f4322850f926f580234d8 : SUCCESS in 23m 48s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ff1d0d7c2a9411c92c7faaea4496441 : SUCCESS in 16m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3663bf4a26b9de5351041263b1e1129060f8bbdb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 16:17:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7a2e878d87d4f0d7015c9be705e391f23884a3d5","revisions":{"7a2e878d87d4f0d7015c9be705e391f23884a3d5":{"kind":"REWORK","_number":1,"created":"2022-04-12 06:59:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/837470/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/70/837470/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/70/837470/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/70/837470/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/70/837470/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/70/837470/1"}}},"commit":{"parents":[{"commit":"34bffaea0075ca45d1d72ee39d1d3d2b1e6d3d37","subject":"Fix setup of bind9 backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/34bffaea0075ca45d1d72ee39d1d3d2b1e6d3d37"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-11 05:30:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-12 06:59:09.000000000","tz":0},"subject":"mdns: Accept array value for the listen parameter","message":"mdns: Accept array value for the listen parameter\n\n... because this parameter is ListOpt which takes a comma-separated\nstring, which represents a list value.\n\nChange-Id: I3e0682b358fe33b7ae341f3d7a5a1de5943c30e3\n(cherry picked from commit 61233cea55834a9536f788d905877c19e994ef4a)\n(cherry picked from commit 8ac604bb93d8a03d112480eac4e27cf2000b8ae5)\n(cherry picked from commit 2b80fed4051dd60f7339e21abc41d30fd2ce7bdd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/7a2e878d87d4f0d7015c9be705e391f23884a3d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/7a2e878d87d4f0d7015c9be705e391f23884a3d5"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
