)]}'
{"id":"openstack%2Fpuppet-designate~840211","triplet_id":"openstack%2Fpuppet-designate~master~I922892d2317d1ea64aaf5965cfeef950dc885504","project":"openstack/puppet-designate","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-05-11 10:08:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I922892d2317d1ea64aaf5965cfeef950dc885504","subject":"Central: Remove deprecated max_domain_name_len","status":"MERGED","created":"2022-05-02 15:30:27.000000000","updated":"2022-05-11 10:10:34.000000000","submitted":"2022-05-11 10:08:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840211","meta_rev_id":"a82b7e1f49d72e597b2cf0b6f852d78915775fcc","_number":840211,"virtual_id_number":840211,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-11 10:08:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-05-06 20:54:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-05-11 09:10:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-05-11 09:10:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-02 16:22:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-06 20:54:00.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-05-11 09:10:19.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"65782c896d147764417a52c36a4d17bca3e5aeb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-02 15:30:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5402b4ca765e072f4d52343a2497736a79f8e5bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 16:22:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/acacb5aa31ce4a2a90c548aab75bfce7 : SUCCESS in 3m 20s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f1f049947df4ce1bde6aee6d041e450 : SUCCESS in 3m 09s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9c9f958c5024b57bfd29643b69ba13e : SUCCESS in 2m 58s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/a37d35c6cdc9435aa8666cf921146df2 : SUCCESS in 11m 52s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8a436941fc44a0c90b5d251bcea30b0 : SUCCESS in 8m 26s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e39777836e44e75b1a97c9df525b995 : SUCCESS in 8m 29s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b985cc3df4a40079560798fe7edfb3f : SUCCESS in 9m 26s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/498a80b4e6f04219afc90de43dfa56d1 : FAILURE in 43m 43s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c1f55d1244cc4eaab12f5385fbff5517 : FAILURE in 10m 24s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49909e20cfc04c4ab3cb817609fc5d4a : SUCCESS in 14m 10s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/0cb643f03f31419ab2f80858c0186e91 : SUCCESS in 21m 23s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/3c7365fd06664a5b85570f65e9525316 : FAILURE in 12m 38s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b412d9e4e439482787eaf35d34b2b5e3 : SUCCESS in 14m 54s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e273e335cd824c28a91b948536fd09a0 : SUCCESS in 15m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3560894ad1e4793914478ea291d5fe2 : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":1},{"id":"40b464aa9de443256c44891516397ea5e0f3bce6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-06 02:37:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ba0df2cf5131d12f2b7e531230da19caa3409878","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 04:02:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/398910ed86644959b04d4bc16adedd41 : SUCCESS in 3m 00s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ced6cceaa9914d288363adfbabb04b4c : SUCCESS in 3m 16s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4aad6e3abb644710be629d8e0c70e95e : SUCCESS in 3m 32s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/7180312fa304403584cc40742e2bc15d : SUCCESS in 12m 16s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5147593f5efc4483a9fab69ff0d07790 : SUCCESS in 8m 59s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe628e1934704c868b6070667885887f : SUCCESS in 9m 44s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8124d6b810d04775b3b111d0b04ec9d1 : SUCCESS in 12m 14s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f73999ce4fd4f45a777ff267c97fe9f : FAILURE in 46m 52s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/aee976ed18d443e5a10dec1226ea0e12 : SUCCESS in 1h 10m 19s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/185b98c7fb3144feb1d4959846e2010e : SUCCESS in 14m 37s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/f2c83cadf1e6437aa01737650db0bccc : SUCCESS in 21m 24s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/4da8b4c7095a4fb4ab68d4b0087e833a : FAILURE in 8m 40s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d1de6e5c35245f3ac6121f2f647e182 : SUCCESS in 13m 33s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8f3c035e42f40a39259141449b11787 : SUCCESS in 14m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21d6f68a98b847e9a48f68d75abffc4b : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"3c09e36fd360a97e356f837762bf3419e9940e75","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-05-06 20:54:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f2a37686cd1ac85e0951b35d50d3feec68b191da","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-05-11 09:10:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dfbf050792a0f1291b08344211a147056d356245","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 09:11:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8f919541aa0cc6c2d925f900ef1f1d152a348413","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 10:08:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15834af0365c4bb99aeca49ec647070c : SUCCESS in 3m 18s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7724674940f64375837fe0a61faee31f : SUCCESS in 3m 42s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb2b8cda3c3c4bd1b0acd569edab2eff : SUCCESS in 3m 25s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/d0d6d38c832f4474835e3b29f89a2c58 : SUCCESS in 10m 05s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb2d44eb6659481baa29901735aff55a : SUCCESS in 9m 58s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c58c5d88a74548d6b886255bf9d334f1 : SUCCESS in 9m 27s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9337f1ff8600481ebb251ec94d6b8876 : SUCCESS in 57m 00s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/1addf7c51d714634b0e4db7ad263afa1 : SUCCESS in 22m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c9c8415441043aaba461494f5f592df : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":1},{"id":"3b6b1e75723d8b9c6567db4c988aff393f6a4a60","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 10:08:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a82b7e1f49d72e597b2cf0b6f852d78915775fcc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 10:10:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ea8f132c4e34cd280c4af75f47a9ca2 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d096b9dc4b46008928f971613af58988c75279da","revisions":{"d096b9dc4b46008928f971613af58988c75279da":{"kind":"REWORK","_number":1,"created":"2022-05-02 15:30:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/840211/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/11/840211/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/11/840211/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/11/840211/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/11/840211/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/11/840211/1"}}},"commit":{"parents":[{"commit":"04e1fb4c235cc49d1f727741cfb08546a97fc119","subject":"Merge \"mdns: Improve parameter coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/04e1fb4c235cc49d1f727741cfb08546a97fc119"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-02 15:29:43.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-02 15:30:08.000000000","tz":540},"subject":"Central: Remove deprecated max_domain_name_len","message":"Central: Remove deprecated max_domain_name_len\n\nThe parameter was deprecated during the previous cycle[1].\n\n[1] 1f0071ef8cccefb57daf2e1e87341e1abb4eb9e0\n\nChange-Id: I922892d2317d1ea64aaf5965cfeef950dc885504\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/d096b9dc4b46008928f971613af58988c75279da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/d096b9dc4b46008928f971613af58988c75279da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
