)]}'
{"id":"openstack%2Fpuppet-designate~846481","triplet_id":"openstack%2Fpuppet-designate~master~Ib67b44cb289e75bfed022cec350fd2cd4bb6374b","project":"openstack/puppet-designate","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-06-20 23:02:21.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ib67b44cb289e75bfed022cec350fd2cd4bb6374b","subject":"bind9 backend: Clean up deprecated rndc_host","status":"MERGED","created":"2022-06-19 12:59:54.000000000","updated":"2022-06-21 00:59:57.000000000","submitted":"2022-06-21 00:58:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"846481","meta_rev_id":"495bda51d4d13bced27113c4e3e14c1ee7ff93e2","_number":846481,"virtual_id_number":846481,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-21 00:58:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-06-20 17:50:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-06-20 18:31:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-06-20 18:31:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-19 14:33:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-20 17:50:59.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-06-20 18:31:26.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"a3701208c5949ca67c557e2f8dac263a7a0d5bba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-19 12:59:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"caae8b7006d6e365cc39ff1b5421fb4189eeb28c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-19 13:00:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5159aee0767cfcb49ac56d3e88823695342c03a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-19 13:38:48.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"d2d2b366bc619fccdbb335f744310088bb28314a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-19 14:33:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3655a4c54d4f484ba426137ab51ac32b : SUCCESS in 3m 47s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d54dfd9a72e4b249e3218a14fe42db5 : SUCCESS in 4m 17s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b23641937f7e49db893ab5ca9bc9317e : SUCCESS in 4m 02s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd0d614d94cf4e618ee48bbce62a8a78 : SUCCESS in 9m 45s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/930b3123fa764c9894c29e0022b68752 : SUCCESS in 10m 22s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0dd36997aeef465bb603bab8524c16ca : SUCCESS in 9m 51s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9599cffcebf04a259bbf0a0b137faeb8 : SUCCESS in 9m 03s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f047320b057f4e2a91cd6db5fad58516 : FAILURE in 38m 10s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b09ebc738f2b4f3d83071eacc19cf1f7 : SUCCESS in 53m 16s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37e782f7b79942b8b859ded51eb8a230 : SUCCESS in 13m 09s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/463d237fe167441eabb5ccf21fa54145 : SUCCESS in 18m 04s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cbf565a7654a4310b62a2bba9692bcc0 : SUCCESS in 16m 41s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1b9b4a1d7b549bc9555682f15220d47 : SUCCESS in 12m 52s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c913893c1ddc4d64b5a388ef40a78368 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":3},{"id":"b0a31e171a79366d37897b48f763a8551e089e91","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-06-20 17:50:59.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"cea29db3da7cdcc317905413e61cdc1184f9c148","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-06-20 18:31:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"92deaf9be299b382ffeaa311b7dc7a7c7ded5d26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 18:31:39.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n\nWarning:\n  Change 846491 in project openstack/puppet-openstack-integration does not share a change queue with 846481 in project openstack/puppet-designate","accounts_in_message":[],"_revision_number":3},{"id":"c1ec5394c890d0d8ae3e6314ce41129d96cf7b09","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-20 23:02:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d557b77b3d5f6c9bb3965b0cd4a9d2b848deed6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 00:09:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d51ff551ed4f46c3917357cf344fdd14 : SUCCESS in 3m 10s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cadbeaeef37e440093425446417701b7 : SUCCESS in 3m 34s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30c5f50d257049628fbdd94ee9f74ab2 : SUCCESS in 3m 04s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/b00e9652ee0e4063b35e0bb9c8968d3e : SUCCESS in 13m 31s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e95319b6320481a984e1e5643e008dc : SUCCESS in 8m 03s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3fc5672d95e411fb7489e19f39bccb4 : SUCCESS in 7m 07s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b177ca8099014e2bb27f28e47a70aec1 : SUCCESS in 7m 53s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/012616ba18744b748567d62d8fccc738 : FAILURE in 39m 20s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/239bad40381843598ede6bece199492f : SUCCESS in 1h 02m 11s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38ca5461306b4839888b37b5d7eb909d : SUCCESS in 15m 46s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/17283d96b87247b1b5c20fb9d99919b3 : SUCCESS in 19m 03s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94703a2c4cc9407292f418839278b89b : SUCCESS in 15m 52s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9e85cc93ac943e6a86eddd396f1116f : SUCCESS in 15m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c619b23af2a240449e964eac0ff3a1e7 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":3},{"id":"b16f98fbd65e5643344c2b3c427919de13b5e18e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 00:10:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"257a6822fe372a7711a620f377742110383e8ed0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 00:58:52.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a5b8e65ca504318b2e583c9a741c4fc : SUCCESS in 2m 33s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/225442329c6e486880bdb722d38fe1da : SUCCESS in 2m 47s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb9cb0732004423d9dde8de73bdcceef : SUCCESS in 2m 53s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/9ae91e30ccb648848b960bcd98ef5331 : SUCCESS in 9m 04s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d23f43572a6e4ab9961fa005a1ebf648 : SUCCESS in 7m 12s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe69caf96eb04d289abeeacfa6623a9a : SUCCESS in 6m 54s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5463cb47c9ab498db80cb40e1a2598fd : SUCCESS in 47m 32s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/44a8a833f4a7429c834604f6a2c4be4f : SUCCESS in 17m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d3c60eb650042a18197ecb115ffb157 : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":3},{"id":"7058ad040971464913d580592507fc652e98e5a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 00:58:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"495bda51d4d13bced27113c4e3e14c1ee7ff93e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 00:59:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8462536af54f4fb59e1f095c323d513e : SUCCESS in 46s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"09eaea3132938aa1354ae7f6f7df314768b3c6f9","revisions":{"c6dcfbbd7cbb94134ad1f02e4a3fae00b35849bf":{"kind":"REWORK","_number":1,"created":"2022-06-19 12:59:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/81/846481/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/81/846481/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/81/846481/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/81/846481/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/81/846481/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/81/846481/1"}}},"commit":{"parents":[{"commit":"c8128823eb61f444cfe812fe7490ffa98ae34df7","subject":"bind9 backend: Manage pool by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/c8128823eb61f444cfe812fe7490ffa98ae34df7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-19 12:59:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-19 12:59:08.000000000","tz":540},"subject":"bind9 backend: Clean up deprecated rndc_host","message":"bind9 backend: Clean up deprecated rndc_host\n\n... and the logics to remove old parameters.\n\nChange-Id: Ib67b44cb289e75bfed022cec350fd2cd4bb6374b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/c6dcfbbd7cbb94134ad1f02e4a3fae00b35849bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/c6dcfbbd7cbb94134ad1f02e4a3fae00b35849bf"}]},"branch":"refs/heads/master"},"d678ff7425a5ce0e7fb473d623dd69fe4451cbd0":{"kind":"REWORK","_number":2,"created":"2022-06-19 13:00:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/81/846481/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/81/846481/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/81/846481/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/81/846481/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/81/846481/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/81/846481/2"}}},"commit":{"parents":[{"commit":"35d366034aa9c99ed08e32742700601e2c219950","subject":"bind9 backend: Manage pool by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/35d366034aa9c99ed08e32742700601e2c219950"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-19 12:59:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-19 13:00:31.000000000","tz":540},"subject":"bind9 backend: Clean up deprecated rndc_host","message":"bind9 backend: Clean up deprecated rndc_host\n\n... and the logics to remove old parameters.\n\nChange-Id: Ib67b44cb289e75bfed022cec350fd2cd4bb6374b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/d678ff7425a5ce0e7fb473d623dd69fe4451cbd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/d678ff7425a5ce0e7fb473d623dd69fe4451cbd0"}]},"branch":"refs/heads/master"},"09eaea3132938aa1354ae7f6f7df314768b3c6f9":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-06-19 13:38:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/81/846481/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/81/846481/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/81/846481/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/81/846481/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/81/846481/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/81/846481/3"}}},"commit":{"parents":[{"commit":"35d366034aa9c99ed08e32742700601e2c219950","subject":"bind9 backend: Manage pool by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/35d366034aa9c99ed08e32742700601e2c219950"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-19 12:59:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-19 13:38:48.000000000","tz":0},"subject":"bind9 backend: Clean up deprecated rndc_host","message":"bind9 backend: Clean up deprecated rndc_host\n\n... and the logics to remove old parameters.\n\nDepends-on: https://review.opendev.org/846491\nChange-Id: Ib67b44cb289e75bfed022cec350fd2cd4bb6374b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/09eaea3132938aa1354ae7f6f7df314768b3c6f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/09eaea3132938aa1354ae7f6f7df314768b3c6f9"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
