)]}'
{"id":"openstack%2Fpuppet-designate~849164","triplet_id":"openstack%2Fpuppet-designate~master~Iccf1bcc19bb1e65fbccf61dc6f17429d6a3b88bd","project":"openstack/puppet-designate","branch":"master","topic":"config-acceptance","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-07-15 20:22:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iccf1bcc19bb1e65fbccf61dc6f17429d6a3b88bd","subject":"Extend coverage of config acceptance tests","status":"MERGED","created":"2022-07-09 03:23:42.000000000","updated":"2022-07-15 20:22:01.000000000","submitted":"2022-07-15 20:22:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849164-config-acceptance","meta_rev_id":"bd498a052007888397c507acc53185530301e16c","_number":849164,"virtual_id_number":849164,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-15 20:22:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-07-11 21:52:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-07-15 19:15:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-07-15 19:15:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-09 03:42:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-11 21:52:40.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-07-15 19:15:46.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"8a8528b097d5d5ebf8ccb34cbb5ef31f345b564f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 03:23:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ca91515e4c6d37c7acca309b05d7d7d0e1f9b42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 03:42:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c180fa568bf4ec58b19bbc7e01b2368\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25398165cc8a424d822e122863655890 : SUCCESS in 3m 02s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b50d0c7ce45341029300014133b1e58c : SUCCESS in 2m 44s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04e0b04080df4c51bc581175c292ab32 : SUCCESS in 2m 42s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05a3102730d5413f8e67cedde3dade2e : FAILURE in 11m 54s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/27858312192a45dba832b18110915111 : FAILURE in 14m 37s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51d41f9cffa243fc813ab651572b2efe : FAILURE in 13m 02s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f14c85eeb4104ee8b789a544848de17b : FAILURE in 13m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0dbcb2c1d9af660966ba8635d5e4e9f539333293","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 04:30:51.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"ac743adc50a7f4042de2136b108f442e5a463ec9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 04:51:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f814517e24c4d80a411f4d07615bfa5\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/515ff8b1ee054c4fa67e3d41693cd7e1 : SUCCESS in 3m 23s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1ffdff72f804957864c3998a52c1846 : SUCCESS in 3m 09s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f20a370e267045cd8ff015ee8361242c : SUCCESS in 3m 05s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eeee1f084df84d3ea8891aa0a6f6273a : SUCCESS in 15m 39s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/d75b1815840843ff915517eb57a5e940 : SUCCESS in 18m 05s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8fe08e826dd423791b06bbaa7aa612b : SUCCESS in 15m 01s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6cb37d4d77b4af4a27a5f50b8cadd80 : SUCCESS in 15m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7688d6f2ada0953f281b56866f29211ee15b1b34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 10:16:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a0241a7a046eee646a791212969d7f6b7bd61972","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 10:23:42.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"251358580da192138435cfde23ad05f9eecef20c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 10:23:56.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"98683f6f662a758d7f1d641b63a86f8102166ee0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 10:24:22.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"1ba7b5dea305ca3a22220fd5678146e891ba12a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 10:54:28.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b964627cba0e4f66af04a5c5a39f0dea\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0976d53e0084d5aa89ec38cd1dadee7 : SUCCESS in 3m 05s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb4d2839b40b436cbb355c853d910c2e : SUCCESS in 3m 15s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ad3df78b3a14850a30f7eb5bd4913bf : SUCCESS in 2m 54s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f88b9f84ba9e4b76a5190dcb46e691ae : SUCCESS in 14m 33s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/e9b2f73e3b7e47d88c201fa02430667e : FAILURE in 28m 23s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/165559471493403bba00fb9e63094ac7 : SUCCESS in 15m 28s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6e3021d97a14893bd8b295f9e2e3d61 : SUCCESS in 14m 42s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0b08c2d40e644197f057a73abf9af0a45d84b29d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 12:05:19.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a36bee0b2abaa1d65aa9e62161609647e43c8feb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 12:30:57.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f7542c069a845ffbbc849510ac2a5f9\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f33bcce8ac9748b3b2f86535a3da7ebf : SUCCESS in 3m 01s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1721c20a7bb745a589f1eeeb27ce844f : SUCCESS in 3m 04s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc7ea8af2f54430399f6b18cec29325f : SUCCESS in 3m 38s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c0999066315434e9e1670d21636cd09 : SUCCESS in 11m 33s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/025dde1c60ae4d8e85c6edcfc382b203 : SUCCESS in 15m 47s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8fd9731e76d94872bebbeb4ddc1e59ee : SUCCESS in 11m 14s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24adf26f310e4f5898a03d7952f1d18f : SUCCESS in 13m 19s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e3b0c90a78f748032765cd2f1b22f46423b40b0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 12:36:02.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"d5d24ed49628e7a9fbef25cadc88c9cfb417a50c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 12:55:36.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6817f012abe54a89a7a2d245ec2f1f8e\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/459704cb86b94f77a9d568b9218512ff : SUCCESS in 2m 58s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1390d8ceced8477586ceefe73e1a0e0a : SUCCESS in 2m 55s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ec5f3914193401ca91c51ca59d441af : SUCCESS in 3m 04s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/414b2472a2874691b69bbf174ac1f09f : SUCCESS in 13m 37s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/23b9a277b7ea42cc850e998f6f04d4ef : SUCCESS in 18m 02s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59393b5f971c4d3795b5c442d9fe1cc5 : SUCCESS in 13m 40s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d2e6274ab88e45bdb4e5a9cd91a78970 : SUCCESS in 14m 09s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"48941d10b05085efc2e69983f7d3e012dab06805","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-10 03:29:18.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"792d5f24ddd2f9a985b43cf4e63a921a9b81aff8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-10 03:32:22.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"fa62128d7ea43a84bf9f4a8bec2b5257786a0af8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-10 03:51:12.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/05838cbc69304782bd83a646f220fb9e\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd243acfefca490db9968fc7102c9809 : SUCCESS in 2m 44s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc30a17ae7464d0aa2dcc58f6730104f : SUCCESS in 2m 41s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7a74c481c7e43778132b97df4a72668 : SUCCESS in 2m 35s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17d80ece03064734aac8e2e41a9d2a58 : SUCCESS in 11m 54s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/b58f11ab84c64a39b0aebcbacdcbfdc2 : FAILURE in 14m 26s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67f52132cffd4b91b53f6f1f94e82e42 : SUCCESS in 11m 46s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4127f85e1cb34a638aeb815c796eeb8e : SUCCESS in 11m 48s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f74fcb660600f102f1956916e3dad89289ceb9f8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-10 03:52:59.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"1c197c0d86d4df76057b94a3c6b91dc949e85116","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-10 04:13:36.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/945e230aeac045878ae9a2912ec83e88\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5374208821f7465eafe1245409e60736 : SUCCESS in 2m 54s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/608b71545e004c579077bef73fdc592a : SUCCESS in 2m 48s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a8cc7082816481abe9c02bc2402ffe3 : SUCCESS in 3m 10s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4845bbb18e14f5182cd1f01d147b951 : SUCCESS in 14m 44s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/dbc7f40e56ec454381bbdb0d90a2300d : SUCCESS in 19m 18s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27a3202fa9e849f7af1046c71b940d1b : SUCCESS in 11m 28s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f6cc7efacd840b594e224cbd5528ab6 : SUCCESS in 12m 03s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"d6b6f8dd1225d3474a2086f5953af619b4132394","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-07-11 21:52:40.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"65937c8e82a744d96d67fd9f18d7701fc7a933ea","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-07-15 19:15:46.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"588a6f32c9ebf4ff3d199efd24f4b1c5c08632b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 19:16:59.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"4a56e0b001807013aa49d4797f65046179ca60be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 20:22:01.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/add161f9195540f882e430f0036f76c4\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/410ea1c14e5a41b696d8e10358a676a4 : SUCCESS in 4m 02s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16ce4c9f807144d1b70e8fd49f6722f1 : SUCCESS in 5m 33s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72c2153d150048a69f112c65db7c4145 : SUCCESS in 4m 01s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/7d9b56149a33485ca4bd56e737d9ca26 : SUCCESS in 21m 23s","accounts_in_message":[],"_revision_number":9},{"id":"bd498a052007888397c507acc53185530301e16c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 20:22:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"ea262853af2f7ba8cbf7005924301df58c2407f4","revisions":{"2414243080169e410b9a2764315a982ca10676a7":{"kind":"REWORK","_number":1,"created":"2022-07-09 03:23:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/64/849164/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/64/849164/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/1"}}},"commit":{"parents":[{"commit":"c3744a4b3efd1e6f14e5d4bb5a58cad5188c7624","subject":"Merge \"bind9 backend: Clean up deprecated rndc_host\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/c3744a4b3efd1e6f14e5d4bb5a58cad5188c7624"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 03:23:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 03:23:28.000000000","tz":540},"subject":"Extend coverage of acceptance tests for config management resources","message":"Extend coverage of acceptance tests for config management resources\n\nChange-Id: Iccf1bcc19bb1e65fbccf61dc6f17429d6a3b88bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/2414243080169e410b9a2764315a982ca10676a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/2414243080169e410b9a2764315a982ca10676a7"}]},"branch":"refs/heads/master"},"e5a9dc14f7ba62cc35c5cd785d900e755e5d302f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-07-09 04:30:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/64/849164/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/64/849164/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/2"}}},"commit":{"parents":[{"commit":"007f12a913d8cec95d9db09dd9df6ec7ea5baa10","subject":"Switch provider to manage rootwrap.conf file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/007f12a913d8cec95d9db09dd9df6ec7ea5baa10"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 03:23:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 04:30:42.000000000","tz":540},"subject":"Extend coverage of acceptance tests for config management resources","message":"Extend coverage of acceptance tests for config management resources\n\nChange-Id: Iccf1bcc19bb1e65fbccf61dc6f17429d6a3b88bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/e5a9dc14f7ba62cc35c5cd785d900e755e5d302f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/e5a9dc14f7ba62cc35c5cd785d900e755e5d302f"}]},"branch":"refs/heads/master"},"15ead913787bd0159bd74049377b234505783823":{"kind":"REWORK","_number":3,"created":"2022-07-09 10:16:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/64/849164/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/64/849164/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/3"}}},"commit":{"parents":[{"commit":"007f12a913d8cec95d9db09dd9df6ec7ea5baa10","subject":"Switch provider to manage rootwrap.conf file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/007f12a913d8cec95d9db09dd9df6ec7ea5baa10"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 03:23:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 10:16:12.000000000","tz":540},"subject":"Extend coverage of acceptance tests for config management resources","message":"Extend coverage of acceptance tests for config management resources\n\nChange-Id: Iccf1bcc19bb1e65fbccf61dc6f17429d6a3b88bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/15ead913787bd0159bd74049377b234505783823"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/15ead913787bd0159bd74049377b234505783823"}]},"branch":"refs/heads/master"},"5494c28d045dd413cad48305e948714ce6731750":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-07-09 10:23:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/64/849164/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/64/849164/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/4"}}},"commit":{"parents":[{"commit":"007f12a913d8cec95d9db09dd9df6ec7ea5baa10","subject":"Switch provider to manage rootwrap.conf file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/007f12a913d8cec95d9db09dd9df6ec7ea5baa10"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 03:23:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 10:23:42.000000000","tz":0},"subject":"Extend coverage of config acceptance tests","message":"Extend coverage of config acceptance tests\n\nChange-Id: Iccf1bcc19bb1e65fbccf61dc6f17429d6a3b88bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/5494c28d045dd413cad48305e948714ce6731750"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/5494c28d045dd413cad48305e948714ce6731750"}]},"branch":"refs/heads/master","description":"Edit commit message"},"04bb65becf853c281fdae098bd123e7d59a01143":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-07-09 10:23:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/64/849164/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/64/849164/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/5"}}},"commit":{"parents":[{"commit":"c3744a4b3efd1e6f14e5d4bb5a58cad5188c7624","subject":"Merge \"bind9 backend: Clean up deprecated rndc_host\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/c3744a4b3efd1e6f14e5d4bb5a58cad5188c7624"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 03:23:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 10:23:56.000000000","tz":0},"subject":"Extend coverage of config acceptance tests","message":"Extend coverage of config acceptance tests\n\nChange-Id: Iccf1bcc19bb1e65fbccf61dc6f17429d6a3b88bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/04bb65becf853c281fdae098bd123e7d59a01143"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/04bb65becf853c281fdae098bd123e7d59a01143"}]},"branch":"refs/heads/master","description":"Rebase"},"2c8618be7a5c1454732f0377a1e5c0ce5347e5a6":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-07-09 10:24:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/64/849164/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/64/849164/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/6"}}},"commit":{"parents":[{"commit":"007f12a913d8cec95d9db09dd9df6ec7ea5baa10","subject":"Switch provider to manage rootwrap.conf file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/007f12a913d8cec95d9db09dd9df6ec7ea5baa10"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 03:23:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 10:24:22.000000000","tz":0},"subject":"Extend coverage of config acceptance tests","message":"Extend coverage of config acceptance tests\n\nChange-Id: Iccf1bcc19bb1e65fbccf61dc6f17429d6a3b88bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/2c8618be7a5c1454732f0377a1e5c0ce5347e5a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/2c8618be7a5c1454732f0377a1e5c0ce5347e5a6"}]},"branch":"refs/heads/master","description":"Rebase"},"49b380663a3f27b640281c626b014fdc5b55b0f5":{"kind":"REWORK","_number":7,"created":"2022-07-09 12:36:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/64/849164/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/64/849164/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/7"}}},"commit":{"parents":[{"commit":"007f12a913d8cec95d9db09dd9df6ec7ea5baa10","subject":"Switch provider to manage rootwrap.conf file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/007f12a913d8cec95d9db09dd9df6ec7ea5baa10"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 03:23:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 12:34:44.000000000","tz":540},"subject":"Extend coverage of config acceptance tests","message":"Extend coverage of config acceptance tests\n\nChange-Id: Iccf1bcc19bb1e65fbccf61dc6f17429d6a3b88bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/49b380663a3f27b640281c626b014fdc5b55b0f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/49b380663a3f27b640281c626b014fdc5b55b0f5"}]},"branch":"refs/heads/master"},"276f2268d68d3b089feef3a904122868cbeaa4d3":{"kind":"REWORK","_number":8,"created":"2022-07-10 03:29:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/64/849164/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/64/849164/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/8"}}},"commit":{"parents":[{"commit":"007f12a913d8cec95d9db09dd9df6ec7ea5baa10","subject":"Switch provider to manage rootwrap.conf file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/007f12a913d8cec95d9db09dd9df6ec7ea5baa10"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 03:23:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-10 03:28:45.000000000","tz":540},"subject":"Extend coverage of config acceptance tests","message":"Extend coverage of config acceptance tests\n\nChange-Id: Iccf1bcc19bb1e65fbccf61dc6f17429d6a3b88bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/276f2268d68d3b089feef3a904122868cbeaa4d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/276f2268d68d3b089feef3a904122868cbeaa4d3"}]},"branch":"refs/heads/master"},"ea262853af2f7ba8cbf7005924301df58c2407f4":{"kind":"REWORK","_number":9,"created":"2022-07-10 03:32:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/64/849164/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/64/849164/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/64/849164/9"}}},"commit":{"parents":[{"commit":"007f12a913d8cec95d9db09dd9df6ec7ea5baa10","subject":"Switch provider to manage rootwrap.conf file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/007f12a913d8cec95d9db09dd9df6ec7ea5baa10"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 03:23:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-10 03:31:59.000000000","tz":540},"subject":"Extend coverage of config acceptance tests","message":"Extend coverage of config acceptance tests\n\nChange-Id: Iccf1bcc19bb1e65fbccf61dc6f17429d6a3b88bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/ea262853af2f7ba8cbf7005924301df58c2407f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/ea262853af2f7ba8cbf7005924301df58c2407f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
