)]}'
{"id":"openstack%2Fpuppet-designate~854727","triplet_id":"openstack%2Fpuppet-designate~master~Idaddbca1acb1daafbf5381fc2e19f7db4748bbea","project":"openstack/puppet-designate","branch":"master","topic":"apache-logging","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-26 17:10:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idaddbca1acb1daafbf5381fc2e19f7db4748bbea","subject":"Add Apache WSGI logging parameters for pipe/syslog","status":"MERGED","created":"2022-08-26 05:51:44.000000000","updated":"2022-08-26 17:10:57.000000000","submitted":"2022-08-26 17:10:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854727-apache-logging","meta_rev_id":"b840a4f5747b07d401b73f49c945c9c19b3dd854","_number":854727,"virtual_id_number":854727,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-26 17:10:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-08-26 09:04:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-08-26 15:58:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-08-26 15:58:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-26 07:45:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-26 09:04:51.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-08-26 15:58:13.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"b08152b30745be15649a35c78a1d7263dce8719f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-26 05:51:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8026b79d8b94dc86939b4ca532420bfb89ac5c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-26 05:54:17.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"5455931d786e80c42c85fccdf5c3398a99cc9163","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-26 06:20:56.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent, and commit message as Patch Set 2.","accounts_in_message":[],"_revision_number":3},{"id":"2ada448289b2d092a3d200c3cedc50a70eebf29e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-26 06:21:22.000000000","message":"Topic apache-param-desc removed","accounts_in_message":[],"_revision_number":3},{"id":"36b03e43a88b5ae4a2dc2d57cba4a4e27f2b8c77","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-26 06:21:24.000000000","message":"Topic set to apache-logging","accounts_in_message":[],"_revision_number":3},{"id":"41eab5c941e66d6b6b7b5733bce248ac77161142","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 07:45:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80ebcd8131714c3881b4e5a7046d5a74\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3514808407d41be9882bb1f7c3e4d4f : SUCCESS in 3m 07s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59637b4ec40e4c629b47672dc0cb9e85 : SUCCESS in 2m 43s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a927c472fc00445b96efb00e9b0a7672 : SUCCESS in 3m 34s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c1da28f103d4de8882b181682af2d2c : SUCCESS in 13m 09s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0e64e41eac343b38a65defb5bfba7b8 : SUCCESS in 9m 00s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd3b93566acd4d98827be264b0981aec : SUCCESS in 7m 55s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a19e1260ef847b6a06aafbf3e1014f5 : SUCCESS in 8m 20s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/afbb52af432f419fa1664a55bcd399be : FAILURE in 49m 14s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c33e79eab12e430188cf82c9bf4b6405 : SUCCESS in 1h 17m 33s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33361fcacc6e49d987022124ed9cebac : SUCCESS in 14m 56s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/c6605ef7b2464ddda872424d9ce67e20 : SUCCESS in 20m 57s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66055e7f3fe34406990ef639ed599825 : SUCCESS in 14m 39s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/053ee8cb036a4279b09d5b779a07af59 : SUCCESS in 14m 22s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43c68e77b05c42d688822104be9aa27b : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":3},{"id":"20da4dd03972b313f690876c5344fd08b1d4e9fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-26 08:40:28.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d172e8a5314fce2fea1d63b9b6a388cfe4ed929b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-08-26 09:04:51.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"09e3e923bf3cbed2cd9abb45635c784a0efa3c5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 09:57:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cde3140b2d514623b0e33deb31ace0ef\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a242fb613a3546bea14e05d56b4b575e : SUCCESS in 3m 33s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b55c34441fa54084a991e0595569c714 : SUCCESS in 3m 35s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4fd95e4fb3c14dfa87446cad7bf53af2 : SUCCESS in 3m 33s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/b781ce2201524e1ca01706b4120f0a35 : SUCCESS in 16m 08s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30b753147e0f4a1b8f5ee1f444e67410 : SUCCESS in 10m 56s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00170dad89ad4e2cb8c57d4ac2c519cc : SUCCESS in 10m 35s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4788b7ddecf941f2bab85bf328b3bcf4 : SUCCESS in 10m 20s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbe7536ea3134087bc13092b47d7d6c5 : FAILURE in 43m 19s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e977aecf8f734378bf58327e96817714 : SUCCESS in 1h 10m 07s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9cc0a731ea8a48109ec7c53f2c800490 : SUCCESS in 16m 32s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/2de2d536818148e9b893d91d0cc7663b : SUCCESS in 30m 57s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/814647a8bde64fefa4cdd98f88e5a3b6 : SUCCESS in 15m 38s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18cc0c39f90b4580bf897b68486e3478 : SUCCESS in 24m 26s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/def635d4fdcb4ea2926ec3c9b7b0172b : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":4},{"id":"b2709ca77ebe3ce8999cc43470e825b5eff1d99f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-08-26 15:58:13.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"8a5105c32fc598ce89bd2fc0ddf98bf4c268c02a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 15:59:19.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d71cfb291e4573a7effad6cbffc722c630eb1797","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 17:10:03.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cab2aeb38f024d298a4c201903783355\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/839d4e3bcbe045f38f31ecda81636770 : SUCCESS in 4m 02s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/812bc15a892a42169103a93feab022d5 : SUCCESS in 3m 53s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1118f190904f4424828749c24290e8de : SUCCESS in 3m 59s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/910cfe5b59ff4e6ba1ba7cb369e5f3eb : SUCCESS in 17m 53s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8659667bfa44c55b5146807cbede787 : SUCCESS in 11m 58s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac3639feb74746c99ce8beba2b63ad8f : SUCCESS in 11m 34s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b22fd7a3f2394666b31e9e6a446b57b8 : SUCCESS in 1h 10m 05s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/12a4fee2a4c04c8b8afaa06eec04e515 : SUCCESS in 23m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66c90437690c4e7c85c7fb5206866903 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":4},{"id":"d2895591c166c127945ba38b6328b923551d5fad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 17:10:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"b840a4f5747b07d401b73f49c945c9c19b3dd854","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 17:10:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3ed4592ad7c436380879df0ad995c26\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4cb3bf431a447b2871022fabf12ed2c : SUCCESS in 35s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"59a16624b79d19bd294de88a5573f121e2775c96","revisions":{"4c74f68f59753513130024d495c0fa1be1956672":{"kind":"REWORK","_number":1,"created":"2022-08-26 05:51:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/854727/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/27/854727/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/27/854727/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/27/854727/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/27/854727/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/27/854727/1"}}},"commit":{"parents":[{"commit":"01e72cbb9a0ab7310f2ac4f7a10caf880f86a41f","subject":"Remove deprecated designate::neutron_network_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/01e72cbb9a0ab7310f2ac4f7a10caf880f86a41f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-26 05:50:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-26 05:51:27.000000000","tz":540},"subject":"Add Apache WSGI logging parameters for pipe/syslog","message":"Add Apache WSGI logging parameters for pipe/syslog\n\nAdd parameters for advanced logging configurations in Apache to\nsupport piped logging and support for syslog (via mod_syslog\navailable in Apache \u003e\u003d 2.5.0)\n\nCo-Authored-By: Andy Botting \u003candy@andybotting.com\u003e\nChange-Id: Idaddbca1acb1daafbf5381fc2e19f7db4748bbea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/4c74f68f59753513130024d495c0fa1be1956672"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/4c74f68f59753513130024d495c0fa1be1956672"}]},"branch":"refs/heads/master"},"5508eae2e7535b352709136dc7661dc40f9f10c8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-08-26 05:54:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/854727/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/27/854727/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/27/854727/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/27/854727/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/27/854727/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/27/854727/2"}}},"commit":{"parents":[{"commit":"4f2583ebe17ea6d9d099075ad271f7d16f3930e8","subject":"Clean up deprecated parameters of designate::worker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/4f2583ebe17ea6d9d099075ad271f7d16f3930e8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-26 05:50:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-26 05:54:17.000000000","tz":0},"subject":"Add Apache WSGI logging parameters for pipe/syslog","message":"Add Apache WSGI logging parameters for pipe/syslog\n\nAdd parameters for advanced logging configurations in Apache to\nsupport piped logging and support for syslog (via mod_syslog\navailable in Apache \u003e\u003d 2.5.0)\n\nCo-Authored-By: Andy Botting \u003candy@andybotting.com\u003e\nChange-Id: Idaddbca1acb1daafbf5381fc2e19f7db4748bbea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/5508eae2e7535b352709136dc7661dc40f9f10c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/5508eae2e7535b352709136dc7661dc40f9f10c8"}]},"branch":"refs/heads/master","description":"Rebase"},"5657e8c0035dcc66d25916af12c576d4dd426cc9":{"kind":"NO_CHANGE","_number":3,"created":"2022-08-26 06:20:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/854727/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/27/854727/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/27/854727/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/27/854727/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/27/854727/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/27/854727/3"}}},"commit":{"parents":[{"commit":"4f2583ebe17ea6d9d099075ad271f7d16f3930e8","subject":"Clean up deprecated parameters of designate::worker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/4f2583ebe17ea6d9d099075ad271f7d16f3930e8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-26 05:50:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-26 06:20:42.000000000","tz":540},"subject":"Add Apache WSGI logging parameters for pipe/syslog","message":"Add Apache WSGI logging parameters for pipe/syslog\n\nAdd parameters for advanced logging configurations in Apache to\nsupport piped logging and support for syslog (via mod_syslog\navailable in Apache \u003e\u003d 2.5.0)\n\nCo-Authored-By: Andy Botting \u003candy@andybotting.com\u003e\nChange-Id: Idaddbca1acb1daafbf5381fc2e19f7db4748bbea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/5657e8c0035dcc66d25916af12c576d4dd426cc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/5657e8c0035dcc66d25916af12c576d4dd426cc9"}]},"branch":"refs/heads/master"},"59a16624b79d19bd294de88a5573f121e2775c96":{"kind":"REWORK","_number":4,"created":"2022-08-26 08:40:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/854727/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/27/854727/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/27/854727/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/27/854727/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/27/854727/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/27/854727/4"}}},"commit":{"parents":[{"commit":"4f2583ebe17ea6d9d099075ad271f7d16f3930e8","subject":"Clean up deprecated parameters of designate::worker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/4f2583ebe17ea6d9d099075ad271f7d16f3930e8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-26 05:50:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-26 08:40:14.000000000","tz":540},"subject":"Add Apache WSGI logging parameters for pipe/syslog","message":"Add Apache WSGI logging parameters for pipe/syslog\n\nAdd parameters for advanced logging configurations in Apache to\nsupport piped logging and support for syslog (via mod_syslog\navailable in Apache \u003e\u003d 2.5.0)\n\nCo-Authored-By: Andy Botting \u003candy@andybotting.com\u003e\nChange-Id: Idaddbca1acb1daafbf5381fc2e19f7db4748bbea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/59a16624b79d19bd294de88a5573f121e2775c96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/59a16624b79d19bd294de88a5573f121e2775c96"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
