)]}'
{"id":"openstack%2Fpuppet-designate~854914","triplet_id":"openstack%2Fpuppet-designate~master~I566dca39f9d0d9b85f33a30a593f08000f705baf","project":"openstack/puppet-designate","branch":"master","topic":"default-opts","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-29 16:32:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I566dca39f9d0d9b85f33a30a593f08000f705baf","subject":"Add support for [DEFAULT] supported_record_type","status":"MERGED","created":"2022-08-28 14:13:23.000000000","updated":"2022-08-29 16:32:56.000000000","submitted":"2022-08-29 16:32:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854914-default-opts","meta_rev_id":"b6eaa51226e359dafc18434688404fac1d0681a2","_number":854914,"virtual_id_number":854914,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-29 16:32:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-08-29 06:38:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-08-29 15:08:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-08-29 15:08:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-28 15:50:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-29 06:38:57.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-08-29 15:08:45.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"a9e24c4beacec22e063d3c285562577f8d211c0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-28 14:13:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a7ef8708f45c303f873c1ecebeee904cca7c477","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-28 14:24:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"aa371f2bc9ae6d979d1c134e9f095f9b92ec7267","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-28 14:35:51.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"c4aadc9d10dbecc600ef845221c854590e8edaa1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-28 15:50:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83fc8ec464464626bea570b185c67377\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b457695d03b4b7d94d44f436da62568 : SUCCESS in 4m 03s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/967c84ece2994638b3f48ae8b2634857 : SUCCESS in 3m 26s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc8253ea7a894cb9900d86a259a8ba81 : SUCCESS in 3m 47s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/859370f8bbb74304b37ed528e43493e5 : SUCCESS in 22m 59s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2fc26d1151b4ce2ba1cb63c0ea8790c : SUCCESS in 10m 31s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47a4c184757e49718c8a3367df4067d5 : SUCCESS in 11m 58s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4de0546eae5445c28b541a4348590692 : SUCCESS in 10m 17s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/840b48c314944dba857131d7667e0a5d : FAILURE in 10m 09s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6d25863022f1410096f3f6419e27e1e2 : SUCCESS in 1h 13m 04s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2fb21b405796437a827b716e90928e16 : FAILURE in 10m 21s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/170f96a384b94cd999805f398598f74b : SUCCESS in 19m 52s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7489c3f6360c401c89100c16c8dfda0d : FAILURE in 17m 59s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/326ddcf0a12f4c10ae6001630986edc7 : FAILURE in 16m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ccb926d8146434099f7e6ba548d0524 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":3},{"id":"3130e2f2eac4fd8ec794d746d9c8838457c255ac","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-08-29 06:38:57.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"fe8ba3672a6b44952ca58c764d5d3adfdbb65148","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-08-29 15:08:45.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"436582451016d336654076423e5bcb9faa40dad3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 15:10:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"04cd01f29f65b9a374e2d758722d2c6cdee1c874","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 16:32:56.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/308b8b5f0866423085339da4c8f7be04\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53251a8bc6554f4b8c2c19eb1a8ca427 : SUCCESS in 3m 28s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb852b4f8c4a4f718b0a20ac69b64c38 : SUCCESS in 4m 07s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5779d4b362143a399fa04b17fd134fb : SUCCESS in 3m 52s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/76b1dca9ecc84dac9d38cf62b311fba8 : SUCCESS in 16m 58s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e574556697f547c0809c2a199d261944 : SUCCESS in 12m 08s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e725bab4a1b4498b991fa69340965b5 : SUCCESS in 11m 59s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e71f6ea81ac94398be3d9db072000ae1 : SUCCESS in 1h 12m 48s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/5421e3235b3d4822af91f15c5aea5949 : SUCCESS in 27m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e01338f79d1a4d66a5e7d231ffb274b2 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":3},{"id":"b6eaa51226e359dafc18434688404fac1d0681a2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 16:32:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"86766620b8481b243076e664b45de2da093eed6e","revisions":{"520c3e8bbf976f8b15abc6d89fd20853ec06832a":{"kind":"REWORK","_number":1,"created":"2022-08-28 14:13:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/854914/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/14/854914/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/14/854914/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/14/854914/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/14/854914/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/14/854914/1"}}},"commit":{"parents":[{"commit":"b2300dd6000e5bd7b546389ec362c3457af24267","subject":"Add support for [DEFAULT] default_ttl","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/b2300dd6000e5bd7b546389ec362c3457af24267"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-28 14:10:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-28 14:13:12.000000000","tz":540},"subject":"Add support for [DEFAULT] supported_record_type","message":"Add support for [DEFAULT] supported_record_type\n\nChange-Id: I566dca39f9d0d9b85f33a30a593f08000f705baf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/520c3e8bbf976f8b15abc6d89fd20853ec06832a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/520c3e8bbf976f8b15abc6d89fd20853ec06832a"}]},"branch":"refs/heads/master"},"6348bcea9e6852061b710f12681b4609172be8a8":{"kind":"REWORK","_number":2,"created":"2022-08-28 14:24:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/854914/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/14/854914/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/14/854914/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/14/854914/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/14/854914/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/14/854914/2"}}},"commit":{"parents":[{"commit":"65286ef786d1422751fb0c0a69cbeefcc728d4a1","subject":"Add support for [DEFAULT] default_ttl","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/65286ef786d1422751fb0c0a69cbeefcc728d4a1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-28 14:10:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-28 14:24:33.000000000","tz":540},"subject":"Add support for [DEFAULT] supported_record_type","message":"Add support for [DEFAULT] supported_record_type\n\nChange-Id: I566dca39f9d0d9b85f33a30a593f08000f705baf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/6348bcea9e6852061b710f12681b4609172be8a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/6348bcea9e6852061b710f12681b4609172be8a8"}]},"branch":"refs/heads/master"},"86766620b8481b243076e664b45de2da093eed6e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-08-28 14:35:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/854914/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/14/854914/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/14/854914/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/14/854914/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/14/854914/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/14/854914/3"}}},"commit":{"parents":[{"commit":"3d68d191e209a198c25feadd5c597d55021c572c","subject":"Add support for [DEFAULT] default_ttl","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/3d68d191e209a198c25feadd5c597d55021c572c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-28 14:10:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-28 14:35:42.000000000","tz":540},"subject":"Add support for [DEFAULT] supported_record_type","message":"Add support for [DEFAULT] supported_record_type\n\nChange-Id: I566dca39f9d0d9b85f33a30a593f08000f705baf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/86766620b8481b243076e664b45de2da093eed6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/86766620b8481b243076e664b45de2da093eed6e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
