)]}'
{"id":"openstack%2Fpuppet-designate~854915","triplet_id":"openstack%2Fpuppet-designate~master~I909e40692aaa2b7a23a0efe8ff5a0392e8fe7293","project":"openstack/puppet-designate","branch":"master","topic":"default-opts","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-29 16:32:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I909e40692aaa2b7a23a0efe8ff5a0392e8fe7293","subject":"Add support for [DEFAULT] host","status":"MERGED","created":"2022-08-28 14:24:41.000000000","updated":"2022-08-29 16:34:06.000000000","submitted":"2022-08-29 16:32:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854915-default-opts","meta_rev_id":"0fb0b39165a6d01565324f67875f7aa842425090","_number":854915,"virtual_id_number":854915,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-29 16:32:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-08-29 06:38:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-08-29 15:08:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-08-29 15:08:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-28 15:29:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-29 06:38:41.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-08-29 15:08:34.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"c197a2b106b701622018d074b7dc92bfc0295c5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-28 14:24:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"67ae23aaf66ba2a794d6f6d8b6de7ec95e5efee6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-28 14:35:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e05a6715ab27c5fbebc535548adeaf4e09d94cac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-28 15:29:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3cbc01f880440e1919a75171db49b5d\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ca3a1c8a92b4f0d8128a43aebc7dd30 : SUCCESS in 4m 01s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62b4ef909cdb44d99fd8f1f3fc365e14 : SUCCESS in 3m 02s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc8912e878394b24a51750dfb4f096df : SUCCESS in 3m 53s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/a233cc2850274592a4a5eb88926c9b26 : SUCCESS in 15m 54s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90e28830c18e41818da3cc1d3617b110 : SUCCESS in 10m 06s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ebb924343d94c65abf71c67e0f68d71 : SUCCESS in 10m 12s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c86d5998e824a8697f0ebf611f2af8e : SUCCESS in 9m 41s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57365a4d2ad442ddad203b047ab406dc : FAILURE in 9m 55s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ab14ea3045a48ef923117a6423b934c : SUCCESS in 51m 56s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1eed1dcebfe44f358f56fd87f6b18f48 : FAILURE in 9m 25s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/834c54fdb53c4ceba3e75c6323ee270c : SUCCESS in 19m 01s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d72bc7c0c6e241c697124777b5c9a350 : FAILURE in 14m 51s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e73866db503846618e55bbb4f505c2b3 : FAILURE in 15m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a2c32e524304ccd932b643328719958 : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":2},{"id":"c2744cc8fc1996e3983287fed3d766db4237938b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-08-29 06:38:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c3fddb8fd3ff1fab432953102743a729a8ef6259","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-08-29 15:08:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8968df028aa61f04ff540c11f6d4ea180263e920","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 15:10:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"882fb669f0dd891239472fe90c35b5c9c318b290","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 16:32:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8475f63bcb364961ad7c00502b3d2710\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20447c1e4dc5466192d9009910be1c6d : SUCCESS in 3m 51s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8074dc587b2046fd9de9bbf0f2375f5f : SUCCESS in 3m 28s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d85a81a33794072b58d20cc384ed1f5 : SUCCESS in 3m 54s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/5aa0edcc255c4111b95bc4b0146d2d38 : SUCCESS in 12m 46s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e51343bd14074ec7afe5007b4a896c3b : SUCCESS in 12m 17s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83af4335cd0d4baa95b6318dfeba5072 : SUCCESS in 8m 23s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1db4e3623a9e49929a326a53e6951e01 : SUCCESS in 1h 17m 11s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/ef61c2788b414825afd5e757ac889466 : SUCCESS in 23m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b6b44ab2e634d51a8b006a6ef193071 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":2},{"id":"b7d648144d6f344e3066e8767b71923bdb7c81c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 16:32:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0fb0b39165a6d01565324f67875f7aa842425090","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 16:34:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce10ee5926b24c10a7df70e12c31c505\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43e9bc6b8bed4e98a0c0a7a20665cc8d : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"54545aacb64d7419f2e2d3c2b756eab29a050046","revisions":{"85f196bcf0ed898c4fc98c7e8e222cea9c07aa54":{"kind":"REWORK","_number":1,"created":"2022-08-28 14:24:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/854915/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/15/854915/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/15/854915/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/15/854915/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/15/854915/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/15/854915/1"}}},"commit":{"parents":[{"commit":"01e72cbb9a0ab7310f2ac4f7a10caf880f86a41f","subject":"Remove deprecated designate::neutron_network_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/01e72cbb9a0ab7310f2ac4f7a10caf880f86a41f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-28 14:17:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-28 14:21:30.000000000","tz":540},"subject":"Add support for [DEFAULT] host","message":"Add support for [DEFAULT] host\n\n... so that users can override this parameter to use static hostname\nrather than the one deteted by socket.gethostname().\n\nChange-Id: I909e40692aaa2b7a23a0efe8ff5a0392e8fe7293\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/85f196bcf0ed898c4fc98c7e8e222cea9c07aa54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/85f196bcf0ed898c4fc98c7e8e222cea9c07aa54"}]},"branch":"refs/heads/master"},"54545aacb64d7419f2e2d3c2b756eab29a050046":{"kind":"REWORK","_number":2,"created":"2022-08-28 14:35:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/854915/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/15/854915/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/15/854915/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/15/854915/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/15/854915/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/15/854915/2"}}},"commit":{"parents":[{"commit":"01e72cbb9a0ab7310f2ac4f7a10caf880f86a41f","subject":"Remove deprecated designate::neutron_network_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/01e72cbb9a0ab7310f2ac4f7a10caf880f86a41f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-28 14:17:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-28 14:35:38.000000000","tz":540},"subject":"Add support for [DEFAULT] host","message":"Add support for [DEFAULT] host\n\n... so that users can override this parameter to use static hostname\nrather than the one deteted by socket.gethostname().\n\nChange-Id: I909e40692aaa2b7a23a0efe8ff5a0392e8fe7293\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/54545aacb64d7419f2e2d3c2b756eab29a050046"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/54545aacb64d7419f2e2d3c2b756eab29a050046"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
