)]}'
{"id":"openstack%2Fpuppet-designate~870734","triplet_id":"openstack%2Fpuppet-designate~master~I9b4e492fe55b2cec34b422e4ca456fd38a99260f","project":"openstack/puppet-designate","branch":"master","hashtags":[],"change_id":"I9b4e492fe55b2cec34b422e4ca456fd38a99260f","subject":"Deprecate support for agent backend","status":"ABANDONED","created":"2023-01-17 09:23:52.000000000","updated":"2023-03-03 10:31:45.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4876c11aa2ab0ec473ed06b8a51668811f8d3b6a","_number":870734,"virtual_id_number":870734,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-01-17 10:33:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-2,"date":"2023-01-17 09:24:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-17 10:33:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"435824135e7f95b610356abffc95265fc9289714","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-17 09:23:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"938a6fa250a660a07a531fd8e00615706a5cc61d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-17 09:24:36.000000000","message":"Patch Set 1: Code-Review-2","accounts_in_message":[],"_revision_number":1},{"id":"1d0faa2e6210da18eaf67aa8b221015ea00cf84c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 10:33:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb3d9abb5d2c42339e3217a879f24bc3\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ebb561f382eb48948be90c68c7706f58 : SUCCESS in 2m 40s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb7851fe26f04ab09edddff2aaaefdc3 : SUCCESS in 3m 24s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/3965349b2fda40e4b1b82621cf602b4e : SUCCESS in 10m 06s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/01431950e6d040da9a20f8490a8ee1b1 : SUCCESS in 7m 38s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c00affc271648f5afcf4ad27866e136 : SUCCESS in 9m 22s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d3df48c1484424ba3ff128ee7f8133b : FAILURE in 34m 55s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b029fab5badc45b989151e8985f1aeee : SUCCESS in 1h 04m 04s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/ba58d683a712480984349967078528f9 : SUCCESS in 20m 30s\n- puppet-openstack-litmus-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0bae1c7c59a046868639a99f9fb309d7 : FAILURE in 8m 54s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/54bea1c89a65403291c6dafeade62097 : FAILURE in 6m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c9f724d0ff146919932695303fe6ede : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":1},{"id":"4876c11aa2ab0ec473ed06b8a51668811f8d3b6a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-03 10:31:45.000000000","message":"Abandoned\n\nduplicate of https://review.opendev.org/c/openstack/puppet-designate/+/874962","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dc9a9c2a38324545d3618a820355f61c0e45c7e5","revisions":{"dc9a9c2a38324545d3618a820355f61c0e45c7e5":{"kind":"REWORK","_number":1,"created":"2023-01-17 09:23:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/870734/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/34/870734/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/34/870734/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/34/870734/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/34/870734/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/34/870734/1"}}},"commit":{"parents":[{"commit":"8fd38799c3f76c7b7b0a9dce458ed9d32876201c","subject":"Switch to Ubuntu Jammy (22.04)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/8fd38799c3f76c7b7b0a9dce458ed9d32876201c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-17 09:22:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-17 09:23:44.000000000","tz":540},"subject":"Deprecate support for agent backend","message":"Deprecate support for agent backend\n\nThe agent backend in designate is being deprecated. This deprecates\nthe manifests to deploy the service, so that we can remove these when\nthe actual implementation is removed from designate.\n\nChange-Id: I9b4e492fe55b2cec34b422e4ca456fd38a99260f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/dc9a9c2a38324545d3618a820355f61c0e45c7e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/dc9a9c2a38324545d3618a820355f61c0e45c7e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
