)]}'
{"id":"openstack%2Fpuppet-designate~875588","triplet_id":"openstack%2Fpuppet-designate~master~I703a2a64ce15ea6adf7cf233e635baf3ad963d31","project":"openstack/puppet-designate","branch":"master","topic":"validate-boolean","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-03-06 16:33:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I703a2a64ce15ea6adf7cf233e635baf3ad963d31","subject":"Add strict validation about boolean parameters","status":"MERGED","created":"2023-02-28 04:17:06.000000000","updated":"2023-03-06 16:33:36.000000000","submitted":"2023-03-06 16:33:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875588-validate-boolean","meta_rev_id":"f80df882ecec94e494f13f539b53045dad93adb9","_number":875588,"virtual_id_number":875588,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-06 16:33:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2023-03-03 08:20:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-03-06 15:21:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2023-03-06 07:13:48.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-03-06 15:21:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-28 05:45:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-28 05:55:23.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-03-01 17:49:11.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"2483523ca83f5e90d1aa37bf5805848d26546276","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-28 04:17:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58b2ee290592decd914edfab6dc1187333fc3b37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 05:45:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bdbc022af6f4a94a40703cda2bfb3c4\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f037f0dbed1461ca9f26483ed7577f3 : SUCCESS in 4m 27s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/71721dce4ac144a99603c6038976180e : SUCCESS in 2m 47s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/dde734e767e7445d85ee1c722251f969 : SUCCESS in 22m 36s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cbce09d154a946b1bb6aaf18e94e7989 : SUCCESS in 7m 43s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9062e53922d40d6be16a5b1e3f98913 : SUCCESS in 8m 16s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c928bb1b51e4ea2b46714906820ba8a : FAILURE in 27m 41s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/54793081ef1f44e18f83de72c0ad9293 : SUCCESS in 1h 10m 37s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/4370545e296c4509948aaaba9557929f : SUCCESS in 21m 57s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5097b190cf94031811eab6761a6e3a7 : FAILURE in 11m 55s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/21f4063454b4446bb3b07ef7bb7d7cf3 : FAILURE in 12m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c45071880fdc3eb346210efd11c264fcee5c869b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-02-28 05:55:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"194f86258cf87e9368dab6f7f45c15b11b0a81d1","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-03-01 17:49:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1bfa3d8868931b2fa675ddd0cf1663a2c0ce0fc1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 17:50:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b1e9f9461171d6bfc8389ff710c4753c72c66601","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 19:02:17.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a203ed507b74d1a90d1315e3de8d3ca\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e0a60805634494fbe7b9ec6b0a840fd : FAILURE in 2m 34s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e7fc228504a4f9884a9c7d3eaa48cdc : SUCCESS in 2m 26s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/b886be743ab841db9d5663a143de7b68 : SUCCESS in 11m 28s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/082e520e458248429a32af9649feeec6 : SUCCESS in 7m 34s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0f8056103641428fa755811c31b423b7 : SUCCESS in 1h 07m 26s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/a962299de5b84af2846ab4aed57f3a19 : SUCCESS in 18m 51s","accounts_in_message":[],"_revision_number":1},{"id":"766d62c77aff84c862e80017874647caadf27be2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-03 04:22:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a6545c9d14c31eebd85a171931acde81f53c47cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-03 04:32:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1dafb3532a7d09e359791e01b30e0260630bee43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-03 04:37:27.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"c6de09ee46fd273d081500628dca3f8487015f2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 05:48:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ac69617a1a247338e70fa825c9d386a\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa7e98c5a6cd493788509de9267d7ea8 : SUCCESS in 2m 47s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b1492f7ff14548dc85236b248c535008 : SUCCESS in 2m 43s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a9a3a987f3a43489a142a95e6149651 : SUCCESS in 9m 51s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/549d72b3e4474fb38e1b012d199edd96 : SUCCESS in 7m 51s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc23624060ad4e8290961cfc0591dfda : SUCCESS in 6m 14s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/26a9d8b30f094817a1aa50f66d401390 : FAILURE in 26m 05s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ad0c639dcbdc48d0ab55d017309b9f7b : SUCCESS in 1h 03m 29s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/e5f2c7b75b634f7c98939d3e07c2188b : SUCCESS in 21m 54s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d993a9da0a641b38beaded9988ba3f9 : FAILURE in 15m 48s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/abd653f8c02241a19b7335fb6376f1e4 : FAILURE in 11m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cb1323243a86f19f162cf88d63beffbbcc0d09a3","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-03-03 08:20:49.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"feca63bc5194ddad5b6075115b8b4f893336df6f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-03-06 07:13:13.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"ca43cf6a6a67e735f0d6e33d470df53468820c53","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-03-06 07:13:26.000000000","message":"Patch Set 4: -Workflow","accounts_in_message":[],"_revision_number":4},{"id":"80463a8c9821c41034ab04819909af138f9caeea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 07:13:38.000000000","message":"Patch Set 4: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":4},{"id":"16c73850b218cf2a75d5001ed5146f0c94242cac","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-03-06 07:13:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1c1245fb4986e97c32d489276f742aabf88a64d0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-06 07:27:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ec5c7a1f31c960292f1f3024576d33214f3e69ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 08:38:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a71e562c0e374918a607b23f5c8657ff\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/72905a02ee644a4ea88f89526101ac46 : SUCCESS in 4m 50s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c674dc456c9c438896883fe096612a25 : SUCCESS in 2m 38s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/3da0f31061d04e30a642c3e0ac6c1618 : SUCCESS in 12m 09s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/678816e93dd84f04bf59017ef9dc1a0d : SUCCESS in 7m 34s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/58dd07250d704a278c128079ba23b305 : SUCCESS in 6m 35s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e5caad1450574163b595443bc36a5138 : FAILURE in 1h 00m 06s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/eac2576af43b4595b42c6bd16a75fa38 : SUCCESS in 1h 05m 22s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/c2a1974db159485db82b5e2408231d2b : SUCCESS in 19m 49s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25edaa2f7f5241beaa89d51209639d8f : FAILURE in 12m 05s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c82e382e004d43c59c8d0624eaf1a8cd : FAILURE in 12m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"056d9d2d71b42ec7927bea5bf429c4f9f4feb7ad","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-03-06 15:21:27.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"221ef8f036807030e749e819ca12007d2d14024d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 15:22:19.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"83b1170a8110ea823ef60755014cfa011f9025ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 16:33:35.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b9844d4f5724fbfbb81530c99fe6413\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a99c0e6b5b7148c59f801ecf8161c1b2 : SUCCESS in 3m 35s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/acbba9a02e0646d2befa85c118b6cc75 : SUCCESS in 2m 51s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/2cd945d6d02f481d8667c17cd378a68f : SUCCESS in 11m 19s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e06a7d8d95fc49df9fb3130241460a42 : SUCCESS in 10m 13s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c7b604c6f3ab48abb323f13ad3bfbde0 : SUCCESS in 1h 05m 55s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/14f7829b34714138bc03ecc18816ba0a : SUCCESS in 29m 24s","accounts_in_message":[],"_revision_number":4},{"id":"f80df882ecec94e494f13f539b53045dad93adb9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 16:33:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5e89485f5d853c62902a4d9c7703ec300165b4d4","revisions":{"da3e6d881da1e38ff73197fe2ff14611479c9d4d":{"kind":"REWORK","_number":1,"created":"2023-02-28 04:17:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/88/875588/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/88/875588/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/88/875588/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/88/875588/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/88/875588/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/88/875588/1"}}},"commit":{"parents":[{"commit":"bf0a7eca6131d14a8c83cecc709b3d2ee2d66730","subject":"Deprecate support for designate-agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/bf0a7eca6131d14a8c83cecc709b3d2ee2d66730"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-28 04:16:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-28 04:16:48.000000000","tz":540},"subject":"Add strict validation about boolean parameters","message":"Add strict validation about boolean parameters\n\nThis ensures the parameters used by if-else logic accept only boolean\nvalues because non-boolean can result in unexpected behavior.\n\nChange-Id: I703a2a64ce15ea6adf7cf233e635baf3ad963d31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/da3e6d881da1e38ff73197fe2ff14611479c9d4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/da3e6d881da1e38ff73197fe2ff14611479c9d4d"}]},"branch":"refs/heads/master"},"af0871b959f86c3f99719e793ebef335abf4359d":{"kind":"REWORK","_number":2,"created":"2023-03-03 04:22:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/88/875588/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/88/875588/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/88/875588/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/88/875588/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/88/875588/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/88/875588/2"}}},"commit":{"parents":[{"commit":"bf0a7eca6131d14a8c83cecc709b3d2ee2d66730","subject":"Deprecate support for designate-agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/bf0a7eca6131d14a8c83cecc709b3d2ee2d66730"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-28 04:16:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-03 04:22:24.000000000","tz":540},"subject":"Add strict validation about boolean parameters","message":"Add strict validation about boolean parameters\n\nThis ensures the parameters used by if-else logic accept only boolean\nvalues because non-boolean can result in unexpected behavior.\n\nChange-Id: I703a2a64ce15ea6adf7cf233e635baf3ad963d31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/af0871b959f86c3f99719e793ebef335abf4359d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/af0871b959f86c3f99719e793ebef335abf4359d"}]},"branch":"refs/heads/master"},"b717277a6b4042f8b381d91afee5f4c792c3917f":{"kind":"REWORK","_number":3,"created":"2023-03-03 04:32:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/88/875588/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/88/875588/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/88/875588/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/88/875588/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/88/875588/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/88/875588/3"}}},"commit":{"parents":[{"commit":"bf0a7eca6131d14a8c83cecc709b3d2ee2d66730","subject":"Deprecate support for designate-agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/bf0a7eca6131d14a8c83cecc709b3d2ee2d66730"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-28 04:16:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-03 04:31:54.000000000","tz":540},"subject":"Add strict validation about boolean parameters","message":"Add strict validation about boolean parameters\n\nThis ensures the parameters used by if-else logic accept only boolean\nvalues because non-boolean can result in unexpected behavior.\n\nChange-Id: I703a2a64ce15ea6adf7cf233e635baf3ad963d31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/b717277a6b4042f8b381d91afee5f4c792c3917f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/b717277a6b4042f8b381d91afee5f4c792c3917f"}]},"branch":"refs/heads/master"},"5e89485f5d853c62902a4d9c7703ec300165b4d4":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-03-03 04:37:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/88/875588/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/88/875588/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/88/875588/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/88/875588/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/88/875588/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/88/875588/4"}}},"commit":{"parents":[{"commit":"dd44ce031e9f92a7999bc8296b4f90657a3ca7c7","subject":"Replace legacy facts and use fact hash","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/dd44ce031e9f92a7999bc8296b4f90657a3ca7c7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-28 04:16:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-03 04:37:27.000000000","tz":0},"subject":"Add strict validation about boolean parameters","message":"Add strict validation about boolean parameters\n\nThis ensures the parameters used by if-else logic accept only boolean\nvalues because non-boolean can result in unexpected behavior.\n\nChange-Id: I703a2a64ce15ea6adf7cf233e635baf3ad963d31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/5e89485f5d853c62902a4d9c7703ec300165b4d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/5e89485f5d853c62902a4d9c7703ec300165b4d4"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
