)]}'
{"id":"openstack%2Fpuppet-designate~886354","triplet_id":"openstack%2Fpuppet-designate~master~I85d7058bf379c6aa0e787b2f806f14aebf0693d3","project":"openstack/puppet-designate","branch":"master","topic":"replace-validate_legacy","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-06-20 00:06:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I85d7058bf379c6aa0e787b2f806f14aebf0693d3","subject":"replace validate_legacy with proper data types","status":"MERGED","created":"2023-06-19 05:00:53.000000000","updated":"2023-06-20 00:06:33.000000000","submitted":"2023-06-20 00:06:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886354-replace-validate_legacy","meta_rev_id":"999bf324b2b4a3162400911fd167e10a47cab332","_number":886354,"virtual_id_number":886354,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-20 00:06:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2023-06-19 19:42:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2023-06-19 23:06:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2023-06-19 23:06:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-19 06:02:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-19 19:42:30.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-06-19 23:06:59.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"9163f15941468cde3a2a31ba2233f36c78b822df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-19 05:00:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e64cf73b10665420a63dd37c4b99d12697c9c500","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-19 05:09:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a7bfd665e182e6d5145c1bccae3f4d23eb1716bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 06:02:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a63a1a18b42498db0d80a551a1b4750\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6e9a77ccaa541789357f411ca4fb869 : SUCCESS in 3m 21s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c245ed19df8f47d8a59b7f3bdb1455d2 : SUCCESS in 2m 42s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/c29e52476a594bbc8415413c6407f8ad : SUCCESS in 5m 39s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/f2b5cbd5309b485c9f63afbfab07044a : SUCCESS in 9m 32s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b334d3e3023f47c5885f05fa06ad7f89 : SUCCESS in 6m 48s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/688766429970469798d9406eeb259af3 : SUCCESS in 10m 34s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b12598a9b974ae1b63e303d8bf4a4c0 : SUCCESS in 6m 51s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/75853bcd2d554ba2bcf5a855d879f4e1 : SUCCESS in 40m 22s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/169d30cd1dd741f69b9019c6bd885066 : SUCCESS in 51m 35s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/9053619bb4e74f8baf0d1a6958341d41 : SUCCESS in 20m 03s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c56d997ae8547f7ab21747a76c99f4f : SUCCESS in 11m 26s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/340ec2bdd7b840629080dc544834aa89 : SUCCESS in 14m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ff66b7e6f01a2168c9b6eda689d2e51c49bef7fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-19 08:27:43.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"ab2f13d67c5793db770e28bbfefc48342b1dae53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 09:26:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/521283ac09d44bc69cabf3095bf7ef24\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5790ab4c765741ddbe88b444f0741572 : SUCCESS in 2m 38s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4e4491368f846fa9da19912a33b3c60 : SUCCESS in 2m 41s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/6432261a6e5140249ecc802b40bf5147 : SUCCESS in 4m 51s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab742052405544198cdc45214bd7151a : SUCCESS in 9m 22s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b04bfc02b6104724969f11d08c28daf7 : SUCCESS in 6m 49s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/b2823e5aa4d44e7aad876fd87d55ce16 : SUCCESS in 11m 08s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c443b0f81d054393b33d4483634a26cc : SUCCESS in 6m 52s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c12541301c44ae0a1372f1703573100 : SUCCESS in 42m 12s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/312805c11a80435b9a07461325b66679 : SUCCESS in 54m 07s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/136963e5598b427985868055070718f4 : SUCCESS in 16m 58s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/136c01dac1c9481983975ea8852066ff : SUCCESS in 13m 18s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b763d43704b4455a74cf2c5c2c8a463 : SUCCESS in 14m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"68851d2661ab2f1fe5419fb2f0b5858573576ab2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-06-19 19:42:30.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"53e6ef9d43f9a452995d10d855efd1d1b7e7b159","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-06-19 23:06:59.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b0c05c8344f0990e09fab0690e2a05d26e8114b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 23:07:51.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f4d6b03404d4fe643d6f23e35fbd8b173d079a39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 00:06:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9eccd9457ec04d7787fa5296a3fabe04\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86ec2d2c655f48a781bdd5927099d68b : SUCCESS in 2m 30s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14438c3ae0a6445c924abac46788afcb : SUCCESS in 2m 22s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e1c871262f246a98a4222caf01a296a : SUCCESS in 12m 00s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ed9ffbccd59040498071bfd8380c975c : SUCCESS in 6m 46s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/5d57c4daaea94602bb31f91539ef6177 : SUCCESS in 11m 33s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b45cbd92d746492eab5f61a6bd813344 : SUCCESS in 57m 42s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/6d3b0fdbaf3d4c6897cb733733fa4b02 : SUCCESS in 17m 36s","accounts_in_message":[],"_revision_number":3},{"id":"999bf324b2b4a3162400911fd167e10a47cab332","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 00:06:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8ed0bcb86753d2e3730b293d98375781e5a3959e","revisions":{"d72bdbb9480c935fb23d6c90b75d56979e2c34b3":{"kind":"REWORK","_number":1,"created":"2023-06-19 05:00:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/54/886354/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/54/886354/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/54/886354/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/54/886354/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/54/886354/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/54/886354/1"}}},"commit":{"parents":[{"commit":"aaa8affdeec65b79df15106836479c11ca0c4506","subject":"Remove support for Puppet 6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/aaa8affdeec65b79df15106836479c11ca0c4506"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-19 05:00:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-19 05:00:11.000000000","tz":540},"subject":"replace validate_legacy with proper data types","message":"replace validate_legacy with proper data types\n\nthe validate_legacy function is marked for deprecation in\nv9.0.0 from puppetlabs-stdlib.\n\nChange-Id: I85d7058bf379c6aa0e787b2f806f14aebf0693d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/d72bdbb9480c935fb23d6c90b75d56979e2c34b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/d72bdbb9480c935fb23d6c90b75d56979e2c34b3"}]},"branch":"refs/heads/master"},"79d5fac1aab5c1c7f839a277f933e006d3e71984":{"kind":"REWORK","_number":2,"created":"2023-06-19 05:09:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/54/886354/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/54/886354/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/54/886354/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/54/886354/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/54/886354/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/54/886354/2"}}},"commit":{"parents":[{"commit":"8f4986cea4a2ce6e8d235312c261d0c18794a3fa","subject":"Deprecate implementations for pool-manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/8f4986cea4a2ce6e8d235312c261d0c18794a3fa"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-19 05:00:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-19 05:08:00.000000000","tz":540},"subject":"replace validate_legacy with proper data types","message":"replace validate_legacy with proper data types\n\nthe validate_legacy function is marked for deprecation in\nv9.0.0 from puppetlabs-stdlib.\n\nNote that this change looses validation about pool ids. Alternatively\nwe can define a new data type to keep the validation, but these defined\ntypes for pool-manager were already deprecated and useless and will be\nremoved in a near future release so we avoid that redundant step.\n\nChange-Id: I85d7058bf379c6aa0e787b2f806f14aebf0693d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/79d5fac1aab5c1c7f839a277f933e006d3e71984"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/79d5fac1aab5c1c7f839a277f933e006d3e71984"}]},"branch":"refs/heads/master"},"8ed0bcb86753d2e3730b293d98375781e5a3959e":{"kind":"REWORK","_number":3,"created":"2023-06-19 08:27:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/54/886354/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/54/886354/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/54/886354/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/54/886354/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/54/886354/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/54/886354/3"}}},"commit":{"parents":[{"commit":"8f4986cea4a2ce6e8d235312c261d0c18794a3fa","subject":"Deprecate implementations for pool-manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/8f4986cea4a2ce6e8d235312c261d0c18794a3fa"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-19 05:00:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-19 08:27:32.000000000","tz":540},"subject":"replace validate_legacy with proper data types","message":"replace validate_legacy with proper data types\n\nthe validate_legacy function is marked for deprecation in\nv9.0.0 from puppetlabs-stdlib.\n\nNote that this change looses validation about pool ids. Alternatively\nwe can define a new data type to keep the validation, but these defined\ntypes for pool-manager were already deprecated and useless and will be\nremoved in a near future release so we avoid that redundant step.\n\nChange-Id: I85d7058bf379c6aa0e787b2f806f14aebf0693d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/8ed0bcb86753d2e3730b293d98375781e5a3959e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/8ed0bcb86753d2e3730b293d98375781e5a3959e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
