)]}'
{"id":"openstack%2Fpuppet-designate~930435","triplet_id":"openstack%2Fpuppet-designate~master~I4d8e37f5f6e02df7f405c555b6f1c8e0c01265f3","project":"openstack/puppet-designate","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-09-25 19:58:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4d8e37f5f6e02df7f405c555b6f1c8e0c01265f3","subject":"Add support for [oslo_messaging_rabbit] enable_cancel_on_failover","status":"MERGED","created":"2024-09-25 10:17:27.000000000","updated":"2024-09-25 19:59:55.000000000","submitted":"2024-09-25 19:58:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930435","meta_rev_id":"dbb67473d5cc6e39423b8ebc0496c94cea59e59c","_number":930435,"virtual_id_number":930435,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-25 19:58:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2024-09-25 12:58:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2024-09-25 18:53:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2024-09-25 18:53:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-25 11:32:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-25 12:58:00.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-09-25 18:53:44.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"cad4e95f2b92bdeb03cf3fc63323dfef5b6d8568","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-25 10:17:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"35d2af4c32989cfce7b802b09b79e90833fd017f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-25 10:26:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cd42b97ea6838403aea1097b86352157474eea78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-25 10:31:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0a5c21a3d8a923bf417b68f31b522784df78db1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 11:32:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/32cc40b4799e4cc38d679cfc8d2a29f0\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70b5fbbe4ea346bb99f94a3cbd5eac64 : SUCCESS in 2m 47s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48bf1a5eeb9a4651b76a94f7aa215404 : SUCCESS in 3m 11s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/259ebb1df6a2401988b8facb03b45558 : SUCCESS in 5m 30s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/d2e9559ec6c74d3b8f9818cb450b9bd9 : FAILURE in 8m 03s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7adc692e04e45b49accfb1dfc6287fc : FAILURE in 5m 01s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/a4d647e53b9f4efda7be3d0f21d1f3a0 : FAILURE in 8m 18s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a249bd2d7a04cecb2c1e71859f4121a : FAILURE in 5m 12s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e4ddc70eea24b8dad4e0759744dc41b : SUCCESS in 41m 32s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6ad0e462883c4cab81ce17988cc6220c : SUCCESS in 51m 47s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/5bb9334ba2fe494e8e38f3165fee28f9 : SUCCESS in 18m 26s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca69012d13b54e5bb8b0e05ec1f225a8 : SUCCESS in 7m 13s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8aab5d0d9d6541e4bdc0d617e9552f01 : SUCCESS in 10m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23efcf14bc85495eab21822f58d5e749 : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":3},{"id":"7ce6c55f19843606f466ffb30c125acdd54da147","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-25 11:50:27.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"bbd43516079a541d91662a2c7a7a618d4aad79e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 12:49:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce8b925e485446a79fe66ff364a6d02a\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1d85d8dd7a24f63871fd1df4dca5fde : SUCCESS in 2m 00s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4efde8fc25745d8bafba01447cf89e2 : SUCCESS in 2m 27s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c02ed1b7eb6476c92af26bc1a491563 : SUCCESS in 5m 03s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/5e926f6109f8493eaf6851ce5b82a1df : SUCCESS in 8m 58s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34bf884de4ce48b9a8492776b6241040 : SUCCESS in 5m 22s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/bbf2b578ab1741f391d093dd069177ce : SUCCESS in 6m 56s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99b417f3f75e4235a1307f7a4179e31e : SUCCESS in 5m 33s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ba0ed01ef43499eb3d0ff01f2818f6c : SUCCESS in 43m 56s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c882e41cfb2a43e8ac11b85138fbfe29 : SUCCESS in 52m 31s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/3c30c047dea4405489617a9903c7931d : SUCCESS in 18m 45s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd7787848aa54ece87e2c2ead09d79a8 : SUCCESS in 9m 35s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cffa8d2f9b484d1780d73e25e31464df : SUCCESS in 11m 22s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ff791a8c774456c8f21169aa3b392a9 : SUCCESS in 2m 05s","accounts_in_message":[],"_revision_number":4},{"id":"6cbc96ccbee764ee0bc12624e1f1a2690a65ebcc","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-09-25 12:58:00.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fd4ea38c71df4ef6a48dee2991b4eeb0576b4874","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-09-25 18:53:44.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d319481102694033e6f1e971c2686915471eb622","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 18:54:33.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9f42df11231a0ea1432c7debbf193311df4ddfee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 19:58:16.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4233a09eb6e49ee902b056a62a4f862\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6958c671dea4f32ab8ec006e56a778d : SUCCESS in 2m 13s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a60b3d9e2e764f2f9812b7eb6ca77098 : SUCCESS in 2m 49s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/0557738b82c243bc9c1f2b6f7c658152 : SUCCESS in 5m 46s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1c337dc7cb545a59114720b77b391ad : SUCCESS in 4m 25s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bef547ca70fb45738d04049d3d40e041 : SUCCESS in 45m 09s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/01f4f98c72514e08b9fe9337550b0c15 : SUCCESS in 9m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a67bcb2faedd4191ac9948999dae25b4 : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":4},{"id":"c9669ce8b8ca8b35b28c5a842050ff3e8baf8b57","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 19:58:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"dbb67473d5cc6e39423b8ebc0496c94cea59e59c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 19:59:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f38a7c5036c4faba19d06b9d461cef4\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52129160d164436c9077554d887ab2a4 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"90bf5ed359f3648a557038f2502f80af927e15c8","revisions":{"45752f3f3b5f90bc2f9744b475c3c5799b7b425b":{"kind":"REWORK","_number":1,"created":"2024-09-25 10:17:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/930435/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/35/930435/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/35/930435/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/35/930435/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/35/930435/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/35/930435/1"}}},"commit":{"parents":[{"commit":"413ac43c4f9cf6f639f8690dc32f91c307777c01","subject":"logging: Deprecate watch_log_file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/413ac43c4f9cf6f639f8690dc32f91c307777c01"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-25 10:16:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-25 10:16:31.000000000","tz":540},"subject":"Add support for [oslo_messaging_rabbit] enable_cancel_on_failover","message":"Add support for [oslo_messaging_rabbit] enable_cancel_on_failover\n\nDepends-on: https://review.opendev.org/928440\nChange-Id: I4d8e37f5f6e02df7f405c555b6f1c8e0c01265f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/45752f3f3b5f90bc2f9744b475c3c5799b7b425b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/45752f3f3b5f90bc2f9744b475c3c5799b7b425b"}]},"branch":"refs/heads/master"},"d516b29c505db81257a23bb0ae9ec3a18d101b57":{"kind":"REWORK","_number":2,"created":"2024-09-25 10:26:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/930435/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/35/930435/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/35/930435/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/35/930435/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/35/930435/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/35/930435/2"}}},"commit":{"parents":[{"commit":"413ac43c4f9cf6f639f8690dc32f91c307777c01","subject":"logging: Deprecate watch_log_file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/413ac43c4f9cf6f639f8690dc32f91c307777c01"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-25 10:16:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-25 10:25:57.000000000","tz":540},"subject":"Add support for [oslo_messaging_rabbit] enable_cancel_on_failover","message":"Add support for [oslo_messaging_rabbit] enable_cancel_on_failover\n\nDepends-on: https://review.opendev.org/928440\nChange-Id: I4d8e37f5f6e02df7f405c555b6f1c8e0c01265f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/d516b29c505db81257a23bb0ae9ec3a18d101b57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/d516b29c505db81257a23bb0ae9ec3a18d101b57"}]},"branch":"refs/heads/master"},"6d5cfae813c13e13dc695fa05d624b1993873eed":{"kind":"REWORK","_number":3,"created":"2024-09-25 10:31:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/930435/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/35/930435/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/35/930435/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/35/930435/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/35/930435/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/35/930435/3"}}},"commit":{"parents":[{"commit":"6b8d8aa77d2eec49b92fe58b92fe8cd1608ccb67","subject":"Expose tunable options for RabbitMQ heartbeat","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/6b8d8aa77d2eec49b92fe58b92fe8cd1608ccb67"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-25 10:16:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-25 10:31:16.000000000","tz":540},"subject":"Add support for [oslo_messaging_rabbit] enable_cancel_on_failover","message":"Add support for [oslo_messaging_rabbit] enable_cancel_on_failover\n\nDepends-on: https://review.opendev.org/928440\nChange-Id: I4d8e37f5f6e02df7f405c555b6f1c8e0c01265f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/6d5cfae813c13e13dc695fa05d624b1993873eed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/6d5cfae813c13e13dc695fa05d624b1993873eed"}]},"branch":"refs/heads/master"},"90bf5ed359f3648a557038f2502f80af927e15c8":{"kind":"REWORK","_number":4,"created":"2024-09-25 11:50:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/930435/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/35/930435/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/35/930435/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/35/930435/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/35/930435/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/35/930435/4"}}},"commit":{"parents":[{"commit":"5fdb3a7a2432277ccd4dfe2b3c90cbd2226f591b","subject":"Expose tunable options for RabbitMQ heartbeat","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/5fdb3a7a2432277ccd4dfe2b3c90cbd2226f591b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-25 10:16:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-25 11:50:15.000000000","tz":540},"subject":"Add support for [oslo_messaging_rabbit] enable_cancel_on_failover","message":"Add support for [oslo_messaging_rabbit] enable_cancel_on_failover\n\nDepends-on: https://review.opendev.org/928440\nChange-Id: I4d8e37f5f6e02df7f405c555b6f1c8e0c01265f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/90bf5ed359f3648a557038f2502f80af927e15c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/90bf5ed359f3648a557038f2502f80af927e15c8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
