)]}'
{"id":"openstack%2Fpuppet-designate~957926","triplet_id":"openstack%2Fpuppet-designate~master~I9705ad334d05a2041f0f9c4b885fc3fd44342720","project":"openstack/puppet-designate","branch":"master","topic":"lint-plugins","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-08-23 02:40:40.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I9705ad334d05a2041f0f9c4b885fc3fd44342720","subject":"Prepare for voxpupuli-puppet-lint-plugins","status":"MERGED","created":"2025-08-19 14:12:18.000000000","updated":"2025-08-23 04:50:58.000000000","submitted":"2025-08-23 04:50:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957926-lint-plugins","meta_rev_id":"43e4f8c1d1eda8056e2dc8bd91a27fbb07a41452","_number":957926,"virtual_id_number":957926,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-23 04:50:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2025-08-20 06:37:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2025-08-22 14:03:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2025-08-22 14:03:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-19 14:54:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-20 06:37:51.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-08-22 14:03:17.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"12cad7dfe469815595ac672d21efd018b79e42be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-19 14:12:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"78fdfebbd005532722e0de08f6d7b779bd8d255b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-19 14:12:52.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"7e23ff7b3f5ae6a326d1e86594de35df6ca56373","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 14:54:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f613729eed0492e98ca23d453644556\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5bb58487383a44da95e39a8c65176266 : FAILURE in 3m 46s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6953cbd0df1a4248aa403d9745658c9e : SUCCESS in 1m 50s","accounts_in_message":[],"_revision_number":1},{"id":"f4e5223a2ad7d54f792f0962fd88af8371a89d43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-19 14:59:20.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"109deea5cb54aeba6f94f533a782dfda9b1f7f8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 15:07:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4bb8fc4e0c444d0ba6e29f3ffffc285f\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76a9f3dc827e495bbd421850b740231e : FAILURE in 3m 47s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/feb00a5ab8e949e0901ff30806470767 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":2},{"id":"91aa2304e066a50e92b1b723810e9aae00302c6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-19 15:11:55.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"8c705f01505c537860157b5b11c166413f5172dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-19 15:39:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"dc7e7b89487d6418247d6b1ce64f1458348aa046","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 19:14:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f4e6ffc58e64d1c8d7bfded34ac3a58\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3d9a0d62e2d4f08b28661296fa31ec6 : SUCCESS in 3m 01s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4718f5e90774996949960fff7d498d4 : SUCCESS in 5m 13s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/b1e2d9d6310d4704b5a89f1a460a9358 : SUCCESS in 8m 51s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0b0f1c886ad4f1cac0a2d2549c64ccf : SUCCESS in 5m 11s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f46baef0981e4a2bbadfc64ef68bf833 : SUCCESS in 4m 23s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/122f6ecd08c14f26a10b7bc716bd08ba : SUCCESS in 48m 51s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/213fbb696cdb4cd6871a9da793a67226 : SUCCESS in 57m 49s\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/46a43843902a43ad8220f8941fc1430f : SUCCESS in 27m 27s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/218768e98d954913abad241ebf1c0f58 : SUCCESS in 16m 14s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53f15b8452864ca1b72a2aadff8d4d4e : SUCCESS in 14m 03s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3fbc958398804888aa9254f8df4d951c : SUCCESS in 7m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0e552da0254e284999940a595435cd63edf0a871","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-08-20 06:37:51.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"d6ad7d3f1988faa96da693f59f2f04a111bbb09c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-08-22 14:03:17.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"064290b9d41de3c5a49c7554b8e87dc0b7ee4f4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 14:04:26.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f795f48a740dee1a4324f8e8c4e3e2bd938bf6c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 14:24:06.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c31a6808bd34e89a876d79bfc4472d9\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94dcd0a45dcf46069dbbb530efab29ba : SUCCESS in 2m 43s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/268511a793ed407b8f6b7e57f1829e2d : SUCCESS in 3m 06s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/add4ea223fad46aeba08c53f6369fa35 : SUCCESS in 9m 58s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/58118b8c3ef643cf89f296adeaf9f244 : SUCCESS in 6m 05s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/516333bc458241f094abf33d2983d9a4 : FAILURE in 14m 55s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/f3b63bba550048ec83d1720b117e7256 : FAILURE in 11m 22s","accounts_in_message":[],"_revision_number":4},{"id":"cc62b4ff6029c76c012ae3d9bcce2ae7e6e32f13","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-23 02:40:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cf1463bb2b01f479417e12f90809b00002affb31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-23 03:48:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/369eee985e16447a871571375e6e2d33\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4fb97c958e964c929cb303809b08b40a : SUCCESS in 2m 55s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b83260cee6949649ad96f67a46a2c9f : SUCCESS in 3m 08s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/0f11ae2e8f494a708e9bfcb79e555887 : SUCCESS in 4m 58s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d7681fb914d42d3a420230c5acb4d5a : SUCCESS in 7m 51s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bbbcf3c3ccf94a36bba57040042ab220 : SUCCESS in 6m 34s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/498141d4ca7f48bb9a53f4d8a28c1acd : FAILURE in 32m 30s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/22fcf4be1b3944ae8669193092c8573d : SUCCESS in 1h 03m 15s\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/225b8bb48fdd4342ae039909d8c6aa27 : SUCCESS in 1h 04m 52s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/109d0a00633842b785284421d99d6e59 : SUCCESS in 17m 30s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dab442da22844990a952feaed771f1b2 : SUCCESS in 12m 10s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/95a9ecd95eab46dab8020ab5271a1950 : SUCCESS in 14m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ee225ed1c70e8b4292e7effd4404dea45234e67d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-23 03:49:15.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e29a6ac7937284b3eb033e0d5783e22fb5696972","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-23 04:50:58.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb6fdf3ff46d4f00a7f1d32258e737ac\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/04a4b7d35dbb406cac07b6e6730a5e3f : SUCCESS in 3m 44s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ffcb07fd76644ec86c71cf09e1755ca : SUCCESS in 2m 41s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/f0a4d908bcc7469f9c7992f7e38b1295 : SUCCESS in 8m 46s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bef1ace91b494b1e97f4fe010cf28d15 : SUCCESS in 3m 37s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/386627a9fb614c8f879ccb5fbf455df3 : SUCCESS in 57m 58s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e02ebb7c8fd40da8d9955cf3754820c : SUCCESS in 17m 08s","accounts_in_message":[],"_revision_number":4},{"id":"43e4f8c1d1eda8056e2dc8bd91a27fbb07a41452","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-23 04:50:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"8f560f1de1c117a6ffcf12be9f4051c9743836b4","revisions":{"6d640b19958113c7ced2a29d283965fd2024609c":{"kind":"REWORK","_number":1,"created":"2025-08-19 14:12:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/957926/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/26/957926/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/26/957926/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/26/957926/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/26/957926/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/26/957926/1"}}},"commit":{"parents":[{"commit":"0d497d714ed6322167786edfef9bd3eed87e7cf6","subject":"Merge \"Remove puppet 7 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/0d497d714ed6322167786edfef9bd3eed87e7cf6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-19 14:11:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-19 14:12:11.000000000","tz":540},"subject":"Prepare for voxpupuli-puppet-lint-plugins","message":"Prepare for voxpupuli-puppet-lint-plugins\n\nDepends-on: https://review.opendev.org/957884\nChange-Id: I9705ad334d05a2041f0f9c4b885fc3fd44342720\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/6d640b19958113c7ced2a29d283965fd2024609c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/6d640b19958113c7ced2a29d283965fd2024609c"}]},"branch":"refs/heads/master"},"f17cfc4710b0d5e82c027d9acaa3156ac4a993b3":{"kind":"REWORK","_number":2,"created":"2025-08-19 14:59:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/957926/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/26/957926/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/26/957926/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/26/957926/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/26/957926/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/26/957926/2"}}},"commit":{"parents":[{"commit":"0d497d714ed6322167786edfef9bd3eed87e7cf6","subject":"Merge \"Remove puppet 7 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/0d497d714ed6322167786edfef9bd3eed87e7cf6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-19 14:11:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-19 14:59:10.000000000","tz":540},"subject":"wip: Prepare for voxpupuli-puppet-lint-plugins","message":"wip: Prepare for voxpupuli-puppet-lint-plugins\n\nDepends-on: https://review.opendev.org/957884\nChange-Id: I9705ad334d05a2041f0f9c4b885fc3fd44342720\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/f17cfc4710b0d5e82c027d9acaa3156ac4a993b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/f17cfc4710b0d5e82c027d9acaa3156ac4a993b3"}]},"branch":"refs/heads/master"},"e775b596f8531bbe43cc5550d3a5da02114ecb6c":{"kind":"REWORK","_number":3,"created":"2025-08-19 15:11:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/957926/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/26/957926/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/26/957926/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/26/957926/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/26/957926/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/26/957926/3"}}},"commit":{"parents":[{"commit":"0d497d714ed6322167786edfef9bd3eed87e7cf6","subject":"Merge \"Remove puppet 7 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/0d497d714ed6322167786edfef9bd3eed87e7cf6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-19 14:11:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-19 15:11:47.000000000","tz":540},"subject":"Prepare for voxpupuli-puppet-lint-plugins","message":"Prepare for voxpupuli-puppet-lint-plugins\n\nFix new lint errors detected when full of the voxpupili lint plugins\nare enabled.\n\nChange-Id: I9705ad334d05a2041f0f9c4b885fc3fd44342720\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/e775b596f8531bbe43cc5550d3a5da02114ecb6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/e775b596f8531bbe43cc5550d3a5da02114ecb6c"}]},"branch":"refs/heads/master"},"8f560f1de1c117a6ffcf12be9f4051c9743836b4":{"kind":"REWORK","_number":4,"created":"2025-08-19 15:39:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/957926/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-designate","ref":"refs/changes/26/957926/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/26/957926/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/26/957926/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-designate refs/changes/26/957926/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-designate refs/changes/26/957926/4"}}},"commit":{"parents":[{"commit":"0d497d714ed6322167786edfef9bd3eed87e7cf6","subject":"Merge \"Remove puppet 7 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/0d497d714ed6322167786edfef9bd3eed87e7cf6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-19 14:11:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-19 15:39:05.000000000","tz":540},"subject":"Prepare for voxpupuli-puppet-lint-plugins","message":"Prepare for voxpupuli-puppet-lint-plugins\n\nFix new lint errors detected when full of the voxpupili lint plugins\nare enabled.\n\nChange-Id: I9705ad334d05a2041f0f9c4b885fc3fd44342720\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/8f560f1de1c117a6ffcf12be9f4051c9743836b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-designate/commit/8f560f1de1c117a6ffcf12be9f4051c9743836b4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
