)]}'
{"id":"openstack%2Fpuppet-freezer~654202","triplet_id":"openstack%2Fpuppet-freezer~master~I25b362b66754df39b558eb21c26ce69d5a2cde56","project":"openstack/puppet-freezer","branch":"master","topic":"move-to-opendev","hashtags":[],"change_id":"I25b362b66754df39b558eb21c26ce69d5a2cde56","subject":"Move to opendev","status":"MERGED","created":"2019-04-21 06:45:56.000000000","updated":"2019-04-22 17:18:54.000000000","submitted":"2019-04-22 17:18:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654202-1555953534093-dfe5b5ea","meta_rev_id":"f7ac598dfa97703d647cafd713627f85212a2428","_number":654202,"virtual_id_number":654202,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-04-22 17:18:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2019-04-22 13:26:54.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-04-22 16:55:35.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2019-04-22 16:55:35.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 13:26:54.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-04-22 16:55:35.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-04-22 17:18:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f23cf6bc92dbbb659cf98f3e585ac45048e842cd","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-21 06:45:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce78d5ec2761abca5e3e32c01212061c94c28670","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 07:24:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/02/654202/1/check/puppet-openstack-lint-ubuntu-bionic/c82b430/ : SUCCESS in 5m 20s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/02/654202/1/check/puppet-openstack-syntax-4-ubuntu-bionic/32d9994/ : SUCCESS in 4m 53s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/02/654202/1/check/puppet-openstack-syntax-5-ubuntu-bionic/00c1559/ : SUCCESS in 4m 14s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/02/654202/1/check/puppet-openstack-syntax-6-ubuntu-bionic/0805ed9/ : SUCCESS in 4m 53s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/02/654202/1/check/puppet-openstack-unit-4.8-centos-7/fb29482/ : SUCCESS in 16m 05s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/02/654202/1/check/puppet-openstack-unit-5.5-centos-7/21523be/ : SUCCESS in 13m 17s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/02/654202/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/c043e20/ : SUCCESS in 7m 01s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/02/654202/1/check/puppet-openstack-unit-latest-ubuntu-bionic/1bb4ed6/ : SUCCESS in 6m 51s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/02/654202/1/check/puppet-openstack-beaker-centos-7/dcdcf17/ : SUCCESS in 8m 41s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/02/654202/1/check/puppet-openstack-beaker-ubuntu-bionic/b8e982d/ : SUCCESS in 5m 09s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/02/654202/1/check/build-openstack-releasenotes/dd4490a/html/ : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":1},{"id":"9918209ebb02f52e5e824b5e95bdaef1681a006f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-04-22 13:26:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"21b73580777ef35d5f4ca887b4a3820d2e731dc7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-04-22 16:55:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"be90565e1db7722da68bdf1fb0debff4eefb4101","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 16:56:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a372b67ac8fb147d63ad131ef0877cc1a00d787f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 17:18:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/02/654202/1/gate/puppet-openstack-lint-ubuntu-bionic/6b4af61/ : SUCCESS in 6m 05s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/02/654202/1/gate/puppet-openstack-syntax-4-ubuntu-bionic/e00926e/ : SUCCESS in 4m 19s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/02/654202/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/b4f866a/ : SUCCESS in 4m 08s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/02/654202/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/19eb894/ : SUCCESS in 4m 17s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/02/654202/1/gate/puppet-openstack-unit-4.8-centos-7/ccccd72/ : SUCCESS in 13m 53s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/02/654202/1/gate/puppet-openstack-unit-5.5-centos-7/89f8544/ : SUCCESS in 9m 57s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/02/654202/1/gate/puppet-openstack-unit-6.0-ubuntu-bionic/857beb9/ : SUCCESS in 10m 03s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/02/654202/1/gate/puppet-openstack-beaker-centos-7/b3bc860/ : SUCCESS in 7m 47s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/02/654202/1/gate/puppet-openstack-beaker-ubuntu-bionic/6ae5f12/ : SUCCESS in 8m 07s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/02/654202/1/gate/build-openstack-releasenotes/1d16cfd/html/ : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":1},{"id":"5f328454bdd738c2a371f2b7907b62a17d6ece00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 17:18:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ee7ada90fbd6d586c7a96d9fd15cadc2f8dbc8c4","revisions":{"ee7ada90fbd6d586c7a96d9fd15cadc2f8dbc8c4":{"kind":"REWORK","_number":1,"created":"2019-04-21 06:45:56.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/02/654202/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-freezer","ref":"refs/changes/02/654202/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-freezer refs/changes/02/654202/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-freezer refs/changes/02/654202/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-freezer refs/changes/02/654202/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-freezer refs/changes/02/654202/1"}}},"commit":{"parents":[{"commit":"7fb7855d0f368b1dec4d0a68f74082b3802b10c1","subject":"Fix Repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-freezer/commit/7fb7855d0f368b1dec4d0a68f74082b3802b10c1"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-21 06:44:48.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-21 06:44:48.000000000","tz":480},"subject":"Move to opendev","message":"Move to opendev\n\nChange-Id: I25b362b66754df39b558eb21c26ce69d5a2cde56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-freezer/commit/ee7ada90fbd6d586c7a96d9fd15cadc2f8dbc8c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-freezer/commit/ee7ada90fbd6d586c7a96d9fd15cadc2f8dbc8c4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
