)]}'
{"id":"openstack%2Fpuppet-freezer~665786","triplet_id":"openstack%2Fpuppet-freezer~master~I3f783632d280cdcdfd84bda1b849796e844d13db","project":"openstack/puppet-freezer","branch":"master","topic":"665786","hashtags":[],"change_id":"I3f783632d280cdcdfd84bda1b849796e844d13db","subject":"assert_private() params and deps classes","status":"ABANDONED","created":"2019-06-17 23:16:32.000000000","updated":"2020-09-18 22:56:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f60fac66675d0eee4cfd55337ee3e2bd3b0dc8ad","_number":665786,"virtual_id_number":665786,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-06-18 00:02:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-18 00:02:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3acc6be7c2f45c184e4960f0a341bc64c3163fc9","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-06-17 23:16:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a719626e5518f4a6c817f81bd329f4fedf9bd431","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-06-17 23:44:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9c734479ccd7057ee844a817a83adba95b9da89e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 00:02:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/86/665786/2/check/puppet-openstack-lint-ubuntu-bionic/74c6d04/ : SUCCESS in 4m 09s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/86/665786/2/check/puppet-openstack-syntax-5-ubuntu-bionic/48954ec/ : SUCCESS in 4m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/86/665786/2/check/puppet-openstack-syntax-6-ubuntu-bionic/b2b92df/ : SUCCESS in 4m 00s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/86/665786/2/check/puppet-openstack-unit-5.5-centos-7/732c856/ : SUCCESS in 10m 04s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/86/665786/2/check/puppet-openstack-unit-6.4-ubuntu-bionic/008ef3a/ : SUCCESS in 7m 13s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/86/665786/2/check/puppet-openstack-unit-latest-ubuntu-bionic/2da57e4/ : SUCCESS in 6m 49s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/86/665786/2/check/puppet-openstack-beaker-centos-7/7193c87/ : SUCCESS in 9m 06s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/86/665786/2/check/puppet-openstack-beaker-ubuntu-bionic/93579ba/ : SUCCESS in 8m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3381be82b2a40c2089d7b7d13b88271a026f5f33","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-18 22:56:00.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ef5914d781406540aa408c0453d34333c0efbf04","revisions":{"ac9871f08bf6c3ee7039e55cc2ee7197f0c7e582":{"kind":"REWORK","_number":1,"created":"2019-06-17 23:16:32.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/86/665786/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-freezer","ref":"refs/changes/86/665786/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-freezer refs/changes/86/665786/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-freezer refs/changes/86/665786/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-freezer refs/changes/86/665786/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-freezer refs/changes/86/665786/1"}}},"commit":{"parents":[{"commit":"a63081763672ebfe8f1feb44d398465bc04ab798","subject":"Prepare Train M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-freezer/commit/a63081763672ebfe8f1feb44d398465bc04ab798"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:16:23.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:16:23.000000000","tz":120},"subject":"assert_private() params and deps classes","message":"assert_private() params and deps classes\n\nThese classes are private and should never be\nused outside of it\u0027s module.\n\nChange-Id: I3f783632d280cdcdfd84bda1b849796e844d13db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-freezer/commit/ac9871f08bf6c3ee7039e55cc2ee7197f0c7e582"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-freezer/commit/ac9871f08bf6c3ee7039e55cc2ee7197f0c7e582"}]},"branch":"refs/heads/master"},"ef5914d781406540aa408c0453d34333c0efbf04":{"kind":"REWORK","_number":2,"created":"2019-06-17 23:44:20.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/86/665786/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-freezer","ref":"refs/changes/86/665786/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-freezer refs/changes/86/665786/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-freezer refs/changes/86/665786/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-freezer refs/changes/86/665786/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-freezer refs/changes/86/665786/2"}}},"commit":{"parents":[{"commit":"a63081763672ebfe8f1feb44d398465bc04ab798","subject":"Prepare Train M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-freezer/commit/a63081763672ebfe8f1feb44d398465bc04ab798"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:16:23.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:44:13.000000000","tz":120},"subject":"assert_private() params and deps classes","message":"assert_private() params and deps classes\n\nThese classes are private and should never be\nused outside of it\u0027s module.\n\nChange-Id: I3f783632d280cdcdfd84bda1b849796e844d13db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-freezer/commit/ef5914d781406540aa408c0453d34333c0efbf04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-freezer/commit/ef5914d781406540aa408c0453d34333c0efbf04"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
