)]}'
{"id":"openstack%2Fpuppet-freezer~684896","triplet_id":"openstack%2Fpuppet-freezer~master~I2453b6fd0182e8bdeb533f44f4f017d7392f29a0","project":"openstack/puppet-freezer","branch":"master","topic":"constraints","hashtags":[],"change_id":"I2453b6fd0182e8bdeb533f44f4f017d7392f29a0","subject":"Update the constraints url","status":"MERGED","created":"2019-09-26 03:06:04.000000000","updated":"2019-09-27 02:17:11.000000000","submitted":"2019-09-27 02:16:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684896-1569550574676-d890bd4f","meta_rev_id":"fac325872e98bce62e9dbd817acf34c4e0186a89","_number":684896,"virtual_id_number":684896,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-27 02:16:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-09-27 01:47:28.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-09-27 01:47:28.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-27 01:47:28.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-09-27 02:16:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"84450e58d5e29b20c432ed7610ee24dcd52c406a","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-09-26 03:06:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"85a70297db0e82aa4ad109b5bb35130d90f29781","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 03:23:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a8fa2ff9544e4c439a9df82020d013fd : SUCCESS in 5m 23s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b0a4c43989f4de38af50116db82ae03 : SUCCESS in 4m 18s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a8986fadcc24864a5909e812ba532e4 : SUCCESS in 4m 33s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/2f27e8448a3a4dad8bb85cb6a456043f : SUCCESS in 10m 23s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/645fe8af360d4354845e38df39efd20b : SUCCESS in 9m 58s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/965dd6588e854f89990a50de51e3840f : SUCCESS in 10m 23s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/5fdbc2c4d24f4bdcb73f94e8a7c14c9e : SUCCESS in 8m 50s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a4e0da7050b49a8818ed9f801afce14 : SUCCESS in 7m 08s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b409e1460e004a94b81a694043b58f7b : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":1},{"id":"19c3457a70772566b2b10043853d1485db17cc37","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-27 01:47:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3b7ab68cc831d10f6a83bd5e699501cfe7fa46f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 01:47:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"508c3c3a27618e5f97e6077d87f60801875626a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 02:16:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/82146ffeecc6407eb5784d47ecb59148 : SUCCESS in 5m 40s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ee3f87f825e74c8f9053438ee51be753 : SUCCESS in 4m 44s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c46170bb1664c599e5ae11abff972ab : SUCCESS in 4m 52s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/320a1e05a2124258bfc3886508ea4e31 : SUCCESS in 11m 14s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1cd3fc4917f2406f87d3481fca7525ee : SUCCESS in 11m 52s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/3189fe340f7c4abd99dc8075c4df8035 : SUCCESS in 10m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97d40447fdca45a7b48ebec504bd7dc1 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":1},{"id":"5c6bc3cf1374400640b90b34c7f20463af1fddd3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 02:16:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2a205caa4bfbd32384a3dae7b2a0dbece3fa8538","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 02:17:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba9819024c0e437e9ff813a19ad4a13c : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0f3ee47f75b91578099bf4a1ae9ccb0ad5c6df34","revisions":{"0f3ee47f75b91578099bf4a1ae9ccb0ad5c6df34":{"kind":"REWORK","_number":1,"created":"2019-09-26 03:06:04.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/96/684896/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-freezer","ref":"refs/changes/96/684896/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-freezer refs/changes/96/684896/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-freezer refs/changes/96/684896/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-freezer refs/changes/96/684896/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-freezer refs/changes/96/684896/1"}}},"commit":{"parents":[{"commit":"e6d4f47f6d93d21a5f23410db6b32cdf458c4e0e","subject":"Prepare Train M4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-freezer/commit/e6d4f47f6d93d21a5f23410db6b32cdf458c4e0e"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-26 03:06:09.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-26 03:06:48.000000000","tz":480},"subject":"Update the constraints url","message":"Update the constraints url\n\nFor more detail, see http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: I2453b6fd0182e8bdeb533f44f4f017d7392f29a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-freezer/commit/0f3ee47f75b91578099bf4a1ae9ccb0ad5c6df34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-freezer/commit/0f3ee47f75b91578099bf4a1ae9ccb0ad5c6df34"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
