)]}'
{"id":"openstack%2Fpuppet-freezer~707972","triplet_id":"openstack%2Fpuppet-freezer~master~I2ff3ceffdbc0e686161cd1acb769e3fbda3bca35","project":"openstack/puppet-freezer","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"I2ff3ceffdbc0e686161cd1acb769e3fbda3bca35","subject":"[ussuri][goal] Cleanup for python 2.7 drop","status":"MERGED","created":"2020-02-16 00:59:09.000000000","updated":"2020-02-18 03:15:49.000000000","submitted":"2020-02-18 03:14:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707972-1581995658525-5f1f21a0","meta_rev_id":"fa89f44bda404890538f3a0958bbbe07c1db56f2","_number":707972,"virtual_id_number":707972,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-18 03:14:18.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-02-18 02:46:36.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-02-17 20:25:15.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-02-18 02:46:36.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-17 20:25:15.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-02-18 02:46:36.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-02-18 03:14:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1113c0ce6accaa436f356ce2a6fdae426552cf18","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-16 00:59:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3636cee71bc00d9eb8f296b4368f2659f9ca56f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 01:19:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ede75c351fd74ef5af89dfa99c66b11a : SUCCESS in 5m 14s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8cea4de1f1fe4385839fd08ad6e8c093 : SUCCESS in 4m 45s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/02b3d195fa214a028a95ff35bc1f2305 : SUCCESS in 4m 48s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/3c7ecf7116174ebd8ef66fede84e8c56 : SUCCESS in 15m 43s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b084f07ddccd4ce0abe8da364cb51837 : SUCCESS in 8m 16s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9d53133e263647fd8202109eb727fbff : SUCCESS in 7m 35s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e804af74a8814d1cbf5c5c9bdc974b64 : SUCCESS in 8m 16s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e7fbfeabfbd4b879039041ad4415210 : SUCCESS in 6m 08s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff6fa1926a3044f79ea5ba9fd9b7a306 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"82812093e85c8e83e04ada4737a2a9b2ddd06e02","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-02-17 20:25:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e86ea2bfee2bf2c05d35be6800b651c9f3aad91f","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-02-18 02:46:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c3ab372dcc8a7927a65b4b4de830b37ebe582d4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 02:46:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e51ada01e63b0acfedab201abb656e139cc8ca4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 03:14:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4186c03757da3d947fb739c91e0370a065025f66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 03:14:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3599269872a640d99626d2d7923898dd : SUCCESS in 6m 00s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/346dd6868d9a4f32ab9af154c176f522 : SUCCESS in 6m 33s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c007ac8b82af4d3ebac05e10210da68a : SUCCESS in 7m 33s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/f5ed71b4d4cd465ca6a3e388f0efa25e : SUCCESS in 20m 27s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c396f22fcbc34fffb7e07f143b826b5f : SUCCESS in 14m 19s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/832c195aed294c5a890d6af5df4b43b7 : SUCCESS in 13m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4afde815cd1343d198c607e705d0e6f2 : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":1},{"id":"06e1a82d651054c28217477c537f01e32659bb1f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 03:15:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b4e1b4decc247d39f96e7ff431fa293 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"06a1272b2572969abd531b5e3ddf71bec6d3a7fe","revisions":{"06a1272b2572969abd531b5e3ddf71bec6d3a7fe":{"kind":"REWORK","_number":1,"created":"2020-02-16 00:59:09.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/72/707972/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-freezer","ref":"refs/changes/72/707972/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-freezer refs/changes/72/707972/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-freezer refs/changes/72/707972/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-freezer refs/changes/72/707972/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-freezer refs/changes/72/707972/1"}}},"commit":{"parents":[{"commit":"4112f4ed25fcb6f584c103f4d1159b81ac7a8b66","subject":"Prepare Ussuri M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-freezer/commit/4112f4ed25fcb6f584c103f4d1159b81ac7a8b66"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-16 00:58:08.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-16 00:59:08.000000000","tz":-360},"subject":"[ussuri][goal] Cleanup for python 2.7 drop","message":"[ussuri][goal] Cleanup for python 2.7 drop\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\npuppet-* repos have to cleanup the requirement and tox\nfor py2.7 drop.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nChange-Id: I2ff3ceffdbc0e686161cd1acb769e3fbda3bca35\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-freezer/commit/06a1272b2572969abd531b5e3ddf71bec6d3a7fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-freezer/commit/06a1272b2572969abd531b5e3ddf71bec6d3a7fe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
