)]}'
{"id":"openstack%2Fpuppet-freezer~724582","triplet_id":"openstack%2Fpuppet-freezer~master~I63ae282c9fa13724056f72f0bd706b4446505d8f","project":"openstack/puppet-freezer","branch":"master","topic":"bump-openstackdocstheme","hashtags":[],"change_id":"I63ae282c9fa13724056f72f0bd706b4446505d8f","subject":"Bump openstackdocstheme to 1.30.0","status":"MERGED","created":"2020-04-30 07:29:26.000000000","updated":"2020-04-30 07:53:41.000000000","submitted":"2020-04-30 07:52:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724582-1588233127397-047f6eff","meta_rev_id":"499bf439326a01ddb7714dfa85fb8b2ea70bc922","_number":724582,"virtual_id_number":724582,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-30 07:52:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-04-30 07:46:18.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-04-30 07:46:18.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-30 07:52:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0f40a0a6d3c2fd984c1ddf6454f757ad6e026d6d","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-04-30 07:29:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e4145392264d4a6ef723e3e14320897f7a0e056","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 07:40:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b10ab912e934e45a93caa7976e82735 : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":1},{"id":"b438f6760895843cbfd30af564a5c16bba86090b","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-04-30 07:46:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5685787af181fa4d37f8fc531d57b8f5910c7878","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 07:46:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"59d50bf305637374a06ee926b3594d54d377ea4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 07:52:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5001d6c91560483fb2cdfb36c6ae02d6 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":1},{"id":"b080436a49e8e1c25a71539915b5d9adfa310a41","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 07:52:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"63dd4c650dceef3556802fbf406ae743ada4967e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 07:53:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89b098abd4504812938dbe025bdde875 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"61caa8e6f2a983936449abe19eaf81c7907adafa","revisions":{"61caa8e6f2a983936449abe19eaf81c7907adafa":{"kind":"REWORK","_number":1,"created":"2020-04-30 07:29:26.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/82/724582/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-freezer","ref":"refs/changes/82/724582/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-freezer refs/changes/82/724582/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-freezer refs/changes/82/724582/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-freezer refs/changes/82/724582/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-freezer refs/changes/82/724582/1"}}},"commit":{"parents":[{"commit":"92bce9ffc99cc51266170f683ff2d0fbe07c766f","subject":"Change rpsec to rspec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-freezer/commit/92bce9ffc99cc51266170f683ff2d0fbe07c766f"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-04-30 07:28:37.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-04-30 07:28:37.000000000","tz":480},"subject":"Bump openstackdocstheme to 1.30.0","message":"Bump openstackdocstheme to 1.30.0\n\n...to pick up many improvements, including the return of table borders.\n\nChange-Id: I63ae282c9fa13724056f72f0bd706b4446505d8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-freezer/commit/61caa8e6f2a983936449abe19eaf81c7907adafa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-freezer/commit/61caa8e6f2a983936449abe19eaf81c7907adafa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
