)]}'
{"id":"openstack%2Fpuppet-freezer~728175","triplet_id":"openstack%2Fpuppet-freezer~stable%2Fussuri~Ifa4f51f09d6651ddcd2167f4b7107d994ccdce1f","project":"openstack/puppet-freezer","branch":"stable/ussuri","hashtags":[],"change_id":"Ifa4f51f09d6651ddcd2167f4b7107d994ccdce1f","subject":"Switch to Ussuri","status":"MERGED","created":"2020-05-14 17:57:11.000000000","updated":"2020-05-18 06:15:38.000000000","submitted":"2020-05-18 06:15:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728175-1589782538527-1d0809e3","meta_rev_id":"7e68f0e2db73459771365d196921b1d1db338efd","_number":728175,"virtual_id_number":728175,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-18 06:15:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-05-18 05:59:36.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-05-15 22:52:18.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-05-18 05:59:36.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-15 22:52:18.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-05-18 05:59:36.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-05-18 06:15:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ead6ac40b14b5e6e7f326791984022f34003f532","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-14 17:57:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9ae1bb435483579a41a348b3e560957268dd695","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 18:13:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4970cadc4e53408e9b22af08c3259ef1 : SUCCESS in 7m 44s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f0d315c0649f4047ab3aa7e3e0b9652e : SUCCESS in 7m 30s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/83af2a7c09f74d78acac2b60e937c62e : SUCCESS in 6m 06s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/168108a4e1c24c08838913d27add7904 : SUCCESS in 12m 54s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/5062da62fd154eca86db6a3127fd2d61 : SUCCESS in 14m 04s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c94c7c46543496d95b1a942e79c44b0 : SUCCESS in 11m 07s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/75dbd19fbb77444db00e3a1e810d50e1 : SUCCESS in 13m 32s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/eef0797e6a914bf983563b649bfd70af : SUCCESS in 10m 45s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/b302b4b1291a4adf8a0bddce7a1d24ea : SUCCESS in 13m 27s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fdfc297fedff4ad4ac459ed6287e612d : SUCCESS in 10m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"23ad9777487ebae6e4ca97787bf84a4df35f1d0d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-15 22:52:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"90761a0c92cc255524fba4381e6bf1fb0ed7e8d2","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-05-18 05:59:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e159f15ecd65b0c6a4a4e74ae4325ddf40b22dae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 05:59:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f1ad59a9ecd5cd186b49299b1bb69165f3ac70e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 06:15:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7e68f0e2db73459771365d196921b1d1db338efd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 06:15:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/32321386f83c4c699d53a01e346a6964 : SUCCESS in 7m 04s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a3ad639928e431e977177a86a54103e : SUCCESS in 5m 11s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a3d976608e0451a80b0ed6434d36f6b : SUCCESS in 6m 49s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/82453afcaf8240388bd63b268f275837 : SUCCESS in 14m 13s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/b7e9b3ae46ff450f8300ea381fd4e9c1 : SUCCESS in 14m 03s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fba1cc16663c42a7a802f8dabfc2eac2 : SUCCESS in 12m 15s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/c4fae9fb0ac24d5cbae19a50b11c6c9a : SUCCESS in 9m 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c328f29aed857fdd1a6fa7a6998250b8a74ec564","revisions":{"c328f29aed857fdd1a6fa7a6998250b8a74ec564":{"kind":"REWORK","_number":1,"created":"2020-05-14 17:57:11.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/75/728175/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-freezer","ref":"refs/changes/75/728175/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-freezer refs/changes/75/728175/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-freezer refs/changes/75/728175/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-freezer refs/changes/75/728175/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-freezer refs/changes/75/728175/1"}}},"commit":{"parents":[{"commit":"f1828721cffc3cfb7becd90eb1b5f8409702e00b","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-freezer/commit/f1828721cffc3cfb7becd90eb1b5f8409702e00b"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-05-14 17:57:03.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-05-14 17:57:03.000000000","tz":120},"subject":"Switch to Ussuri","message":"Switch to Ussuri\n\nChange-Id: Ifa4f51f09d6651ddcd2167f4b7107d994ccdce1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-freezer/commit/c328f29aed857fdd1a6fa7a6998250b8a74ec564"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-freezer/commit/c328f29aed857fdd1a6fa7a6998250b8a74ec564"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
