)]}'
{"id":"openstack%2Fpuppet-glance~369692","triplet_id":"openstack%2Fpuppet-glance~master~I1e75eafc598d102892b9c111c687646deab931a6","project":"openstack/puppet-glance","branch":"master","topic":"puppet-lint_check","hashtags":[],"change_id":"I1e75eafc598d102892b9c111c687646deab931a6","subject":"Test latest puppet-lint","status":"ABANDONED","created":"2016-09-13 21:03:47.000000000","updated":"2016-11-10 15:35:48.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"275608e0fd7f652f5adf0c0735f506e01bd25804","_number":369692,"virtual_id_number":369692,"owner":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":1,"date":"2016-11-07 11:34:48.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"date":"2016-11-07 10:48:52.000000000","_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"date":"2016-11-07 12:17:42.000000000","_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-11-07 11:34:48.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2016-11-07 12:17:42.000000000","updated_by":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"27eb7287a3b49de9583d88ab19a60c0047c93585","author":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"date":"2016-09-13 21:03:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ca9873d17474e233cf13cabe56c31e0bd3fd9bc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-13 21:42:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-glance-puppet-lint http://logs.openstack.org/92/369692/1/check/gate-puppet-glance-puppet-lint/e495041/ : FAILURE in 3m 04s\n- gate-puppet-glance-puppet-syntax-3-centos-7 http://logs.openstack.org/92/369692/1/check/gate-puppet-glance-puppet-syntax-3-centos-7/dca9191/ : SUCCESS in 6m 16s\n- gate-puppet-glance-puppet-syntax-4-centos-7 http://logs.openstack.org/92/369692/1/check/gate-puppet-glance-puppet-syntax-4-centos-7/2fef67c/ : SUCCESS in 2m 50s\n- gate-puppet-glance-puppet-unit-3.6-centos-7 http://logs.openstack.org/92/369692/1/check/gate-puppet-glance-puppet-unit-3.6-centos-7/e17f032/ : SUCCESS in 19m 29s\n- gate-puppet-glance-puppet-unit-3.8-centos-7 http://logs.openstack.org/92/369692/1/check/gate-puppet-glance-puppet-unit-3.8-centos-7/8561664/ : SUCCESS in 33m 55s\n- gate-puppet-glance-puppet-unit-4.5-centos-7 http://logs.openstack.org/92/369692/1/check/gate-puppet-glance-puppet-unit-4.5-centos-7/67cce89/ : SUCCESS in 21m 47s\n- gate-puppet-glance-puppet-unit-latest-centos-7 http://logs.openstack.org/92/369692/1/check/gate-puppet-glance-puppet-unit-latest-centos-7/e25ff55/ : SUCCESS in 20m 50s (non-voting)\n- gate-puppet-glance-puppet-beaker-rspec-centos-7 http://logs.openstack.org/92/369692/1/check/gate-puppet-glance-puppet-beaker-rspec-centos-7/1a183ba/ : SUCCESS in 9m 12s\n- gate-puppet-glance-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/92/369692/1/check/gate-puppet-glance-puppet-beaker-rspec-ubuntu-xenial/f0b41bc/ : SUCCESS in 7m 56s\n- gate-puppet-openstack-integration-3-scenario001-tempest-centos-7 http://logs.openstack.org/92/369692/1/check/gate-puppet-openstack-integration-3-scenario001-tempest-centos-7/c563ec5/ : SUCCESS in 25m 07s\n- gate-puppet-openstack-integration-3-scenario002-tempest-centos-7 http://logs.openstack.org/92/369692/1/check/gate-puppet-openstack-integration-3-scenario002-tempest-centos-7/b7185c1/ : SUCCESS in 34m 32s\n- gate-puppet-openstack-integration-3-scenario003-tempest-centos-7 http://logs.openstack.org/92/369692/1/check/gate-puppet-openstack-integration-3-scenario003-tempest-centos-7/2930eb9/ : SUCCESS in 22m 06s\n- gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/92/369692/1/check/gate-puppet-openstack-integration-3-scenario001-tempest-ubuntu-xenial/5ede45b/ : SUCCESS in 22m 03s\n- gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/92/369692/1/check/gate-puppet-openstack-integration-3-scenario002-tempest-ubuntu-xenial/b55d371/ : SUCCESS in 26m 11s\n- gate-puppet-openstack-integration-3-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/92/369692/1/check/gate-puppet-openstack-integration-3-scenario003-tempest-ubuntu-xenial/3dd5735/ : SUCCESS in 20m 48s","accounts_in_message":[],"_revision_number":1},{"id":"8cb325206e0ffbf2e8b42d7e92e521b2035d0cc1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-13 22:59:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (check-tripleo pipeline).\n\n- gate-tripleo-ci-centos-7-ovb-nonha http://logs.openstack.org/92/369692/1/check-tripleo/gate-tripleo-ci-centos-7-ovb-nonha/68e6487/ : SUCCESS in 1h 55m 30s","accounts_in_message":[],"_revision_number":1},{"id":"d0afbf36c50bc34a01309d1f00a045f95004c044","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-09-13 23:13:21.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library/2578/ : SUCCESS\n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/4667/ : SUCCESS\n\n* master.puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/master.puppet-openstack.fuel.noop/4730/ : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"ef12381392a92f2ee9d005e538dac7856d5bd43c","author":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"date":"2016-09-14 08:00:51.000000000","message":"Topic set to puppet-lint_check","accounts_in_message":[],"_revision_number":1},{"id":"69178b0b18335430fd8c3676d4f00dcbc6951aba","author":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"date":"2016-09-26 08:30:34.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"e3839f0ed321dc3ad83d8c3c61e768cfc6a90f73","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-26 09:16:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-glance-puppet-lint http://logs.openstack.org/92/369692/2/check/gate-puppet-glance-puppet-lint/e8095a2/ : SUCCESS in 3m 47s\n- gate-puppet-glance-puppet-syntax-3-centos-7 http://logs.openstack.org/92/369692/2/check/gate-puppet-glance-puppet-syntax-3-centos-7/081b81b/ : SUCCESS in 2m 59s\n- gate-puppet-glance-puppet-syntax-4-centos-7 http://logs.openstack.org/92/369692/2/check/gate-puppet-glance-puppet-syntax-4-centos-7/119f682/ : SUCCESS in 5m 48s\n- gate-puppet-glance-puppet-unit-3.6-centos-7 http://logs.openstack.org/92/369692/2/check/gate-puppet-glance-puppet-unit-3.6-centos-7/d3ad332/ : SUCCESS in 21m 49s\n- gate-puppet-glance-puppet-unit-3.8-centos-7 http://logs.openstack.org/92/369692/2/check/gate-puppet-glance-puppet-unit-3.8-centos-7/865a532/ : SUCCESS in 23m 08s\n- gate-puppet-glance-puppet-unit-4.5-centos-7 http://logs.openstack.org/92/369692/2/check/gate-puppet-glance-puppet-unit-4.5-centos-7/aeef165/ : SUCCESS in 17m 01s\n- gate-puppet-glance-puppet-unit-latest-centos-7 http://logs.openstack.org/92/369692/2/check/gate-puppet-glance-puppet-unit-latest-centos-7/b8e3c6c/ : SUCCESS in 19m 27s (non-voting)\n- gate-puppet-glance-puppet-beaker-rspec-centos-7 http://logs.openstack.org/92/369692/2/check/gate-puppet-glance-puppet-beaker-rspec-centos-7/ae75773/ : SUCCESS in 9m 03s\n- gate-puppet-glance-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/92/369692/2/check/gate-puppet-glance-puppet-beaker-rspec-ubuntu-xenial/cda9e9e/ : SUCCESS in 8m 34s\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/92/369692/2/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7/7b7d315/ : SUCCESS in 30m 40s\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/92/369692/2/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7/a1d6e6f/ : SUCCESS in 39m 32s\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/92/369692/2/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7/e91928d/ : SUCCESS in 35m 36s\n- gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/92/369692/2/check/gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/0d9c72d/ : SUCCESS in 28m 46s\n- gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/92/369692/2/check/gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/e86ccb9/ : SUCCESS in 34m 17s\n- gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/92/369692/2/check/gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/771e532/ : SUCCESS in 23m 48s","accounts_in_message":[],"_revision_number":2},{"id":"201cd5b8bc793e7dba5781332bd87d5f4157821c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-26 10:14:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (check-tripleo pipeline).\n\n- gate-tripleo-ci-centos-7-ovb-nonha http://logs.openstack.org/92/369692/2/check-tripleo/gate-tripleo-ci-centos-7-ovb-nonha/578bfa4/ : SUCCESS in 1h 43m 47s","accounts_in_message":[],"_revision_number":2},{"id":"95d7c5063dd9aa505c9f3eb74be3ecdd715bc028","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-09-26 12:15:39.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library/2929/ : FAILURE\n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/5022/ : SUCCESS\n\n* master.puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/master.puppet-openstack.fuel.noop/5085/ : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"07c07d0e7f96bd8fbfd882343652be11838cd228","author":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"date":"2016-11-07 08:48:33.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"0910912a99953da0455afad9c76874fc2ee9eb90","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-07 09:44:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-glance-puppet-lint http://logs.openstack.org/92/369692/3/check/gate-puppet-glance-puppet-lint/d0d918b/ : SUCCESS in 1m 31s\n- gate-puppet-glance-puppet-syntax-3-centos-7 http://logs.openstack.org/92/369692/3/check/gate-puppet-glance-puppet-syntax-3-centos-7/ce8833c/ : FAILURE in 30s\n- gate-puppet-glance-puppet-syntax-4-centos-7 http://logs.openstack.org/92/369692/3/check/gate-puppet-glance-puppet-syntax-4-centos-7/db47cf4/ : FAILURE in 27s\n- gate-puppet-glance-puppet-unit-3.6-centos-7 http://logs.openstack.org/92/369692/3/check/gate-puppet-glance-puppet-unit-3.6-centos-7/98d17da/ : FAILURE in 29s\n- gate-puppet-glance-puppet-unit-3.8-centos-7 http://logs.openstack.org/92/369692/3/check/gate-puppet-glance-puppet-unit-3.8-centos-7/0ac8036/ : FAILURE in 36s\n- gate-puppet-glance-puppet-unit-4.5-centos-7 http://logs.openstack.org/92/369692/3/check/gate-puppet-glance-puppet-unit-4.5-centos-7/0d1246e/ : FAILURE in 28s\n- gate-puppet-glance-puppet-unit-latest-centos-7 http://logs.openstack.org/92/369692/3/check/gate-puppet-glance-puppet-unit-latest-centos-7/97cbb37/ : FAILURE in 34s (non-voting)\n- gate-puppet-glance-puppet-beaker-rspec-centos-7 http://logs.openstack.org/92/369692/3/check/gate-puppet-glance-puppet-beaker-rspec-centos-7/e7fe10c/ : FAILURE in 40s\n- gate-puppet-glance-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/92/369692/3/check/gate-puppet-glance-puppet-beaker-rspec-ubuntu-xenial/6b441c5/ : SUCCESS in 5m 22s\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/92/369692/3/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7/4b69379/ : SUCCESS in 32m 13s\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/92/369692/3/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7/3bea2d6/ : SUCCESS in 53m 18s\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/92/369692/3/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7/5869774/ : SUCCESS in 31m 33s\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/92/369692/3/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7/5638106/ : SUCCESS in 24m 14s\n- gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/92/369692/3/check/gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/c73637c/ : SUCCESS in 21m 15s\n- gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/92/369692/3/check/gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/80d2bb3/ : SUCCESS in 23m 30s\n- gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/92/369692/3/check/gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/b96cbe8/ : SUCCESS in 20m 46s\n- gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/92/369692/3/check/gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/695474f/ : SUCCESS in 20m 10s","accounts_in_message":[],"_revision_number":3},{"id":"76e1790336ecacb782704d7bd27149dbce9be2ad","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-07 09:52:25.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"00b1f333087ede4f320ab92a2e3131730fcf5027","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-11-07 10:45:32.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/5970/ : SUCCESS\n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library/3878/ : SUCCESS\n\n* master.puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/master.puppet-openstack.fuel.noop/6034/ : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"d478ca890f3dbfbc9150ce9d84f8f329b2a64d9b","author":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"date":"2016-11-07 10:48:52.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ee94b9aa12b7def9bec54f39d7f0da9a23a4414f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-07 11:34:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-glance-puppet-lint http://logs.openstack.org/92/369692/3/check/gate-puppet-glance-puppet-lint/a4f48ed/ : SUCCESS in 1m 55s\n- gate-puppet-glance-puppet-syntax-3-centos-7 http://logs.openstack.org/92/369692/3/check/gate-puppet-glance-puppet-syntax-3-centos-7/7e094d6/ : SUCCESS in 3m 04s\n- gate-puppet-glance-puppet-syntax-4-centos-7 http://logs.openstack.org/92/369692/3/check/gate-puppet-glance-puppet-syntax-4-centos-7/7f88579/ : SUCCESS in 3m 55s\n- gate-puppet-glance-puppet-unit-3.6-centos-7 http://logs.openstack.org/92/369692/3/check/gate-puppet-glance-puppet-unit-3.6-centos-7/8b91452/ : SUCCESS in 18m 17s\n- gate-puppet-glance-puppet-unit-3.8-centos-7 http://logs.openstack.org/92/369692/3/check/gate-puppet-glance-puppet-unit-3.8-centos-7/d610ed8/ : SUCCESS in 18m 55s\n- gate-puppet-glance-puppet-unit-4.5-centos-7 http://logs.openstack.org/92/369692/3/check/gate-puppet-glance-puppet-unit-4.5-centos-7/4b1c4e2/ : SUCCESS in 18m 55s\n- gate-puppet-glance-puppet-unit-latest-centos-7 http://logs.openstack.org/92/369692/3/check/gate-puppet-glance-puppet-unit-latest-centos-7/082d172/ : SUCCESS in 17m 03s (non-voting)\n- gate-puppet-glance-puppet-beaker-rspec-centos-7 http://logs.openstack.org/92/369692/3/check/gate-puppet-glance-puppet-beaker-rspec-centos-7/af957e9/ : SUCCESS in 9m 42s\n- gate-puppet-glance-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/92/369692/3/check/gate-puppet-glance-puppet-beaker-rspec-ubuntu-xenial/b801ccc/ : SUCCESS in 8m 04s\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/92/369692/3/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7/2fbd1cc/ : SUCCESS in 33m 14s\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/92/369692/3/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7/4f77683/ : SUCCESS in 41m 52s\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/92/369692/3/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7/1ac0aa0/ : SUCCESS in 39m 18s\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/92/369692/3/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7/f5c0695/ : SUCCESS in 26m 46s\n- gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/92/369692/3/check/gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/8feaa48/ : SUCCESS in 31m 46s\n- gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/92/369692/3/check/gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/f41be44/ : SUCCESS in 24m 40s\n- gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/92/369692/3/check/gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/b74c973/ : SUCCESS in 21m 22s\n- gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/92/369692/3/check/gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/9fad04e/ : SUCCESS in 15m 17s","accounts_in_message":[],"_revision_number":3},{"id":"ed449d7a179df61df7d3f3f8508a7ff717003fd2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-07 11:37:25.000000000","message":"Patch Set 3:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- gate-tripleo-ci-centos-7-ovb-nonha http://logs.openstack.org/92/369692/3/check-tripleo/gate-tripleo-ci-centos-7-ovb-nonha/aa781fd/ : FAILURE in 48m 12s","accounts_in_message":[],"_revision_number":3},{"id":"02f3907f43cfe0510ae045e84ac26fae8d6c1b03","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2016-11-07 12:17:42.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/5988/ : ABORTED\n\n* master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library https://ci.fuel-infra.org/job/master.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library/3896/ : FAILURE\n\n* master.puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/master.puppet-openstack.fuel.noop/6052/ : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"275608e0fd7f652f5adf0c0735f506e01bd25804","author":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"date":"2016-11-10 15:35:48.000000000","message":"Abandoned\n\nNot needed anymore","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"781c70a5cb1b9538bc3952a410ebef1fe53835f5","revisions":{"ada316eb0d1c4a1c3cf96919e84856e25b364183":{"kind":"REWORK","_number":1,"created":"2016-09-13 21:03:47.000000000","uploader":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"ref":"refs/changes/92/369692/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/92/369692/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/92/369692/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/92/369692/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/92/369692/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/92/369692/1"}}},"commit":{"parents":[{"commit":"ffa154c77a179797a04807ad8d13e56aca3001f0","subject":"Update log_file documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/ffa154c77a179797a04807ad8d13e56aca3001f0"}]}],"author":{"name":"Carlos Camacho","email":"ccamacho@redhat.com","date":"2016-09-13 21:03:28.000000000","tz":120},"committer":{"name":"Carlos Camacho","email":"ccamacho@redhat.com","date":"2016-09-13 21:03:44.000000000","tz":120},"subject":"Test latest puppet-lint","message":"Test latest puppet-lint\n\nDO NOT MERGE\n\nChange-Id: I1e75eafc598d102892b9c111c687646deab931a6\nDepends-On: Ica920efbded947c1263bc3b5b218a324bda5bc47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/ada316eb0d1c4a1c3cf96919e84856e25b364183"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/ada316eb0d1c4a1c3cf96919e84856e25b364183"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"ffa154c77a179797a04807ad8d13e56aca3001f0","is_merged_in_target_branch":true,"change_id":"I2b7f3ad6f04b24e357948bd23782b89764e632e5","change_number":367696,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"0f6c4e7ef3067932a2f781aa733bafbff57d43a2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2016-09-26 08:30:34.000000000","uploader":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"ref":"refs/changes/92/369692/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/92/369692/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/92/369692/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/92/369692/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/92/369692/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/92/369692/2"}}},"commit":{"parents":[{"commit":"372fb933208dc21b3be7cf3efa877047f856ea45","subject":"Merge \"Fix puppet-lint before upgrading gem\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/372fb933208dc21b3be7cf3efa877047f856ea45"}]}],"author":{"name":"Carlos Camacho","email":"ccamacho@redhat.com","date":"2016-09-13 21:03:28.000000000","tz":120},"committer":{"name":"Carlos Camacho","email":"ccamacho@redhat.com","date":"2016-09-26 08:30:31.000000000","tz":120},"subject":"Test latest puppet-lint","message":"Test latest puppet-lint\n\nDO NOT MERGE\n\nChange-Id: I1e75eafc598d102892b9c111c687646deab931a6\nDepends-On: Ica920efbded947c1263bc3b5b218a324bda5bc47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/0f6c4e7ef3067932a2f781aa733bafbff57d43a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/0f6c4e7ef3067932a2f781aa733bafbff57d43a2"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"372fb933208dc21b3be7cf3efa877047f856ea45","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"781c70a5cb1b9538bc3952a410ebef1fe53835f5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2016-11-07 08:48:33.000000000","uploader":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"ref":"refs/changes/92/369692/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/92/369692/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/92/369692/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/92/369692/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/92/369692/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/92/369692/3"}}},"commit":{"parents":[{"commit":"800dcf4d4aa39d1d57faca58d8cb4ff22fca2eb2","subject":"Fix puppet version for requirements in metadata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/800dcf4d4aa39d1d57faca58d8cb4ff22fca2eb2"}]}],"author":{"name":"Carlos Camacho","email":"ccamacho@redhat.com","date":"2016-09-13 21:03:28.000000000","tz":120},"committer":{"name":"Carlos Camacho","email":"ccamacho@redhat.com","date":"2016-11-07 08:48:31.000000000","tz":60},"subject":"Test latest puppet-lint","message":"Test latest puppet-lint\n\nDO NOT MERGE\n\nChange-Id: I1e75eafc598d102892b9c111c687646deab931a6\nDepends-On: Ica920efbded947c1263bc3b5b218a324bda5bc47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/781c70a5cb1b9538bc3952a410ebef1fe53835f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/781c70a5cb1b9538bc3952a410ebef1fe53835f5"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"800dcf4d4aa39d1d57faca58d8cb4ff22fca2eb2","is_merged_in_target_branch":true,"change_id":"I356613a64b99c5e57e211fd1f24e2e484a8cc05b","change_number":393065,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
