)]}'
{"id":"openstack%2Fpuppet-glance~515917","triplet_id":"openstack%2Fpuppet-glance~stable%2Fpike~I773355a74ea45cf90c5a8b43e143766a3f35d2b7","project":"openstack/puppet-glance","branch":"stable/pike","topic":"bindep/puppet","hashtags":[],"change_id":"I773355a74ea45cf90c5a8b43e143766a3f35d2b7","subject":"[pike] Add Puppet package to bindep, for module build","status":"MERGED","created":"2017-10-27 20:42:46.000000000","updated":"2017-10-30 03:14:56.000000000","submitted":"2017-10-30 03:14:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"515917-1509333296352-9631ecff","meta_rev_id":"b1f7598b9fabc4bc6517de9461231e73be34eb86","_number":515917,"virtual_id_number":515917,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2017-10-30 03:14:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2017-10-28 14:52:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2017-10-30 01:06:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-10-30 01:06:44.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2017-10-30 03:14:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"503d6b7c01a7af16031a2aed1a36e19b2e0c8eb4","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2017-10-27 20:42:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09ac64d9485865039c0146ba332b8a01452f9bdb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-28 02:08:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint http://logs.openstack.org/17/515917/1/check/puppet-openstack-lint/5983719/ : SUCCESS in 5m 25s\n- puppet-openstack-syntax-4 http://logs.openstack.org/17/515917/1/check/puppet-openstack-syntax-4/dec5a12/ : SUCCESS in 4m 20s\n- legacy-puppet-unit-4.8-centos-7 http://logs.openstack.org/17/515917/1/check/legacy-puppet-unit-4.8-centos-7/b6d0099/ : SUCCESS in 17m 26s\n- legacy-puppet-beaker-rspec-centos-7 http://logs.openstack.org/17/515917/1/check/legacy-puppet-beaker-rspec-centos-7/76ca6bd/ : SUCCESS in 27m 18s\n- legacy-puppet-beaker-rspec http://logs.openstack.org/17/515917/1/check/legacy-puppet-beaker-rspec/d5d7373/ : SUCCESS in 17m 35s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/17/515917/1/check/puppet-openstack-integration-4-scenario001-tempest-centos-7/e0005fd/ : FAILURE in 22m 03s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/17/515917/1/check/puppet-openstack-integration-4-scenario002-tempest-centos-7/9bd5995/ : FAILURE in 16m 54s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/17/515917/1/check/puppet-openstack-integration-4-scenario003-tempest-centos-7/b105b3c/ : FAILURE in 9m 54s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/17/515917/1/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/dfdbad3/ : FAILURE in 10m 14s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/17/515917/1/check/puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/4dbd1e4/ : FAILURE in 11m 19s\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/17/515917/1/check/puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/2c571fd/ : FAILURE in 14m 59s\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/17/515917/1/check/puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/cd848ae/ : FAILURE in 14m 56s\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/17/515917/1/check/puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/4d0085e/ : FAILURE in 7m 16s (non-voting)\n- legacy-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/17/515917/1/check/legacy-tripleo-ci-centos-7-nonha-multinode-oooq/06114cd/ : SUCCESS in 2h 05m 08s","accounts_in_message":[],"_revision_number":1},{"id":"48fc92b018a7787a6121c33b3a7e507a5a943cc1","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2017-10-28 14:43:26.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"4b39db7cc3631688efcf128c78b4cf9546f8e3a7","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2017-10-28 14:52:19.000000000","message":"Patch Set 2: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5516d3bf08b94b137c695ea46fec7c603d5ee49e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-28 16:46:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/17/515917/2/check/puppet-openstack-lint/9cb3c6a/ : SUCCESS in 3m 35s\n- puppet-openstack-syntax-4 http://logs.openstack.org/17/515917/2/check/puppet-openstack-syntax-4/7cefa6c/ : SUCCESS in 3m 32s\n- legacy-puppet-unit-4.8-centos-7 http://logs.openstack.org/17/515917/2/check/legacy-puppet-unit-4.8-centos-7/5cd4405/ : SUCCESS in 20m 43s\n- legacy-puppet-beaker-rspec-centos-7 http://logs.openstack.org/17/515917/2/check/legacy-puppet-beaker-rspec-centos-7/522bc41/ : SUCCESS in 22m 51s\n- legacy-puppet-beaker-rspec http://logs.openstack.org/17/515917/2/check/legacy-puppet-beaker-rspec/97ebaf7/ : SUCCESS in 15m 42s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/17/515917/2/check/puppet-openstack-integration-4-scenario001-tempest-centos-7/a929325/ : SUCCESS in 1h 08m 14s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/17/515917/2/check/puppet-openstack-integration-4-scenario002-tempest-centos-7/d21e4f5/ : SUCCESS in 1h 03m 11s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/17/515917/2/check/puppet-openstack-integration-4-scenario003-tempest-centos-7/d89b45e/ : SUCCESS in 1h 06m 40s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/17/515917/2/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/f98b08e/ : SUCCESS in 39m 48s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/17/515917/2/check/puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/912f34c/ : SUCCESS in 38m 22s\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/17/515917/2/check/puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/22ade5c/ : SUCCESS in 41m 39s\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/17/515917/2/check/puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/89ea339/ : SUCCESS in 35m 31s\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/17/515917/2/check/puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/43e6033/ : SUCCESS in 27m 11s (non-voting)\n- legacy-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/17/515917/2/check/legacy-tripleo-ci-centos-7-nonha-multinode-oooq/c3d70ea/ : SUCCESS in 2h 01m 00s","accounts_in_message":[],"_revision_number":2},{"id":"dba9460e96d791d2f60a6e5fe9a54a64defd766b","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2017-10-30 01:00:03.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"c4fc8cc92fdda2dbeba6dfc1ab01a0bd887445fe","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2017-10-30 01:06:44.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"636c41ee61c53402e3f0fe0616f6248e2d1afedb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-30 01:06:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d1921d72a034909c27b18ac6381b15a9c69fde6a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-30 03:14:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/17/515917/2/gate/puppet-openstack-lint/2cdf7f5/ : SUCCESS in 4m 20s\n- puppet-openstack-syntax-4 http://logs.openstack.org/17/515917/2/gate/puppet-openstack-syntax-4/e32acf2/ : SUCCESS in 6m 00s\n- legacy-puppet-unit-4.8-centos-7 http://logs.openstack.org/17/515917/2/gate/legacy-puppet-unit-4.8-centos-7/fd60400/ : SUCCESS in 16m 44s\n- legacy-puppet-beaker-rspec-centos-7 http://logs.openstack.org/17/515917/2/gate/legacy-puppet-beaker-rspec-centos-7/07cdb27/ : SUCCESS in 16m 58s\n- legacy-puppet-beaker-rspec http://logs.openstack.org/17/515917/2/gate/legacy-puppet-beaker-rspec/340985e/ : SUCCESS in 10m 30s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/17/515917/2/gate/puppet-openstack-integration-4-scenario001-tempest-centos-7/0e59ae3/ : SUCCESS in 1h 01m 54s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/17/515917/2/gate/puppet-openstack-integration-4-scenario002-tempest-centos-7/ea8e5ff/ : SUCCESS in 58m 22s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/17/515917/2/gate/puppet-openstack-integration-4-scenario003-tempest-centos-7/e1279ad/ : SUCCESS in 1h 05m 35s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/17/515917/2/gate/puppet-openstack-integration-4-scenario004-tempest-centos-7/9c844d1/ : SUCCESS in 47m 58s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/17/515917/2/gate/puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/8c74861/ : SUCCESS in 44m 42s\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/17/515917/2/gate/puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/de23a0f/ : SUCCESS in 46m 12s\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/17/515917/2/gate/puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/88590f5/ : SUCCESS in 32m 43s\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/17/515917/2/gate/puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/33cb9e7/ : SUCCESS in 29m 08s (non-voting)\n- legacy-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/17/515917/2/gate/legacy-tripleo-ci-centos-7-nonha-multinode-oooq/81b4b94/ : SUCCESS in 2h 03m 17s","accounts_in_message":[],"_revision_number":2},{"id":"bab498705c18ea2fdd101a6380ac0c14f74da27d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-30 03:14:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"780f0a569f5f27c85fc35a1d9a0c0d4b44da98b2","revisions":{"098c81950286571be829ca2f2af504518eb9bb01":{"kind":"REWORK","_number":1,"created":"2017-10-27 20:42:46.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/17/515917/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/17/515917/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/17/515917/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/17/515917/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/17/515917/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/17/515917/1"}}},"commit":{"parents":[{"commit":"b3bb1db2a81cab561aa64b8f7ff87f6431cef4b3","subject":"Merge \"Update UPPER_CONSTRAINTS_FILE for stable/pike\" into stable/pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/b3bb1db2a81cab561aa64b8f7ff87f6431cef4b3"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2017-10-27 20:42:41.000000000","tz":-420},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2017-10-27 20:42:41.000000000","tz":-420},"subject":"[pike] Add Puppet package to bindep, for module build","message":"[pike] Add Puppet package to bindep, for module build\n\nWe need Puppet package deployed from bindep so we can\nrun puppet module build with the new zuul v3 job.\n\nChange-Id: I773355a74ea45cf90c5a8b43e143766a3f35d2b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/098c81950286571be829ca2f2af504518eb9bb01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/098c81950286571be829ca2f2af504518eb9bb01"}]},"parents_data":[{"branch_name":"refs/heads/stable/pike","commit_id":"b3bb1db2a81cab561aa64b8f7ff87f6431cef4b3","is_merged_in_target_branch":false}],"branch":"refs/heads/stable/pike"},"780f0a569f5f27c85fc35a1d9a0c0d4b44da98b2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2017-10-28 14:43:26.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/17/515917/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/17/515917/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/17/515917/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/17/515917/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/17/515917/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/17/515917/2"}}},"commit":{"parents":[{"commit":"b3bb1db2a81cab561aa64b8f7ff87f6431cef4b3","subject":"Merge \"Update UPPER_CONSTRAINTS_FILE for stable/pike\" into stable/pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/b3bb1db2a81cab561aa64b8f7ff87f6431cef4b3"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2017-10-27 20:42:41.000000000","tz":-420},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2017-10-28 14:43:23.000000000","tz":0},"subject":"[pike] Add Puppet package to bindep, for module build","message":"[pike] Add Puppet package to bindep, for module build\n\nWe need Puppet package deployed from bindep so we can\nrun puppet module build with the new zuul v3 job.\n\nDepends-On: Iccff9449202345462369d71c7fe97c2f82d909cb\nChange-Id: I773355a74ea45cf90c5a8b43e143766a3f35d2b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/780f0a569f5f27c85fc35a1d9a0c0d4b44da98b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/780f0a569f5f27c85fc35a1d9a0c0d4b44da98b2"}]},"parents_data":[{"branch_name":"refs/heads/stable/pike","commit_id":"b3bb1db2a81cab561aa64b8f7ff87f6431cef4b3","is_merged_in_target_branch":false}],"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
