)]}'
{"id":"openstack%2Fpuppet-glance~660002","triplet_id":"openstack%2Fpuppet-glance~master~I7549012ec06f5e20328cc9a32190a8b9ba89953c","project":"openstack/puppet-glance","branch":"master","topic":"fix-client-test","hashtags":[],"change_id":"I7549012ec06f5e20328cc9a32190a8b9ba89953c","subject":"Fix client spec test","status":"MERGED","created":"2019-05-20 03:04:04.000000000","updated":"2019-05-20 07:20:39.000000000","submitted":"2019-05-20 07:20:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660002-1558336839794-354c625b","meta_rev_id":"ced39c4a256a13ec8a8d6e766e27a0d82ef46b95","_number":660002,"virtual_id_number":660002,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-05-20 07:20:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2019-05-20 05:34:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2019-05-20 05:34:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-20 07:20:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a100146992265b81c3857b7b2a3a62759b8fe7bf","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-20 03:04:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f99be6f414a84f38fdd90d094aef8567d7ee50b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 04:37:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/02/660002/1/check/puppet-openstack-lint-ubuntu-bionic/3f2f545/ : SUCCESS in 4m 33s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/02/660002/1/check/puppet-openstack-syntax-4-ubuntu-bionic/4b0e1c0/ : SUCCESS in 3m 54s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/02/660002/1/check/puppet-openstack-syntax-5-ubuntu-bionic/0efcf19/ : SUCCESS in 5m 31s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/02/660002/1/check/puppet-openstack-syntax-6-ubuntu-bionic/283c6a3/ : SUCCESS in 3m 23s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/02/660002/1/check/puppet-openstack-unit-4.8-centos-7/f3294b4/ : SUCCESS in 13m 03s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/02/660002/1/check/puppet-openstack-unit-5.5-centos-7/f45ec3d/ : SUCCESS in 12m 03s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/02/660002/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/dbe9e46/ : SUCCESS in 11m 14s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/02/660002/1/check/puppet-openstack-unit-latest-ubuntu-bionic/1468c91/ : SUCCESS in 8m 31s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/02/660002/1/check/tripleo-puppet-ci-centos-7-undercloud-containers/293adcc/ : SUCCESS in 1h 28m 27s (non-voting)\n- puppet-glance-tripleo-standalone http://logs.openstack.org/02/660002/1/check/puppet-glance-tripleo-standalone/bb11e50/ : SUCCESS in 1h 27m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3c0d1ac955102ff30209c7a332f0c547eeb3042a","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-20 05:34:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d7b0d9a696c1cd82561b63f63cef98391e79faa2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 05:34:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d96ed55d3a7d4879399c7b63849e6f56ab73a258","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 07:20:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fa49cb366fc938cdcc6784eebdbd50c8722253f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 07:20:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/02/660002/1/gate/puppet-openstack-lint-ubuntu-bionic/be3ad42/ : SUCCESS in 4m 28s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/02/660002/1/gate/puppet-openstack-syntax-4-ubuntu-bionic/a1a77bf/ : SUCCESS in 6m 18s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/02/660002/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/f133dbb/ : SUCCESS in 3m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/02/660002/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/f383891/ : SUCCESS in 6m 14s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/02/660002/1/gate/puppet-openstack-unit-4.8-centos-7/be24f08/ : SUCCESS in 13m 09s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/02/660002/1/gate/puppet-openstack-unit-5.5-centos-7/6403f48/ : SUCCESS in 16m 13s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/02/660002/1/gate/puppet-openstack-unit-6.4-ubuntu-bionic/a9c8f60/ : SUCCESS in 6m 37s\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/02/660002/1/gate/tripleo-puppet-ci-centos-7-undercloud-containers/88f0557/ : SUCCESS in 1h 34m 39s (non-voting)\n- puppet-glance-tripleo-standalone http://logs.openstack.org/02/660002/1/gate/puppet-glance-tripleo-standalone/73d1145/ : SUCCESS in 1h 31m 40s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"757d987e509ad5a6b32f3904d4ee6504325238c3","revisions":{"757d987e509ad5a6b32f3904d4ee6504325238c3":{"kind":"REWORK","_number":1,"created":"2019-05-20 03:04:04.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/02/660002/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/02/660002/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/02/660002/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/02/660002/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/02/660002/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/02/660002/1"}}},"commit":{"parents":[{"commit":"a6c77dc3818b0d638468c4630049de6bd8d90013","subject":"Port puppet-glance-tripleo-standalone to os_tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/a6c77dc3818b0d638468c4630049de6bd8d90013"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-20 03:03:19.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-20 03:03:19.000000000","tz":480},"subject":"Fix client spec test","message":"Fix client spec test\n\nChange-Id: I7549012ec06f5e20328cc9a32190a8b9ba89953c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/757d987e509ad5a6b32f3904d4ee6504325238c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/757d987e509ad5a6b32f3904d4ee6504325238c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
