)]}'
{"id":"openstack%2Fpuppet-glance~663048","triplet_id":"openstack%2Fpuppet-glance~stable%2Fstein~I0d892663dd9aee86f56348175fd31016a17fabf5","project":"openstack/puppet-glance","branch":"stable/stein","topic":"bp/split-controlplane-glance-cache-stable/stein","hashtags":[],"change_id":"I0d892663dd9aee86f56348175fd31016a17fabf5","subject":"Add configuration options for the image cache","status":"MERGED","created":"2019-06-04 14:26:00.000000000","updated":"2019-06-07 23:15:17.000000000","submitted":"2019-06-07 23:15:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663048-1559949317586-2d8656ec","meta_rev_id":"2c13da53106b8d2f8d253018217aa064f09f384a","_number":663048,"virtual_id_number":663048,"owner":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":2,"date":"2019-06-07 23:15:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"recommended":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"all":[{"value":1,"date":"2019-06-06 12:41:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":2,"date":"2019-06-04 17:15:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-07 13:54:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2019-06-04 14:42:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-06-07 13:54:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-04 14:42:17.000000000","updated_by":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"reviewer":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"state":"REVIEWER"},{"updated":"2019-06-04 17:15:22.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-06 12:41:22.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2019-06-07 13:54:18.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-06-07 23:15:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"98ae608e2051a3e4f0c20394adf1a342e162b1be","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2019-06-04 14:26:00.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e47822c6952ad48f56e63d9e87cbb923254dc57d","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2019-06-04 14:42:17.000000000","message":"Patch Set 1: Code-Review+1\n\nthanks","accounts_in_message":[],"_revision_number":1},{"id":"ffa1b2759a07a5be5f4494585ea1495674c70679","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 16:36:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/48/663048/1/check/puppet-openstack-lint-ubuntu-bionic/3607613/ : SUCCESS in 4m 08s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/48/663048/1/check/puppet-openstack-syntax-4-ubuntu-bionic/18769a6/ : SUCCESS in 5m 29s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/48/663048/1/check/puppet-openstack-syntax-5-ubuntu-bionic/0fd8a56/ : SUCCESS in 4m 44s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/48/663048/1/check/puppet-openstack-syntax-6-ubuntu-bionic/6981115/ : SUCCESS in 4m 56s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/48/663048/1/check/puppet-openstack-unit-4.8-centos-7/4747757/ : SUCCESS in 21m 43s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/48/663048/1/check/puppet-openstack-unit-5.5-centos-7/44adec6/ : SUCCESS in 21m 36s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/48/663048/1/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/c65380d/ : SUCCESS in 1h 05m 10s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/48/663048/1/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/8bb959b/ : SUCCESS in 1h 06m 07s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/48/663048/1/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/8e7eb2f/ : SUCCESS in 1h 10m 22s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus http://logs.openstack.org/48/663048/1/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/a9ff224/ : SUCCESS in 45m 44s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/48/663048/1/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/2efd78f/ : FAILURE in 18m 53s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/48/663048/1/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/d501cc0/ : FAILURE in 18m 20s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/48/663048/1/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/22e6f5b/ : FAILURE in 28m 10s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/48/663048/1/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/ae0a925/ : FAILURE in 21m 36s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/48/663048/1/check/puppet-openstack-beaker-centos-7/37534db/ : SUCCESS in 15m 49s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/48/663048/1/check/puppet-openstack-beaker-ubuntu-bionic/9d9645f/ : FAILURE in 6m 59s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/48/663048/1/check/tripleo-puppet-ci-centos-7-undercloud-containers/0ac3d80/ : SUCCESS in 1h 51m 39s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/48/663048/1/check/build-openstack-releasenotes/b3fb004/html/ : SUCCESS in 2m 39s\n- puppet-glance-tripleo-standalone http://logs.openstack.org/48/663048/1/check/puppet-glance-tripleo-standalone/26a8707/ : FAILURE in 2h 07m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3878e870356691848378634982b0be25a45fd0cc","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-04 17:15:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b1f83b4e1dfcedae567d3acf8135ce376d752bdb","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2019-06-06 12:41:22.000000000","message":"Patch Set 1: Code-Review+1\n\nThanks, Alan.","accounts_in_message":[],"_revision_number":1},{"id":"2b95561de41b94c746f1056c9926726f96ba22b4","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-06-07 13:54:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"136bc8654fc7b5e52175dfebf0d7e60476df4c1a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 13:54:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"28aff2053d46b8bfc6b654f8eeaf93ed055ca7c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 16:06:23.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/48/663048/1/gate/puppet-openstack-lint-ubuntu-bionic/4c7bb87/ : SUCCESS in 5m 22s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/48/663048/1/gate/puppet-openstack-syntax-4-ubuntu-bionic/27d5d2b/ : SUCCESS in 4m 14s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/48/663048/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/177c35f/ : SUCCESS in 4m 04s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/48/663048/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/5a0bd26/ : SUCCESS in 4m 24s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/48/663048/1/gate/puppet-openstack-unit-4.8-centos-7/ccda1e9/ : SUCCESS in 19m 28s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/48/663048/1/gate/puppet-openstack-unit-5.5-centos-7/67d0955/ : SUCCESS in 18m 57s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/48/663048/1/gate/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/a88b0b5/ : SUCCESS in 49m 29s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/48/663048/1/gate/puppet-openstack-integration-5-scenario002-tempest-centos-7/e955328/ : SUCCESS in 1h 03m 09s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/48/663048/1/gate/puppet-openstack-integration-5-scenario003-tempest-centos-7/8851828/ : FAILURE in 1h 08m 29s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus http://logs.openstack.org/48/663048/1/gate/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/f8e3e3d/ : SUCCESS in 46m 07s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/48/663048/1/gate/puppet-openstack-beaker-centos-7/b4c3b5b/ : SUCCESS in 21m 46s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/48/663048/1/gate/puppet-openstack-beaker-ubuntu-bionic/d998622/ : FAILURE in 10m 14s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/48/663048/1/gate/build-openstack-releasenotes/c77ee7e/html/ : SUCCESS in 2m 35s\n- puppet-glance-tripleo-standalone http://logs.openstack.org/48/663048/1/gate/puppet-glance-tripleo-standalone/742e8b6/ : FAILURE in 2h 03m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"329ee9e12a0eead178ec10399cff55f841ca0182","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-07 16:07:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e119a51921bd678c4626a57ea7c431150644f238","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 18:09:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/48/663048/1/check/puppet-openstack-lint-ubuntu-bionic/37f7aae/ : SUCCESS in 4m 47s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/48/663048/1/check/puppet-openstack-syntax-4-ubuntu-bionic/560ffa2/ : SUCCESS in 3m 57s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/48/663048/1/check/puppet-openstack-syntax-5-ubuntu-bionic/26e3139/ : SUCCESS in 3m 37s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/48/663048/1/check/puppet-openstack-syntax-6-ubuntu-bionic/1eff56e/ : SUCCESS in 5m 08s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/48/663048/1/check/puppet-openstack-unit-4.8-centos-7/13fdd3a/ : SUCCESS in 19m 03s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/48/663048/1/check/puppet-openstack-unit-5.5-centos-7/d123325/ : SUCCESS in 12m 59s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/48/663048/1/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/2b04fb3/ : SUCCESS in 1h 03m 56s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/48/663048/1/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/e6cd523/ : SUCCESS in 1h 03m 27s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/48/663048/1/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/5e9dbc2/ : SUCCESS in 58m 21s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus http://logs.openstack.org/48/663048/1/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/e22fbf9/ : SUCCESS in 45m 09s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/48/663048/1/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/70a2e9b/ : FAILURE in 3m 31s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/48/663048/1/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/80a6991/ : FAILURE in 4m 24s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/48/663048/1/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/84974dd/ : FAILURE in 46m 15s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/48/663048/1/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/38493a2/ : FAILURE in 22m 39s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/48/663048/1/check/puppet-openstack-beaker-centos-7/6a80200/ : SUCCESS in 19m 27s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/48/663048/1/check/puppet-openstack-beaker-ubuntu-bionic/413eb86/ : FAILURE in 7m 49s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/48/663048/1/check/tripleo-puppet-ci-centos-7-undercloud-containers/5ff1c13/ : SUCCESS in 1h 31m 53s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/48/663048/1/check/build-openstack-releasenotes/86d41e0/html/ : SUCCESS in 2m 30s\n- puppet-glance-tripleo-standalone http://logs.openstack.org/48/663048/1/check/puppet-glance-tripleo-standalone/cf9213d/ : FAILURE in 1h 55m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8c664f8a44d75d937155b5ea9c501589baab55de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 18:09:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4e8dfa1b60ea3e52d3e31cc5e52adf5cf67b137d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 21:07:29.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"54d5e8724c93f811c2f3401fa2245100f245bb84","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 23:15:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2a861746b07a9be8140c75403ba026d4b6348787","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 23:15:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/48/663048/1/gate/puppet-openstack-lint-ubuntu-bionic/19c8625/ : SUCCESS in 5m 36s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/48/663048/1/gate/puppet-openstack-syntax-4-ubuntu-bionic/49082c8/ : SUCCESS in 7m 10s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/48/663048/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/88ca88a/ : SUCCESS in 4m 46s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/48/663048/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/339c1de/ : SUCCESS in 7m 14s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/48/663048/1/gate/puppet-openstack-unit-4.8-centos-7/b266751/ : SUCCESS in 20m 58s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/48/663048/1/gate/puppet-openstack-unit-5.5-centos-7/5598b1a/ : SUCCESS in 20m 04s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/48/663048/1/gate/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/60ff82b/ : SUCCESS in 56m 51s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/48/663048/1/gate/puppet-openstack-integration-5-scenario002-tempest-centos-7/c9fd4a8/ : SUCCESS in 1h 03m 15s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/48/663048/1/gate/puppet-openstack-integration-5-scenario003-tempest-centos-7/0b5e420/ : SUCCESS in 1h 09m 25s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus http://logs.openstack.org/48/663048/1/gate/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/af9b511/ : SUCCESS in 48m 32s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/48/663048/1/gate/puppet-openstack-beaker-centos-7/f7715f3/ : SUCCESS in 20m 11s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/48/663048/1/gate/puppet-openstack-beaker-ubuntu-bionic/f974fe7/ : FAILURE in 11m 24s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/48/663048/1/gate/build-openstack-releasenotes/76b8ad9/html/ : SUCCESS in 3m 49s\n- puppet-glance-tripleo-standalone http://logs.openstack.org/48/663048/1/gate/puppet-glance-tripleo-standalone/f002034/ : FAILURE in 2h 05m 39s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fb15d58effb8187a3a0c21364544ad0bf709a836","revisions":{"fb15d58effb8187a3a0c21364544ad0bf709a836":{"kind":"REWORK","_number":1,"created":"2019-06-04 14:26:00.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/48/663048/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/48/663048/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/48/663048/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/48/663048/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/48/663048/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/48/663048/1"}}},"commit":{"parents":[{"commit":"e1c293d22acc54e51cae41531e2b3c4c3f758458","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e1c293d22acc54e51cae41531e2b3c4c3f758458"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2019-05-10 04:42:37.000000000","tz":330},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2019-06-04 14:26:00.000000000","tz":0},"subject":"Add configuration options for the image cache","message":"Add configuration options for the image cache\n\nThe Glance cache uses two conf files:\n1. glance-api.conf is for the server and\n2. glance-cache.conf is for the utilities.\n\nThe following options are already in both configuration files but\nthese need the same values otherwise the cache will potentially run\ninto problems,\n\n - image_cache_dir\n - image_cache_max_size\n - image_cache_stall_time\n\nEnabling above parameters in both conf files wherever they are missing.\n\nPartial-Implements: blueprint split-controlplane-glance-cache\nChange-Id: I0d892663dd9aee86f56348175fd31016a17fabf5\n(cherry picked from commit c8ce009bcdaa9072bce772e8852c9bf876237a77)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/fb15d58effb8187a3a0c21364544ad0bf709a836"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/fb15d58effb8187a3a0c21364544ad0bf709a836"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
