)]}'
{"id":"openstack%2Fpuppet-glance~665787","triplet_id":"openstack%2Fpuppet-glance~master~I5a3c6b4502549e028c47293d5ffb5286a5eec3c0","project":"openstack/puppet-glance","branch":"master","hashtags":[],"change_id":"I5a3c6b4502549e028c47293d5ffb5286a5eec3c0","subject":"assert_private() params and deps classes","status":"ABANDONED","created":"2019-06-17 23:17:39.000000000","updated":"2020-09-18 22:56:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c543eba0468cd7d8fcb8a8e4216b392e4692b875","_number":665787,"virtual_id_number":665787,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-06-18 00:52:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-06-18 00:19:24.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-18 00:19:24.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-18 00:52:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"63795b37e9b2eaeacc79df79115151f76c321f8b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-06-17 23:17:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"839a88e6c545295387c157d3ab2e79fb3635ae91","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-18 00:19:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/87/665787/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/9043920/ : SUCCESS in 38m 29s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/87/665787/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/bf37e80/ : SUCCESS in 1h 00m 10s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/87/665787/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/f91b8e1/ : SUCCESS in 42m 37s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/87/665787/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/7abae27/ : SUCCESS in 34m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2c2886ed2fd2f0a39672918f67a8aafecfc87016","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 00:52:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/87/665787/1/check/puppet-openstack-lint-ubuntu-bionic/863a38d/ : SUCCESS in 4m 27s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/87/665787/1/check/puppet-openstack-syntax-5-ubuntu-bionic/92d8599/ : SUCCESS in 4m 22s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/87/665787/1/check/puppet-openstack-syntax-6-ubuntu-bionic/67a4f09/ : SUCCESS in 4m 39s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/87/665787/1/check/puppet-openstack-unit-5.5-centos-7/911e873/ : SUCCESS in 13m 30s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/87/665787/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/3454c9c/ : SUCCESS in 10m 04s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/87/665787/1/check/puppet-openstack-unit-latest-ubuntu-bionic/16af4e0/ : SUCCESS in 8m 30s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/87/665787/1/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/0c389e5/ : SUCCESS in 56m 53s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/87/665787/1/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/3e92572/ : SUCCESS in 55m 12s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/87/665787/1/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/f4bad3b/ : SUCCESS in 1h 06m 52s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus http://logs.openstack.org/87/665787/1/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/72133b7/ : SUCCESS in 42m 25s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/87/665787/1/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/e9705b0/ : FAILURE in 30m 16s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/87/665787/1/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/e1a26cd/ : FAILURE in 32m 21s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/87/665787/1/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/ad7ef14/ : FAILURE in 3m 32s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/87/665787/1/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/9f412dc/ : FAILURE in 26m 22s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/87/665787/1/check/puppet-openstack-beaker-centos-7/b021967/ : SUCCESS in 16m 49s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/87/665787/1/check/puppet-openstack-beaker-ubuntu-bionic/4d6214b/ : FAILURE in 9m 06s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/87/665787/1/check/tripleo-puppet-ci-centos-7-undercloud-containers/aea9080/ : SUCCESS in 1h 17m 44s (non-voting)\n- puppet-glance-tripleo-standalone http://logs.openstack.org/87/665787/1/check/puppet-glance-tripleo-standalone/35d34f1/ : SUCCESS in 1h 25m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c543eba0468cd7d8fcb8a8e4216b392e4692b875","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-18 22:56:03.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"972cdacfa9b4fcee4f1dc78a6ad35560cd856762","revisions":{"972cdacfa9b4fcee4f1dc78a6ad35560cd856762":{"kind":"REWORK","_number":1,"created":"2019-06-17 23:17:39.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/87/665787/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/87/665787/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/87/665787/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/87/665787/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/87/665787/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/87/665787/1"}}},"commit":{"parents":[{"commit":"c5a4e73193fa63403036d54deb78891051871509","subject":"Prepare Train M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/c5a4e73193fa63403036d54deb78891051871509"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:17:30.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:17:30.000000000","tz":120},"subject":"assert_private() params and deps classes","message":"assert_private() params and deps classes\n\nThese classes are private and should never be\nused outside of it\u0027s module.\n\nChange-Id: I5a3c6b4502549e028c47293d5ffb5286a5eec3c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/972cdacfa9b4fcee4f1dc78a6ad35560cd856762"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/972cdacfa9b4fcee4f1dc78a6ad35560cd856762"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
