)]}'
{"id":"openstack%2Fpuppet-glance~669037","triplet_id":"openstack%2Fpuppet-glance~master~If15f7cc9cef9a3bca855a3cd02068d45eedd171f","project":"openstack/puppet-glance","branch":"master","topic":"bug/1835161","hashtags":[],"change_id":"If15f7cc9cef9a3bca855a3cd02068d45eedd171f","subject":"Manage oslo_middleware/max_request_body_size","status":"MERGED","created":"2019-07-04 03:44:21.000000000","updated":"2019-07-05 02:35:34.000000000","submitted":"2019-07-05 02:35:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669037-1562294134376-769ffea3","meta_rev_id":"2a2918a3451b147a6bb86a340ca20cd73db7a3f8","_number":669037,"virtual_id_number":669037,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-07-05 02:35:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-04 06:48:46.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-07-04 13:59:30.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-07-05 00:51:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-04 06:48:46.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-04 13:59:30.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-07-05 02:35:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e5826357e29a1f309ace773e4b376a707ac069ea","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-07-04 03:44:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"460c85704508dfaf713c14aab6278d852556f67c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-04 04:45:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/37/669037/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/f3d7709/ : SUCCESS in 50m 24s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/37/669037/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/f5890fe/ : SUCCESS in 57m 01s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/37/669037/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/a4a47e5/ : SUCCESS in 57m 31s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/37/669037/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/6f13030/ : SUCCESS in 42m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8ba4154659f8f3c02ea12337d45f3e54023b95e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 05:53:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/37/669037/1/check/puppet-openstack-lint-ubuntu-bionic/ebbac99/ : SUCCESS in 6m 28s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/37/669037/1/check/puppet-openstack-syntax-5-ubuntu-bionic/fc0e021/ : SUCCESS in 6m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/37/669037/1/check/puppet-openstack-syntax-6-ubuntu-bionic/352d7cb/ : SUCCESS in 5m 15s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/37/669037/1/check/puppet-openstack-unit-5.5-centos-7/0e92a76/ : SUCCESS in 14m 07s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/37/669037/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/d373124/ : SUCCESS in 10m 40s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/37/669037/1/check/puppet-openstack-unit-latest-ubuntu-bionic/471ded2/ : SUCCESS in 10m 28s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/37/669037/1/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/8ed002b/ : SUCCESS in 51m 22s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/37/669037/1/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/90b185e/ : SUCCESS in 56m 53s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/37/669037/1/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/3d8c727/ : FAILURE in 26m 23s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus http://logs.openstack.org/37/669037/1/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/4f94514/ : SUCCESS in 51m 41s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/37/669037/1/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/830a2a8/ : FAILURE in 10m 17s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/37/669037/1/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/1f2fced/ : FAILURE in 9m 23s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/37/669037/1/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/b46940a/ : FAILURE in 11m 43s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/37/669037/1/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/ad9964d/ : FAILURE in 11m 00s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/37/669037/1/check/puppet-openstack-beaker-centos-7/ee4014b/ : FAILURE in 21m 12s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/37/669037/1/check/puppet-openstack-beaker-ubuntu-bionic/8ceda7e/ : FAILURE in 14m 43s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/37/669037/1/check/tripleo-puppet-ci-centos-7-undercloud-containers/061c228/ : SUCCESS in 1h 30m 05s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/37/669037/1/check/build-openstack-releasenotes/a501200/html/ : SUCCESS in 4m 06s\n- puppet-glance-tripleo-standalone http://logs.openstack.org/37/669037/1/check/puppet-glance-tripleo-standalone/99f1fe3/ : SUCCESS in 1h 28m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dbe6650ff7bd82b4fac5fb4e249329ef53ec75b1","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-07-04 06:08:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"780115eb1da39d4edfd01ccb060c5c70ba418756","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-04 06:48:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/37/669037/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/3d95b88/ : SUCCESS in 33m 08s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/37/669037/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/b698652/ : SUCCESS in 37m 06s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/37/669037/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/6ee069b/ : SUCCESS in 37m 55s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/37/669037/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/341ee7a/ : SUCCESS in 34m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6463a1716d957544ce5db2dd8b1275faa5946dc3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 08:08:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/37/669037/1/check/puppet-openstack-lint-ubuntu-bionic/59832db/ : SUCCESS in 4m 51s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/37/669037/1/check/puppet-openstack-syntax-5-ubuntu-bionic/f788938/ : SUCCESS in 4m 20s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/37/669037/1/check/puppet-openstack-syntax-6-ubuntu-bionic/1090206/ : SUCCESS in 5m 17s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/37/669037/1/check/puppet-openstack-unit-5.5-centos-7/6ac8fe9/ : SUCCESS in 13m 26s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/37/669037/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/b81a022/ : SUCCESS in 13m 20s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/37/669037/1/check/puppet-openstack-unit-latest-ubuntu-bionic/3f535dd/ : SUCCESS in 14m 10s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/37/669037/1/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/2dcad0a/ : SUCCESS in 51m 14s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/37/669037/1/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/9158118/ : SUCCESS in 1h 06m 16s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/37/669037/1/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/e779d67/ : SUCCESS in 1h 02m 39s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus http://logs.openstack.org/37/669037/1/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/7409a98/ : SUCCESS in 45m 42s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/37/669037/1/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/62d5ad3/ : FAILURE in 4m 16s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/37/669037/1/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/ae01227/ : FAILURE in 4m 23s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/37/669037/1/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/eef54e6/ : FAILURE in 5m 34s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/37/669037/1/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/501697b/ : FAILURE in 6m 04s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/37/669037/1/check/puppet-openstack-beaker-centos-7/500c2cb/ : SUCCESS in 26m 25s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/37/669037/1/check/puppet-openstack-beaker-ubuntu-bionic/a3a795f/ : FAILURE in 9m 28s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/37/669037/1/check/tripleo-puppet-ci-centos-7-undercloud-containers/bfebc89/ : SUCCESS in 1h 50m 43s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/37/669037/1/check/build-openstack-releasenotes/fbeabba/html/ : SUCCESS in 3m 31s\n- puppet-glance-tripleo-standalone http://logs.openstack.org/37/669037/1/check/puppet-glance-tripleo-standalone/8dc1994/ : SUCCESS in 1h 35m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f80ff109b409312f4590a368809702662a7290ad","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-04 13:59:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e51e4838390275269bcf3bb13663dc6f3f8d4ef0","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-07-05 00:51:34.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"04c6a7f6d2d0b4915f16d6057842cb38a0cea387","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 00:51:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8f379fbcfe1c7e96de6603fe811edc1eae0a0505","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 02:35:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f04606dfc653f9986fccc07431dab542c27d2fb5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 02:35:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/37/669037/1/gate/puppet-openstack-lint-ubuntu-bionic/7f9cadf/ : SUCCESS in 4m 07s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/37/669037/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/7b8e48d/ : SUCCESS in 4m 31s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/37/669037/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/99f0033/ : SUCCESS in 3m 59s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/37/669037/1/gate/puppet-openstack-unit-5.5-centos-7/93e3e6c/ : SUCCESS in 12m 18s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/37/669037/1/gate/puppet-openstack-unit-6.4-ubuntu-bionic/839c963/ : SUCCESS in 11m 27s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/37/669037/1/gate/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/71aaffc/ : SUCCESS in 51m 46s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/37/669037/1/gate/puppet-openstack-integration-5-scenario002-tempest-centos-7/a772b5a/ : SUCCESS in 1h 02m 01s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/37/669037/1/gate/puppet-openstack-integration-5-scenario003-tempest-centos-7/b7a3e03/ : SUCCESS in 1h 11m 54s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus http://logs.openstack.org/37/669037/1/gate/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/01edb7e/ : SUCCESS in 43m 09s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/37/669037/1/gate/puppet-openstack-beaker-centos-7/99d64f9/ : SUCCESS in 17m 50s\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/37/669037/1/gate/tripleo-puppet-ci-centos-7-undercloud-containers/89e5c4e/ : SUCCESS in 1h 37m 56s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/37/669037/1/gate/build-openstack-releasenotes/c23939d/html/ : SUCCESS in 3m 15s\n- puppet-glance-tripleo-standalone http://logs.openstack.org/37/669037/1/gate/puppet-glance-tripleo-standalone/86873c4/ : SUCCESS in 1h 32m 57s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"643f342304f6ed704202f065a9794e397d12e5b9","revisions":{"643f342304f6ed704202f065a9794e397d12e5b9":{"kind":"REWORK","_number":1,"created":"2019-07-04 03:44:21.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/37/669037/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/37/669037/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/37/669037/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/37/669037/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/37/669037/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/37/669037/1"}}},"commit":{"parents":[{"commit":"c5a4e73193fa63403036d54deb78891051871509","subject":"Prepare Train M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/c5a4e73193fa63403036d54deb78891051871509"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-07-04 03:34:42.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-07-04 03:36:14.000000000","tz":480},"subject":"Manage oslo_middleware/max_request_body_size","message":"Manage oslo_middleware/max_request_body_size\n\nSo that we can increase it from the default 114688\n\nUseful in case for example the OS-Federation mapping is too large.\n\nIf this limit is breached glance will return a 413 Entity Too Large\nand not log anything to glance.log.\n\nChange-Id: If15f7cc9cef9a3bca855a3cd02068d45eedd171f\nCloses-Bug: #1835161\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/643f342304f6ed704202f065a9794e397d12e5b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/643f342304f6ed704202f065a9794e397d12e5b9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
