)]}'
{"id":"openstack%2Fpuppet-glance~676308","triplet_id":"openstack%2Fpuppet-glance~master~I0516f2fa0604df84dd3c2c641597a0ea184e9efb","project":"openstack/puppet-glance","branch":"master","topic":"dev","hashtags":[],"change_id":"I0516f2fa0604df84dd3c2c641597a0ea184e9efb","subject":"Add bugs url link to README","status":"ABANDONED","created":"2019-08-14 02:22:24.000000000","updated":"2019-08-15 08:28:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"198774d60951f4cfcdf467524eff8bad3eabf0c8","_number":676308,"virtual_id_number":676308,"owner":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":1,"date":"2019-08-14 02:27:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":-1,"date":"2019-08-15 08:04:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-14 02:27:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-15 08:04:47.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"}],"messages":[{"id":"5deaa9ccca664a8a7b2e97475f2f590b79e2059d","author":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"date":"2019-08-14 02:22:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e89dcac9e16e7908543254a8405ac0fc4fd74c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 02:27:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bcbac2547b2642aa8057cf0f3745b2f7 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"25c9903001c4697d4c0b3e7ebe0152e47b7eafd0","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-15 08:04:47.000000000","message":"Patch Set 1: Code-Review-1\n\nWe shouldn\u0027t do this, we are planning to move to StoryBoard which would require us to change these all over again.","accounts_in_message":[],"_revision_number":1},{"id":"b444c1c49e3f3e10188f78ed89fc358e1c7c8b40","author":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"date":"2019-08-15 08:28:11.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e0a73b6f316084de6293f4f0fdf3a602cd3aabb5","revisions":{"e0a73b6f316084de6293f4f0fdf3a602cd3aabb5":{"kind":"REWORK","_number":1,"created":"2019-08-14 02:22:24.000000000","uploader":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"ref":"refs/changes/08/676308/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/08/676308/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/08/676308/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/08/676308/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/08/676308/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/08/676308/1"}}},"commit":{"parents":[{"commit":"17fee69289fe607718b96bcc6e94ed0a6d14d06b","subject":"Bump the openstackdocstheme extension to 1.20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/17fee69289fe607718b96bcc6e94ed0a6d14d06b"}]}],"author":{"name":"ShangXiao","email":"shangxiaobj@inspur.com","date":"2019-08-14 02:20:57.000000000","tz":-420},"committer":{"name":"ShangXiao","email":"shangxiaobj@inspur.com","date":"2019-08-14 02:22:17.000000000","tz":-420},"subject":"Add bugs url link to README","message":"Add bugs url link to README\n\nChange-Id: I0516f2fa0604df84dd3c2c641597a0ea184e9efb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e0a73b6f316084de6293f4f0fdf3a602cd3aabb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e0a73b6f316084de6293f4f0fdf3a602cd3aabb5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
