)]}'
{"id":"openstack%2Fpuppet-glance~676564","triplet_id":"openstack%2Fpuppet-glance~master~Ib692f55fa267e9fbe17d94c5116f244be02b2107","project":"openstack/puppet-glance","branch":"master","topic":"remove_deprecate_pki_related_options","hashtags":[],"change_id":"Ib692f55fa267e9fbe17d94c5116f244be02b2107","subject":"Remove deprecated pki related options","status":"MERGED","created":"2019-08-15 03:52:34.000000000","updated":"2019-08-16 09:49:30.000000000","submitted":"2019-08-16 09:49:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676564-1565948970171-32243520","meta_rev_id":"4a3d00d6d04653e6022085087e2b443b6cdc2c14","_number":676564,"virtual_id_number":676564,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-08-16 09:49:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-15 05:27:52.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2019-08-15 16:21:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-16 08:15:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-08-16 08:15:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-15 05:27:52.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-15 16:21:38.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-16 08:15:35.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-08-16 09:49:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a7195bebc147de14eb9f70b4c43edf71d280d9f2","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-15 03:52:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"451d1f83940f2b2a3d51f2a4669fb09d3abbf422","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-15 05:27:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/64/676564/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/a5441bf/ : SUCCESS in 46m 52s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/64/676564/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/b74f1ae/ : SUCCESS in 1h 31m 44s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/64/676564/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/4a95090/ : SUCCESS in 1h 31m 07s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/64/676564/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/77398ea/ : SUCCESS in 40m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"11f9d795884a2387e3c0009a761f914c13dd7247","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 05:40:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f450134c43ba4a64b1372080f33b6452 : SUCCESS in 5m 06s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/67449e602c3b4b83a98c39ad86d6330d : SUCCESS in 5m 05s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb8379e9474a4a9fb4aea2c88c830b02 : SUCCESS in 4m 13s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/242aa927a6604746ae69fc96e69ccad6 : SUCCESS in 20m 30s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/26dc8ce429e4486aa275726a765b5bb0 : SUCCESS in 13m 04s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b1b1f53da1ff45d7b4148ba73e1c588f : SUCCESS in 13m 28s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/a2baefcd15b44463bf7607e68e21546f : SUCCESS in 52m 07s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/7920a9bb10134ad9a0ed92c6b9d77b52 : SUCCESS in 56m 46s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/b1c74e5b2edc492d8d972b998a3d2c52 : SUCCESS in 53m 53s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/df7c4b44aaf746e0be2df902a514b0c0 : SUCCESS in 41m 24s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/5596d48d2d57452d9b0dc68cbd242583 : FAILURE in 32m 39s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7cdac8d6cfe4d00bc3b917413a763a6 : FAILURE in 28m 34s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3cd0418e3e8d43d4ac3ea06c0eab0ef6 : FAILURE in 25m 27s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/a3e82ae6ede843948f8ae76f9b5cb402 : FAILURE in 22m 53s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/40e77d611dfb4662828d5327c00085fb : SUCCESS in 17m 23s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ff2a5fbd48c4ffb8f5923b0200a1dd3 : FAILURE in 7m 46s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/956cb5904943456cb527ab51101896ef : SUCCESS in 1h 42m 05s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c462e4a4270f48368bca0e115f95b83c : SUCCESS in 3m 56s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b0eadb3c84bf493eb0f421c7cbc78847 : SUCCESS in 1h 26m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0b8b287f8b9dea21f0da2bcf3b5d387271659dfe","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-15 16:21:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a9586687b9e69f2111ac8d6afc43d2d3466002bd","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-08-16 08:15:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0c0615f188264a662d4df33f165eb08f38337a54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 08:15:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3fe6773caf43e52ca4aa4aa16827d5558c191ff8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 09:49:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3090a93b7e804cd39ad94725af3e9a47 : SUCCESS in 4m 20s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/46daa6703471455d9b6c9991dcf04239 : SUCCESS in 4m 29s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/117d5b3d07fb402f9963142d640de322 : SUCCESS in 5m 00s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/de2181b12bf8425f81a758b559fd4533 : SUCCESS in 12m 42s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/080e656cbea34fb094a6a9c508463ba9 : SUCCESS in 8m 59s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/271c7fc3d2e14510adf41193e300c22e : SUCCESS in 1h 01m 51s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/9d024cde04ff4502b39d05a33e3041d2 : SUCCESS in 1h 03m 05s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/42282f18e99d4050bc30498ebf140f3d : SUCCESS in 56m 31s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/7d06847a37fc4e88af61431e56755bc5 : SUCCESS in 45m 43s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/22e82772364f427e82c4e3721e0107ac : SUCCESS in 17m 25s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/9193f762e07442cf8cc6e29278b38e0d : SUCCESS in 1h 25m 14s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/347ca75a7dd1424ab399d17aeb96571b : SUCCESS in 5m 15s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/93fdbe18912e4d11b0d670c2f3cf436b : SUCCESS in 1h 25m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dac385d5cadc505fe5bf85c636672f1b66cd3bcf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 09:49:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e82d93c1c0c0696fc1644a3b16cc1817b0f585a9","revisions":{"e82d93c1c0c0696fc1644a3b16cc1817b0f585a9":{"kind":"REWORK","_number":1,"created":"2019-08-15 03:52:34.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/64/676564/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/64/676564/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/64/676564/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/64/676564/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/64/676564/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/64/676564/1"}}},"commit":{"parents":[{"commit":"17fee69289fe607718b96bcc6e94ed0a6d14d06b","subject":"Bump the openstackdocstheme extension to 1.20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/17fee69289fe607718b96bcc6e94ed0a6d14d06b"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-08-15 03:51:36.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-08-15 03:51:36.000000000","tz":480},"subject":"Remove deprecated pki related options","message":"Remove deprecated pki related options\n\nThe deprecated pki related options check_revocations_for_cached and\nhash_algorithms option has been removed.\n\nChange-Id: Ib692f55fa267e9fbe17d94c5116f244be02b2107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e82d93c1c0c0696fc1644a3b16cc1817b0f585a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e82d93c1c0c0696fc1644a3b16cc1817b0f585a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
