)]}'
{"id":"openstack%2Fpuppet-glance~684917","triplet_id":"openstack%2Fpuppet-glance~master~I5962e42a8e2b56b987db3ddfa9093b67ec13c821","project":"openstack/puppet-glance","branch":"master","topic":"constraints","hashtags":[],"change_id":"I5962e42a8e2b56b987db3ddfa9093b67ec13c821","subject":"Update the constraints url","status":"MERGED","created":"2019-09-26 03:41:05.000000000","updated":"2019-10-16 22:11:27.000000000","submitted":"2019-10-16 22:09:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684917-1571263792478-33a7a8e1","meta_rev_id":"e602f11d43405bc83c3fa2c381850a262e56e1c5","_number":684917,"virtual_id_number":684917,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-10-09 14:02:32.000000000","_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"date":"2019-10-16 15:52:45.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-16 22:09:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-16 16:47:19.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-10 07:22:41.000000000","_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-09-27 01:47:37.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2019-10-09 19:26:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-09-27 01:47:37.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-27 01:47:37.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-10-09 14:02:32.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-10-09 19:26:25.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-10-16 15:52:45.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-10-16 16:47:19.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-16 22:09:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"74c591e1fd9ce93bb395d12b277d686ac0aac475","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-09-26 03:41:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ddc26c6c850311c8cc5367c0d54ad112beba2c34","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-26 04:39:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/5909f58/ : SUCCESS in 40m 57s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/66b300c/ : FAILURE in 44m 20s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/8109279/ : SUCCESS in 41m 43s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/ced47ab/ : SUCCESS in 35m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0ca5727298c65746706fa977245ecce7fda5eee3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 06:43:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/26f2384e369f4457bb427959076e5bdf : SUCCESS in 5m 17s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4dcd569b5d194b72ab0b25d958338f97 : SUCCESS in 4m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d957182cfd3a4056ba78d73dd59153ac : SUCCESS in 5m 34s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/8c945c2e8f894e2cbe21a964f2d51aa2 : SUCCESS in 16m 09s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c4717dc360084576aab21c9655a2ddc9 : SUCCESS in 9m 46s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c371f985657420b918a3f1653bbb009 : SUCCESS in 11m 46s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/2716f9f832dc40558a7fbba8431b8868 : SUCCESS in 51m 02s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/ea3035b484b14deb81c15f2b68863661 : SUCCESS in 1h 03m 19s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/41c03ee5833244f0a4b0c42da7a9c01f : SUCCESS in 1h 04m 44s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/942eda1f4c8b45da92ef9317ea1143d9 : SUCCESS in 52m 32s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/7c8ebe6cd22a4053bb925029d481fc00 : FAILURE in 34m 26s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b23f054fff184184b160e45c53dbf7c9 : FAILURE in 30m 24s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/79b56451bfe54be0aefd40dfd2fa4df5 : FAILURE in 24m 29s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/1bd0dd35b2a24120abe99f93cff0a8c4 : FAILURE in 37m 50s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/b2f02e9e73454612b2d190a2ab3331c4 : FAILURE in 19m 12s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4aed78eddc2e4a40a5c100793f2ea71e : FAILURE in 7m 29s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/68680ea3e4744b9690dfef3b38f3b309 : SUCCESS in 2h 47m 21s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4044f07c9e24212ac37e8699520694f : SUCCESS in 2m 56s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5c4a0146e8a741f8a2996ca0355a36a0 : FAILURE in 2h 13m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8dd7cfd9e92786b5deee1484585a305f38d47fa3","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-27 01:47:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"62f2a7e4b43a2b56ad7b809fc874e1ee9dc9545c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 04:34:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8fd0392171d54cb79b860e4e7b110657 : SUCCESS in 5m 05s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ec3178cee7354bd7b3fd9d7b1e782d56 : SUCCESS in 5m 35s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1bf1eadb64ae47ef8336a2a61fe9e17e : SUCCESS in 5m 50s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/d3cba64d247f4823b0726043c0925808 : SUCCESS in 14m 01s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b28fe8ec7ede4283848366c386e852b9 : SUCCESS in 19m 44s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1380a305f4a848b5885f1b948a4170a3 : SUCCESS in 10m 55s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/8ce1e83b904a4953b55a4c55a3f800f0 : SUCCESS in 51m 05s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/d96211bcff6d4f0e82568e194d9ae68a : SUCCESS in 57m 36s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/551c5088ff7f4f6da4a92efdb0d3a03f : SUCCESS in 1h 13m 57s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/aed4cd28a0ad45a2b64e358edc408786 : SUCCESS in 46m 49s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/a506849907c344e89d29de1a602d3b6e : FAILURE in 35m 35s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e9111c3908d147f891c8e9b5e6ff9a06 : FAILURE in 32m 10s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7502d4d04064ef18bb722d613776ab2 : FAILURE in 27m 30s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/d4119ba77dad495883c90d873e2ce3b5 : FAILURE in 24m 47s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/4d2e1e15138f4b4c8d7e9d53bbd1ce0d : FAILURE in 21m 50s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/352e8f26687042a597ef73e706945002 : FAILURE in 10m 00s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/29fe686e483a4891b873e349116b9396 : SUCCESS in 2h 36m 50s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c7f073f095b48d8967a7fabdc959221 : SUCCESS in 4m 33s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8cad024432b14e31906c6b5b924189b3 : SUCCESS in 2h 17m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"584f3feb03f2534dace1e4e0feaace300af5f91d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-09-27 21:29:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"25d174d8e89aee54d585b699437fbd388963f54f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-27 22:49:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/7310976/ : SUCCESS in 1h 13m 09s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/962b915/ : SUCCESS in 44m 30s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/0b1013d/ : SUCCESS in 1h 18m 27s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/5f4b6d4/ : SUCCESS in 32m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ffd31383b7b179b2f8bdbd95084dd8a0b0050fb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 23:58:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa3cf95c3d0e48ab972110ba79d18f0c : SUCCESS in 3m 31s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/28e54e6129af4dc48d7c143386afa5d6 : SUCCESS in 5m 49s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f7573a53500482fbca07964bf91d831 : SUCCESS in 4m 54s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/9dfc6bb87f7e44d7bb4b2ad0e9f4880c : SUCCESS in 17m 28s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a6ad612c9c5141dcb936144b7fa0e263 : SUCCESS in 9m 56s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5db5a96962e745fb8ea5e515736ebc72 : SUCCESS in 10m 51s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/451cc31693354bc4ba12c4148b9e2790 : SUCCESS in 1h 08m 36s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/c0d51a6349fb442998fc7e03d7a4b58e : SUCCESS in 1h 01m 34s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/854859092d4d43b2993d6b0feb2ebcf8 : SUCCESS in 1h 07m 48s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/9a812a817d564ab9851b570f0955364c : SUCCESS in 42m 33s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/5e8b63a268094e42b1db123f256559d2 : FAILURE in 28m 58s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cdc3a534b0c144d89e10a47cfcc6c06c : FAILURE in 45m 13s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e72678fee9142a6ab00546ee2a7641e : FAILURE in 29m 18s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/5e3b7f4f17df44fca4f05f6b25342e81 : FAILURE in 21m 31s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/0d9c1dcea01d4f8a8cf141df1ba02614 : FAILURE in 20m 35s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f7721f8076a545ffa86b52eb8e74e614 : FAILURE in 9m 16s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/00d17866cd884f41b1aecac33f00688b : SUCCESS in 2h 20m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca266c6181324197a534c22866e895b7 : SUCCESS in 3m 55s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/46cf2d34f24b423aa937e52d9bf20c25 : SUCCESS in 2h 27m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7a5d9f5b5412870a9f59ab00c5a92fba42b09619","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-09-29 01:38:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a38d6c0c189733844c926b2c3af55ea2e1eaacdd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-29 02:25:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/9867cb6/ : SUCCESS in 33m 55s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/05a63b1/ : SUCCESS in 40m 54s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/03f7067/ : SUCCESS in 45m 38s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/89a10e2/ : SUCCESS in 35m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"136832ef34fd7bbe3bf1934d0d9b8a99c95a1b95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-29 04:05:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3ef06827bfdf4a0cb3447fe4f55eeac4 : SUCCESS in 3m 46s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6e1795a971a14c3d9ba811af28a80e67 : SUCCESS in 3m 51s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9af9b431fa9d4b97b6f1c9238744e60e : SUCCESS in 4m 25s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/a4af04df792e4516bf1851cd72329210 : FAILURE in 6m 56s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/40d75fbac8a044b5849472b2972381ec : SUCCESS in 9m 09s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/266437e7b16e4f8ea9120a537c3cc98a : SUCCESS in 8m 56s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/dfaa82ac98c647de807ee5c237c2e692 : SUCCESS in 49m 53s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/4bd7cc6bc16945f7ab2ad2223b650a4c : FAILURE in 1h 11m 12s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/0b45d9d81af14a7a8f803abb7cb66e25 : SUCCESS in 1h 11m 43s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/912def3475ca465b9d6fa10346fac1ce : SUCCESS in 49m 49s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/6f81db5563a94d5ebcdb55e33ddd3b24 : FAILURE in 37m 00s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6d063e01002c44a69a4c7372575671f9 : FAILURE in 33m 42s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/723f96d242514509bb21430675536ae8 : FAILURE in 20m 49s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/72fc3769124c45ef917791f1a547b945 : FAILURE in 27m 54s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/0783ef8070f74e7aa08e571ba29f7122 : FAILURE in 9m 45s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1dfb059e62c742dcba744a892371f051 : FAILURE in 9m 27s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/6f66ec4b144241c48425127e89cf268c : SUCCESS in 2h 19m 17s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8193ebbb30740c992e4a35837f57d99 : SUCCESS in 3m 09s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/62fa74f43cb94cdf958663f363baa932 : SUCCESS in 2h 09m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8a257030442caa6bed256126ec1b2b69c1479014","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-09-29 06:34:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ee62116c25f21a785c8c2347b4a7a9ad04221099","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-29 07:21:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/cbad65a/ : SUCCESS in 34m 43s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/8014da2/ : SUCCESS in 46m 07s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/943eb02/ : FAILURE in 41m 30s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/c230e0d/ : SUCCESS in 35m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"90eea579600db35068ddb11b70862bcbca3b582c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-29 08:43:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/026d9b2e9960464fad277f69f8633531 : SUCCESS in 4m 11s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d58e5e2c45834c209966dc9c9a6ddf8b : SUCCESS in 3m 52s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/66b8c7dc111240e292d406bf0e0bebfa : SUCCESS in 4m 05s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/33b34bb4dab4445bac3c99aaa75a30d6 : SUCCESS in 19m 57s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6351e2a4cd58428b825211ddebdd94e2 : SUCCESS in 11m 14s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f682336cf80f4fa5a30a67a8a117fdc6 : SUCCESS in 8m 57s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/f0e9a38c80c845539def24adf29642b8 : SUCCESS in 55m 20s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/a206a38c6b2444fca0876e8423e5467a : FAILURE in 7m 41s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/a0ff706f22c74ce48c6df0c13c50378f : FAILURE in 18m 53s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/77ea6a188e00490d9da23d14dec789c0 : SUCCESS in 48m 14s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/18ce073b95454d10b24303f64ff0ef4d : FAILURE in 31m 35s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0172a13948ab47029920b459345a37e7 : FAILURE in 31m 22s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/063954360fd248208e080c0cdf4f17b1 : FAILURE in 25m 17s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/d4d3a6e69a814633b64e48f7ea38deac : FAILURE in 22m 14s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ca2fc837afb0460fba1402f5e96ff005 : FAILURE in 7m 16s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/139484879b3740b89cd2be932bfb1463 : FAILURE in 8m 31s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/5ab408cec5e0420ba68bf42339ba5bff : SUCCESS in 2h 00m 06s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0ae5ea5edfd4650812a118d4b0419c8 : SUCCESS in 3m 16s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/bf12955601934d40a4b8eb2d13eb7f98 : SUCCESS in 2h 04m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2862be7f696a5083017ddad302c793c8687943f5","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-09-30 06:16:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2ddf13c86fbbdb0e85c076bdd26f7f16e7238615","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-30 07:04:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/53addf3/ : SUCCESS in 35m 14s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/0977727/ : FAILURE in 45m 49s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/3b2edeb/ : FAILURE in 46m 06s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/7385ec5/ : SUCCESS in 37m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"40f6a534851ac17f9b26cbc4b1f0df350f6ebd05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 09:08:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d3443559cce847e59b3b5aed09aa41af : SUCCESS in 4m 06s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/736ec69049a54dd08ae2d0f51b49e12e : SUCCESS in 5m 21s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a2a920a1e85c453d9fdad512df6f29b4 : SUCCESS in 3m 50s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/10ebfa231792498bbbdce7a2d6624d29 : FAILURE in 9m 37s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3bc5eb022596457a93ffb13724b8a836 : SUCCESS in 11m 52s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7fb6b57710da455082ff266af836e31b : SUCCESS in 13m 09s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/2b3049bde58a4a8492e3552ecaa09ea9 : SUCCESS in 57m 39s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/34772389058a4ee7af10878a7aa74878 : FAILURE in 15m 05s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/70b8cfa7d19843918e7c6951eeb84dce : FAILURE in 14m 47s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/666a11c5a03c49a88f3e89b90bcb94dd : SUCCESS in 43m 59s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/ef92dac393ed4bbd9b5b2571300f525b : FAILURE in 32m 56s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/947f1345505749d7a0b1cc133f69f8d6 : FAILURE in 36m 24s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e48ca47b44694a7d98f26fe730e70461 : FAILURE in 30m 07s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/391b40b7e0bb46b1bb4eef9550c6b6ee : FAILURE in 26m 13s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1cf5aba7230b49c0b0c7a6d9fff61ef1 : FAILURE in 18m 57s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef840cbb58ae4e4e818d2a1df775c213 : FAILURE in 8m 30s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/1aaa4ed07ea746bab79e300b59dfeba5 : SUCCESS in 2h 13m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a773e3e945dd4f72b64ca45618c7f65e : SUCCESS in 4m 11s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/bb189f0873744c16af0a36e307935ee0 : SUCCESS in 2h 45m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae443e78288c30497db0cf314ca3f47a50305ee3","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-09-30 10:15:56.000000000","message":"Patch Set 1:\n\nTests are broken due to https://github.com/openstack/python-openstackclient/commit/f0fafec54f0d0ad5bc241d5d93c98b4d23f94dfc, need to fix like https://review.opendev.org/#/q/topic:osc_train+(status:open+OR+status:merged)","accounts_in_message":[],"_revision_number":1},{"id":"8d7859632c54fdd110d55388cfd3677b51b9260b","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-10-08 01:31:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6f2df74613e46e6a4b6499091e75cd54cef98a15","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-08 02:36:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/39ff1e5/ : SUCCESS in 42m 49s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/b6521e2/ : SUCCESS in 1h 01m 15s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/b9c4b15/ : FAILURE in 1h 03m 33s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/047dcef/ : SUCCESS in 48m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5f11d98d6ca396115f4f038d91bac7638491259a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 03:45:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/16c4774fad7846b695370b5419ccff9f : SUCCESS in 3m 43s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3c6faec595a74f818d1b7ce4f3cde6e4 : SUCCESS in 3m 42s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/289d1a6a2206480ea7d8a2ea7fa9190c : SUCCESS in 3m 45s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/d345d9387a984429aecbd57ecd2f43d7 : SUCCESS in 19m 00s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2eadc9138051499a99292cc163e078ee : SUCCESS in 11m 28s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ca8ca71e0914c699887c3a0a084bc5a : SUCCESS in 12m 24s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/fa67e054319a46729a06e569fe31f643 : SUCCESS in 49m 02s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/fd71ec96fbeb4d98bd5afc5467cc2e32 : SUCCESS in 53m 34s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/0afa8b5141a04e86a9c159063868fb63 : SUCCESS in 1h 03m 24s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/6755b94f377f43f8ad3003af31dccd94 : SUCCESS in 45m 45s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/9488cfff34c5498eb1a0442753f0900f : FAILURE in 29m 03s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/06c906f7797d48959e4176df970a29d3 : FAILURE in 27m 11s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/51d1b04a380a4c22ba67d42dd9fe07b5 : FAILURE in 19m 05s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/ab8b1d55f2d84e3184238391a545b696 : FAILURE in 19m 56s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e1128fa1e78b4592bf677b84d89b6851 : FAILURE in 18m 59s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7df05593a97f4fabb7af701f038ca86e : FAILURE in 6m 54s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/89da13d8ab894e0388735c95fc399d90 : SUCCESS in 2h 07m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1503d76a1553434cba62cff09efd1f5b : SUCCESS in 3m 07s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/329a27191f2c4ddc9d5799ab25052a47 : SUCCESS in 2h 02m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4d4c629bcdd64279ec1a87ca1e2aeaad11efa2ba","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-10-09 14:02:32.000000000","message":"Patch Set 1:\n\n\u003e Tests are broken due to https://github.com/openstack/python-openstackclient/commit/f0fafec54f0d0ad5bc241d5d93c98b4d23f94dfc,\n \u003e need to fix like https://review.opendev.org/#/q/topic:osc_train+(status:open+OR+status:merged)\n\nhttps://review.opendev.org/687573 should fix it","accounts_in_message":[],"_revision_number":1},{"id":"02c800c41f84860510b3ec3d03f2aacbbe4fb8c0","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-10-09 19:26:25.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b5d55aa8c909aaa8f5c0e1a653f754ae396184c7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-09 20:13:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/1eb63d6/ : SUCCESS in 34m 58s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/3570fbf/ : SUCCESS in 45m 05s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/fa11885/ : FAILURE in 44m 46s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/010a94a/ : SUCCESS in 29m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"85246dea54f35b76d03e22211ec620a631a60aae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 21:43:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0df4766968b645e9bd0ff2d6c20c8363 : SUCCESS in 5m 06s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81183caf5b71412c9d5f576762909436 : SUCCESS in 4m 26s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b931171cde9d4c0382421c507fef0818 : SUCCESS in 3m 42s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/395bfa4c3ebf4a7ca0d9e996c3c2eb20 : SUCCESS in 19m 17s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c9f1093d7a04b70b4792ebdf16b9064 : SUCCESS in 12m 00s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/502f87ae33634878a14a6aced0ee986d : SUCCESS in 12m 29s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/80558d6ed58148cda4f269add84e283d : SUCCESS in 55m 54s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/2f5727e101344f23b395f7f9e2a37272 : SUCCESS in 1h 07m 44s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/b1cefb4de95c4aa8bc9928bedceaaa37 : SUCCESS in 1h 04m 52s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/dcb8889de9394b078bb345833bce4348 : SUCCESS in 46m 52s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/bd291d31d801443488568e8f2945eddc : FAILURE in 32m 54s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0cbc9e55dd364d589f6b89a796ece623 : FAILURE in 34m 58s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/de4d1fdcf5a54a54bbc7e0c47bf7881b : FAILURE in 21m 26s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/8b38ddd3eeb248cd92177557ea1d737b : FAILURE in 21m 52s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/38ee7df8119b4536b5b29030ce3e1379 : FAILURE in 19m 55s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ac2778abc9a34ec39ac6e1158c47adbe : FAILURE in 9m 15s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/27b342b7911c4539a5cda90a82127095 : SUCCESS in 2h 11m 42s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b75020ae7b2b4e54b9f1908af36efdf4 : SUCCESS in 3m 13s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e5f3c2dd50044d81b032787f08fc818e : SUCCESS in 2h 02m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b46f4d97c8c1a2da53590d79b1bd16eaf66daced","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-10-10 07:22:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d4b3271e7d35f89f35e15fcfa47dee6980c4982a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-10 08:11:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/c5da6ee/ : SUCCESS in 40m 48s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/daccf85/ : SUCCESS in 47m 01s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/08a5509/ : SUCCESS in 38m 27s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/5863384/ : SUCCESS in 28m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fe3f3f291e997bcb0372286e68e6ec7ca5ab0ac7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 09:36:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/250e30831709475ca493b346a2c070b9 : SUCCESS in 4m 09s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c054f148b36440c3a6f8932aac3abb95 : SUCCESS in 4m 02s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/441dee73364b4e03adb82da146769c25 : SUCCESS in 4m 21s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/3547f8708b304b7b80fa59cb2e6b54a2 : SUCCESS in 19m 54s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/48d0aa6bb20948a382616dae3dbf09d3 : SUCCESS in 11m 34s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/11037d729508411a93788ec72e29ae4c : SUCCESS in 12m 45s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/7fa44e1eeeb1443980b3d247c8932e8c : SUCCESS in 55m 17s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/f6281c4802174ee583bbe68919ad47f6 : SUCCESS in 54m 51s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/1348e564f0d34a818ed91a4fadf5044e : SUCCESS in 1h 03m 21s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/350807966fb24889b6088a41768af63c : SUCCESS in 54m 43s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/bd7f0e0f7263442f9769078605a03de6 : FAILURE in 30m 02s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b8dde0ca0094dadab0fadace45acf6d : FAILURE in 31m 11s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2df680da551549eeb40784c913efd5db : FAILURE in 23m 00s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/996a99c339a047e38989cb601de91b14 : FAILURE in 23m 30s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/2e447e3db493414baa42307213cb0aa0 : FAILURE in 6m 18s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2404726fa0fc49009d3256e1b6b02305 : FAILURE in 7m 41s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/bbc81fb630f4441289d4c24ad7a1166a : SUCCESS in 2h 09m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/adf9db71a72d4d0cbe6ea8ae09d1fcd6 : SUCCESS in 3m 14s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/78d0fd5e091e40edbc9afb90c95a285c : SUCCESS in 1h 50m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c7f37c3566f31a20ba9a9a2f1f9499f130bb9fb7","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-16 11:49:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cdf987e78ce343fca2e6328ab07acb565b30403f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-16 12:37:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/d92508d/ : SUCCESS in 36m 36s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/a79cc50/ : SUCCESS in 39m 44s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/8396e35/ : FAILURE in 46m 07s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/49b679c/ : SUCCESS in 34m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7afb7ef5886bdee5aa927e11c954aacc5e2d2943","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 14:53:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b60bec5fd894c38bf57941df90cf12a : SUCCESS in 4m 59s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c162c50eb551454b8797584c830c58ee : SUCCESS in 4m 14s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3398c1237775474fbad9ef3d2c7ed961 : SUCCESS in 5m 38s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/3ba8c67e4aec4f049d50991e682af4ce : SUCCESS in 17m 43s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bca3f71ed70a4823a2dbf626dc35aa25 : SUCCESS in 12m 58s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/571f618a12574cf5ba1800f67a710cbc : SUCCESS in 12m 03s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/323b5c89344c4b6dad120ee65feabf6e : FAILURE in 1h 09m 46s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/4269402aeb394895a6fe5c614bc96565 : SUCCESS in 1h 03m 32s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/b8e5dc4f28014880aa2034d3a7f7a64c : SUCCESS in 1h 09m 39s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/0aa321ccd0254369879297ae6e937a1f : SUCCESS in 44m 57s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/6a244b76587c464ca52a78bfa41120fd : FAILURE in 27m 25s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b81eef15b4a642429bdd9cff83ed1600 : FAILURE in 41m 57s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f7862d85cf8c47299ca1b528d4517232 : FAILURE in 18m 34s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/6ac62cd59ca44c81999b4b7097db0621 : FAILURE in 38m 29s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/b338588632a840dbb4b580e012277142 : SUCCESS in 16m 42s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1fef366665d04ab3925d331b6c559552 : FAILURE in 8m 53s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/06174ed43e7840a8b625390258f5c25f : SUCCESS in 3h 02m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0b437fb119640e5b018e5e530ae7cef : SUCCESS in 5m 40s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9988617984084d09bf45674b81e08e61 : SUCCESS in 2h 42m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a7e8ac2af7cfae615dbf1a5cdc60185d868bcebc","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-16 15:52:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3d5eb229fa8505c15d17752af16a849813771a99","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-16 16:47:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/17f6384/ : SUCCESS in 39m 28s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/6349a18/ : FAILURE in 52m 45s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/c0ab96e/ : SUCCESS in 39m 20s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/17/684917/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/e77c426/ : SUCCESS in 30m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d1882fb871e3fce2686c521d958af28ef092b880","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 18:01:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/096eb20285e54d1da653d9bcf708a0b0 : SUCCESS in 3m 59s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f475da77f0674444a0449d1bc68d8b8d : SUCCESS in 4m 13s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e1ed154a0414b65946571d973a3a793 : SUCCESS in 3m 58s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/f4513e92920c42e5b537900b6b34536e : SUCCESS in 18m 58s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d0c9fa8ee7484fa48c2ddfac0d257e7d : SUCCESS in 11m 49s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/26753b5b0d854e3ca48ad91100d3332b : SUCCESS in 13m 51s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/b82a333b6de44e60831ebac9ac82ec64 : SUCCESS in 1h 02m 34s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/2ba5be2cd6cc4aeeab17f6489c8e038e : SUCCESS in 1h 00m 53s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/b766e0f715cc4da38e14f5ea4ae4ae15 : SUCCESS in 1h 10m 23s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/1f5199521c6145a499cbc85dad3281c3 : SUCCESS in 40m 56s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/7c9fdabb6367400ea9f1b52551739b4f : FAILURE in 27m 43s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/674bc759eeef4510b7bff589cb3cb7a2 : FAILURE in 28m 51s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/105ae69271164b8e88a11546d0b72539 : FAILURE in 22m 44s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/915a8e7b7910436a89f3bb98c721651f : FAILURE in 23m 21s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/efea288f32374af4852e1fa0228276f9 : SUCCESS in 16m 23s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b9f2ba7077a4550a7e25770abe8a606 : FAILURE in 13m 07s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/a8378ac5987544b1bac8ae2116394603 : SUCCESS in 2h 06m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b052e15c9704df08f4f0facd7da6bb5 : SUCCESS in 4m 08s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ef9a5e0eaae648cc850e8b5a2e5e0fbf : SUCCESS in 1h 55m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2ec7dd6d66a33a833f6b11b53e3ce03fa59ec340","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 18:02:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"75b15e413d1643019c3e2c5b1bb42b237f70071b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 20:00:03.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"22c463e9581d462407e9c36ddd583b3a1a4cb713","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 22:09:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cb5d2299d9f74ae6855b62bb954103d9 : SUCCESS in 6m 41s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b5b89aaea514d38966f674576652a83 : SUCCESS in 5m 29s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e657dc51a0cf4f3fb8c5501839420c20 : SUCCESS in 4m 44s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/1524e19b3f4446eb974a444f74b1c68e : SUCCESS in 21m 07s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b219be68763947b6ad8cbab6198b0118 : SUCCESS in 10m 14s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/2e6a2d9b69de4087847e22c47595a689 : SUCCESS in 58m 57s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/770b95031de045288c863f5403ba3277 : SUCCESS in 1h 09m 43s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/18017056af1f45d2989c139611ca69f5 : SUCCESS in 1h 16m 26s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/7c88ca09148a4f60be30d19804da39eb : SUCCESS in 1h 12m 09s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/217b42d27a384b53b3b1a5de02f6f64f : SUCCESS in 20m 10s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/797f3fefff354819ac152c4c90127c59 : SUCCESS in 1h 45m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16c1578e44bb4dada30ab4cf7e620017 : SUCCESS in 7m 16s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ed79c6cad0a74ac390e880db36e92e1d : FAILURE in 2h 00m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d7a3eec39b2eaedc5416b1c142a10a646e69f4b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 22:09:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"70490f8578322d6b022b2f051c42a346bb3656b2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 22:11:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c075e61f2659429989d30001d0f91a06 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"809ce20cbc36e55379f19dfb2897a2aba6603453","revisions":{"809ce20cbc36e55379f19dfb2897a2aba6603453":{"kind":"REWORK","_number":1,"created":"2019-09-26 03:41:05.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/17/684917/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/17/684917/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/17/684917/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/17/684917/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/17/684917/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/17/684917/1"}}},"commit":{"parents":[{"commit":"5bf2fa317bb4e5cee5aafdafcb10d99926a84743","subject":"Merge \"Add support to configure service_token_roles in authtoken middleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/5bf2fa317bb4e5cee5aafdafcb10d99926a84743"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-26 03:40:36.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-26 03:41:49.000000000","tz":480},"subject":"Update the constraints url","message":"Update the constraints url\n\nFor more detail, see http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: I5962e42a8e2b56b987db3ddfa9093b67ec13c821\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/809ce20cbc36e55379f19dfb2897a2aba6603453"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/809ce20cbc36e55379f19dfb2897a2aba6603453"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
