)]}'
{"id":"openstack%2Fpuppet-glance~686822","triplet_id":"openstack%2Fpuppet-glance~master~I4f3ec9ee7c4147a4e62de961a9532bdc45526f6b","project":"openstack/puppet-glance","branch":"master","topic":"bug/1846808","hashtags":[],"change_id":"I4f3ec9ee7c4147a4e62de961a9532bdc45526f6b","subject":"Update the method of configuring glance backends","status":"MERGED","created":"2019-10-04 18:40:51.000000000","updated":"2019-10-23 23:17:50.000000000","submitted":"2019-10-23 23:16:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":37,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686822-1571872598492-c8ddf85e","meta_rev_id":"e0a966b3b7602b2daf1da71c7bf70a81c6d08486","_number":686822,"virtual_id_number":686822,"owner":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"date":"2019-10-22 23:19:36.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-23 23:16:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"date":"2019-10-23 20:47:05.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-10-22 13:51:00.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-10-21 14:21:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-10-22 13:51:00.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-07 15:44:08.000000000","updated_by":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"reviewer":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"state":"REVIEWER"},{"updated":"2019-10-14 08:53:45.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-10-21 14:21:29.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-10-22 13:51:00.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-10-23 20:47:05.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-23 23:16:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"53422f64a7b873de5c74bac767e5bcde879af11d","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2019-10-04 18:40:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"53cdf5ad07a81f42f1bcbc490c77a736641d6380","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2019-10-04 18:51:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3584035c6342ace95ad1fc89d5c39da34f6d8efe","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-04 19:36:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/22/686822/2/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/5941df0/ : FAILURE in 5m 12s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/22/686822/2/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/167612b/ : SUCCESS in 40m 42s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/22/686822/2/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/a039383/ : FAILURE in 43m 27s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/22/686822/2/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/94a1b78/ : SUCCESS in 30m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"19747e2097e10daeefc9f08ace08499756b79b42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 22:51:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bea27a9929744705a91865cef174773f : FAILURE in 4m 39s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/71f1af92176f49518cbab1fa6c28dcb8 : SUCCESS in 4m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/afae07fefa4648809007f5a6e60582f8 : SUCCESS in 5m 18s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/11dc3c2798724bbeb5256aa86dc5aa60 : SUCCESS in 14m 30s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c84e8bdbc2cb455d851c7050bc0c147e : SUCCESS in 11m 38s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0ab940fbb245466481fdafe989f952b9 : SUCCESS in 10m 20s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/80f2e8ddfd244df5b085cc2f1681c88f : FAILURE in 1h 03m 31s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/fd75222c4fc64cca85a555022c038c63 : SUCCESS in 55m 26s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/0c23e405c0ba4624801741cf5123d3f7 : SUCCESS in 1h 18m 06s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/bdcbeb647d2f44c4be3ba3cf6e6c9d37 : SUCCESS in 50m 07s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/274b6509c8d64234a9a6f82aec6e0ffc : FAILURE in 33m 40s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3cc28812656d49b190daac665b65d4f4 : FAILURE in 38m 23s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/476d400dceee45c3b2b20bd225b95323 : FAILURE in 30m 39s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/97f52bc4b24a48c8bf51c3a99f120b32 : FAILURE in 32m 04s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/09fed99f7267449bb2d5e5fbcb45b5f9 : FAILURE in 23m 53s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4aedf0d336b04c1aa75decbe55539e51 : FAILURE in 9m 41s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/7598982d87be44bea1eaed60652cb8d3 : SUCCESS in 2h 18m 54s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/272596ca753540599fde7ef8c45a2853 : SUCCESS in 3m 10s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c3a719fba809492a85a85762746adf09 : SUCCESS in 2h 08m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f6f8c64f9b02b3c25a8997cf1894db05ca1d44c7","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2019-10-07 19:15:45.000000000","message":"Patch Set 2:\n\nI\u0027ll be posting an update to fix the lint issue.\n\nThe puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous failure is unrelated, but should be fixed by [1]\n\n[1] https://review.rdoproject.org/r/22999","accounts_in_message":[],"_revision_number":2},{"id":"73c5579ed785dc1023ff71532559442d36dc4e2d","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2019-10-07 19:20:14.000000000","message":"Patch Set 2:\n\n\u003e I\u0027ll be posting an update to fix the lint issue.\n \u003e \n \u003e The puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous\n \u003e failure is unrelated, but should be fixed by [1]\n \u003e \n \u003e [1] https://review.rdoproject.org/r/22999\n\nCorrection: puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous failure looks spurious. The puppet-openstack-beaker-centos-7 failure was due to missing oslo-reports RPM, fixed by [1]","accounts_in_message":[],"_revision_number":2},{"id":"33293d38d2a9bfce37a4565bbda709f57d809a5a","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2019-10-11 12:59:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8a5e35c939e7e0acc46939242a03a40a620fed7c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-11 13:49:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/22/686822/3/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/71e7da7/ : FAILURE in 5m 09s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/22/686822/3/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/4da5146/ : SUCCESS in 43m 02s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/22/686822/3/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/51b5171/ : SUCCESS in 48m 36s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/22/686822/3/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/14e2222/ : SUCCESS in 30m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e9af69b4e7e4408c7f3df1152a0b741982178f5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 15:07:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b7d665c9ba64343820f8b8d76160ac4 : FAILURE in 6m 06s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e5eee18080444eec995ef351501dcabf : SUCCESS in 5m 49s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c7179a658ffd40c18c30e79f69cda442 : SUCCESS in 4m 36s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/853493aa7f6345d597b835686351f886 : SUCCESS in 16m 03s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5faf82f167a6451389cad62468cf9d51 : SUCCESS in 12m 25s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/90b73cd8f019468cb16626576b2356b1 : SUCCESS in 12m 40s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/0c42df0576aa424dbaa9d90f92f0fce4 : SUCCESS in 55m 45s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/c86bd89e00e341bb84623fe8b1e915cc : SUCCESS in 1h 03m 49s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/7b5cd8b5e8da40e885e771d1fe6c01cb : SUCCESS in 1h 09m 46s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/1f334425b3a048f9a05893288fbc896d : SUCCESS in 46m 11s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/5892e55bd6a34b8e8b787f4cb9d8e0e2 : FAILURE in 31m 17s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5f91624dad194a03803ebe64228971f0 : FAILURE in 30m 42s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/625bf1f3685947dbba768308a96d54ce : FAILURE in 19m 58s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/4fb2d92e334a49498acd6f8db6b4157e : FAILURE in 22m 27s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/68909d609cc142968b28e3071d77ee14 : SUCCESS in 20m 58s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d5c80e81624745558fca49cdb0e88807 : FAILURE in 7m 32s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/dfeb6bcfe1b347fd929851cd6b0db9b8 : SUCCESS in 2h 03m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b31ff2dbf4ac45d095ea24a478d6bf98 : SUCCESS in 3m 14s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/eef81cd9c15149e4ac6cd8943414d7bb : SUCCESS in 2h 06m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f28f1746f78feda752b2538bfd86d0d66846d7a2","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2019-10-11 15:33:34.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ef16fe9d89962e94e62e82d7658414c9c8343c1b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-11 16:17:20.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/22/686822/4/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/53bc8f4/ : FAILURE in 5m 30s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/22/686822/4/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/e894a1b/ : SUCCESS in 41m 36s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/22/686822/4/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/b9f8a5e/ : SUCCESS in 41m 25s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/22/686822/4/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/fa73238/ : SUCCESS in 32m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"730aee378520e388884fcaf999a35d12d5ef1893","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 17:50:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/50127daec4e844b8aa78180d4b74c49c : SUCCESS in 4m 48s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ff8121d51e14232863e104e1d87c91b : SUCCESS in 4m 41s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cd3e18371d4d4f2e8785de417c54cb38 : SUCCESS in 5m 16s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/afde593432b8440c9dc59fc6d4366249 : SUCCESS in 18m 11s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8ffb19705b754eaebe7f630372192351 : SUCCESS in 15m 49s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f2bb5d2e566e4809a31bc902bc9b4cbb : SUCCESS in 15m 17s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/8d90853acb96439abbeabaf63cc1e23e : SUCCESS in 52m 23s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/f4ce01858e6a4a348b792d1c813e3347 : SUCCESS in 58m 28s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/7b43ad6e7fbd403094d149041299bc53 : FAILURE in 1h 17m 28s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/44006b9431b24283a3b859f82dd87c8b : SUCCESS in 53m 41s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/a49ebf96ed0e4d3f818a6475315cabfc : FAILURE in 49m 15s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15cfcf9a478e4c5cbff08a29b2506edb : FAILURE in 49m 16s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e08c9603ea5e4dd09eb5a429236435fc : FAILURE in 37m 37s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/60a3475bf0174d6a92ac0cf12f6df5a9 : FAILURE in 37m 27s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/30d858c4966544d0ae7a44705725a939 : SUCCESS in 22m 17s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e9108cac57fc46078489b41fd4eb2a58 : FAILURE in 13m 55s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/d797deaa14b847ee9fca58c7342c772f : SUCCESS in 2h 11m 50s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/499debb4d6524eb89c1aaf2cd27757b6 : SUCCESS in 4m 21s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e5c4131bffaa44d69cde621bdc90f926 : SUCCESS in 1h 59m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8de0f966bb10aa49935ef4d4e76652b66385f776","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2019-10-11 18:31:09.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"1d2c59be629b1b3f2675860913b3a0826179ab39","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-11 19:23:34.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/22/686822/4/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/e78c82d/ : FAILURE in 5m 47s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/22/686822/4/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/39bcf49/ : FAILURE in 50m 25s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/22/686822/4/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/8294d48/ : SUCCESS in 47m 16s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/22/686822/4/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/a0783db/ : SUCCESS in 39m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"824b48656df68160553d4e3480c2276415d127fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 20:43:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cd9400edd8ad429289395759a624a268 : SUCCESS in 4m 25s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4a18d2a959244cf3936a27e5335df38a : SUCCESS in 5m 24s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c26a28f8021a450999c2aa2b8e10d546 : SUCCESS in 4m 27s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/e36556a55e2f423895068d336c5bccf6 : SUCCESS in 21m 16s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/79c4e75d46094d7283cd5f51ddceed3d : SUCCESS in 13m 16s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/990fc296cc6842f99cf26a07546045db : SUCCESS in 14m 53s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/31e32a3912b847d88d431a7995f26c46 : SUCCESS in 56m 00s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/e0ed10124dda48798f01f163dcafd7b3 : SUCCESS in 1h 15m 47s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/75e0ea592e7a4e4eb4fbff6384c08014 : SUCCESS in 1h 29m 11s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/63d990887a694d5aabcbff25da38a291 : SUCCESS in 43m 43s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/17f3dbf64b6b4c8385250e6c77eb94b1 : FAILURE in 29m 58s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0385bb29687f4fe5972ba27a9ad6d254 : FAILURE in 32m 09s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/026c681d8a394e90b3faae54d2f5a9d1 : FAILURE in 33m 29s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/f30ad8a624e54df2a34aba68bd5d2e03 : FAILURE in 35m 29s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e31c7238542b468c944fb270e4dbc2a8 : SUCCESS in 23m 52s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d012faf14fc24d6d96c627ab5ece0087 : FAILURE in 9m 58s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/dc0483261b214c3aa1f1ef94c095f3b4 : SUCCESS in 2h 09m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0b7d89c4ac84e7aaaa76ce3de48668f : SUCCESS in 4m 10s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3e053b5fe141474c94bef656d50b0166 : SUCCESS in 1h 56m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"305a4ab2df6eec9afa9215a181189a109debe82c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-10-11 22:16:08.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"31e633d5f1df91bdb410fe1982fd4478ec836092","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-14 08:53:45.000000000","message":"Patch Set 4: Code-Review+2\n\n(17 comments)\n\nnice work! just nits that u can fix if you need to push another fix somewhere, im fine either way","accounts_in_message":[],"_revision_number":4},{"id":"cd449751eb82f7a96a4c08bd1def53973e82d9d9","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2019-10-15 15:14:15.000000000","message":"Patch Set 4:\n\n(18 comments)","accounts_in_message":[],"_revision_number":4},{"id":"66f880e39a15342ac5ebf3ad902b57d402ea2e70","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2019-10-15 15:14:24.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8736db1fba89f0c7d016b28fe373a663900a61c2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-15 16:12:51.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/22/686822/5/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/d3c4e78/ : FAILURE in 5m 22s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/22/686822/5/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/7478507/ : SUCCESS in 56m 52s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/22/686822/5/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/391c2a5/ : FAILURE in 46m 52s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/22/686822/5/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/336fa16/ : SUCCESS in 38m 48s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d3286db668cc3ebc11d565765dc0c174787e6823","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2019-10-15 17:27:25.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"220136116ea4ae4283ca65e7da24cd9b014c8893","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-15 18:14:18.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/22/686822/6/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/e3e985e/ : FAILURE in 5m 09s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/22/686822/6/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/e9c5dbf/ : SUCCESS in 45m 07s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/22/686822/6/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/fdbe5ab/ : FAILURE in 41m 11s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/22/686822/6/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/1fd12be/ : SUCCESS in 34m 10s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a23910dcc8c8e2a1ab3280139a66270485d90dc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 18:41:07.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b39555e7cb3b4381a42ab84357323824 : SUCCESS in 6m 01s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7403b559404e459481405865af8520c8 : SUCCESS in 4m 20s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cb1913e1a3504267aab4eacfe1aa90c9 : SUCCESS in 3m 56s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/6bb9036c1f1d414bb16562a5d0c42ac6 : FAILURE in 22m 46s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fd61218e12c44cf8b2cc98086c476d1f : FAILURE in 15m 23s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc2e24ca08704b8c971b2c00a81970ca : FAILURE in 17m 03s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/57765b8b33cf4833b8f347374c0bc18d : SUCCESS in 55m 21s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/281c3d7648504e789a5a1424fa854784 : SUCCESS in 1h 09m 32s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/863aabcb70d14fc6b1db5562de85e277 : SUCCESS in 1h 07m 05s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/aa9837829e63446a8d7ceca60aaab5c6 : FAILURE in 56m 02s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/97ae6d6293ba4f76ab31e6be464c0290 : FAILURE in 32m 09s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f1b630213ea424d8e5e8ed8bfcad15d : FAILURE in 32m 35s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/325b167161c14424bcebf299148ba3d9 : FAILURE in 21m 15s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/31793d14a8bf43a49f7b1b4a44ca60da : FAILURE in 22m 20s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9674d08e36474cc49d3e88e420d690ba : SUCCESS in 21m 49s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6e5f31ebea51484897d79964f95fc4e4 : FAILURE in 11m 19s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/ff2039ca1efa492fb8916d0391e66e7a : FAILURE in 37m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47fc045df94e4d4fb108eacda03c3103 : SUCCESS in 3m 38s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7ecdeea24e834324a1ad157faf78ad28 : FAILURE in 30m 55s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f81aa1380eded6ea7efed7add78bbe50127dc921","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2019-10-16 21:46:46.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"52b25d96e67ca284ae868e4fc7d294c9af3fddfb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-16 22:33:23.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/22/686822/7/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/9267732/ : SUCCESS in 40m 28s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/22/686822/7/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/0c6a56a/ : SUCCESS in 44m 54s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/22/686822/7/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/f8601fa/ : SUCCESS in 43m 40s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/22/686822/7/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/828b863/ : SUCCESS in 34m 35s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"cf8347a3522d5e7f6eeb250f21a7b032a7a29ec1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 23:56:58.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8caf5dc91aec486a948cc58cd5a0f470 : SUCCESS in 5m 18s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95b71da644f44010a8e02a9a8036b362 : SUCCESS in 4m 35s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/423ca9e78ef1441e90629e3a8f224580 : SUCCESS in 4m 15s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/d9603124c2fa4b99a81579a2928f840e : SUCCESS in 21m 51s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e2363017ca564068bfe6e59cc0d73fad : SUCCESS in 11m 24s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e74faad2e9ac4d72952ae69b330b69f1 : SUCCESS in 11m 57s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/7494c92c097b4a0ea3f02180d25d2cd1 : SUCCESS in 56m 23s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/fc7fac25a56547d9ae2ba244ee61adf9 : SUCCESS in 1h 16m 25s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/29ef921b277446f8972079a2ddc8dd0b : SUCCESS in 1h 20m 12s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/cc04d5016bc34981ac97ba17d29a1c4d : SUCCESS in 51m 29s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/c871f4a265104d97b35d5739066f5f89 : FAILURE in 31m 23s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/526780a491834f0880e4b3fa82f7f361 : FAILURE in 30m 44s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7084064fed804ff6a095e2394ed5f170 : FAILURE in 21m 34s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/88aaaffdec8642d199f7411301f6af36 : FAILURE in 25m 45s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/40980c6115354a59a41c2b17516fa8e9 : SUCCESS in 20m 33s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e582f59dcaf47cf9702197eb6b7d18f : FAILURE in 7m 57s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/da44379fecc84fe795e6662de61fe59c : SUCCESS in 1h 53m 14s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6c23c70ba8d4daaaa0d305a4f90e0af : SUCCESS in 3m 31s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/61fd745e250f4cd3b9f2b7d77285acc8 : SUCCESS in 1h 57m 10s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"dc0b77f2ef8199c98b4c8cfccf94ec7ebcfa2694","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-10-21 14:21:29.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"1163f31983623ec7dbc0ea8e9e17f977ea222f68","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-10-22 13:51:00.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"039540810e800d73788e524b9b25221a0f95a554","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 13:51:12.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"16da480037b76262c6ca8ed4ba91dc331c904b64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 15:41:31.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c0f756ae33af463980074ecf4c9dd09b : SUCCESS in 4m 56s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ddac4517c6c45f0872b866bd7a4489b : SUCCESS in 4m 22s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b77df4ca80e4ba0bcba3129f12303fa : SUCCESS in 4m 43s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/2fcddbf936234e4b9fdc32a81c714edc : SUCCESS in 16m 53s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fd2984a3baac4c008f944666ba92684c : SUCCESS in 10m 01s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/30d5c4777b694f9d9b5838ffca86713e : SUCCESS in 1h 00m 32s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/da6913b05b634d3e93205eed8fc8af98 : SUCCESS in 1h 05m 30s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/d0ccb583b3094405937c1bd6ffacd1ca : SUCCESS in 1h 13m 08s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/28417f6489c442cc903bd5fe8387e897 : SUCCESS in 49m 59s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a95dff8763fc4b52a69c3d26d6f9ab08 : FAILURE in 18m 50s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/a8f83769196a4956afebbbd0b53b66b2 : SUCCESS in 1h 43m 01s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b6b46d104c9456dbec3c2552a131ae2 : SUCCESS in 3m 40s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a9d92c8cccf04a94bf5ab4ce5451d8f1 : SUCCESS in 1h 48m 14s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"38eaa0639c8fafa3eaadee73f4c8fb1b0f2c47b7","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2019-10-22 23:19:36.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"eaf3514665c2bee793c011588b607ca029c39343","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-23 00:01:39.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/22/686822/7/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/3d4255f/ : SUCCESS in 34m 43s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/22/686822/7/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/b60e902/ : SUCCESS in 39m 54s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/22/686822/7/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/a8867b8/ : SUCCESS in 38m 08s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/22/686822/7/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/f5dc401/ : SUCCESS in 32m 15s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e518a60006e4823b83f4f1675fc469e6e007a8dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 01:11:27.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb43bc75d4a147bdaa9989742a3de58b : SUCCESS in 5m 02s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/70e47211eb714b22aefdd15965a1d42d : SUCCESS in 5m 15s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6e5c4976629843b1a1635f419df196b8 : SUCCESS in 7m 02s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/12803d9b951e40cd8a86bb80e44151f9 : SUCCESS in 19m 31s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f18d28328cc443e90cd7b5779e58aa4 : SUCCESS in 14m 33s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bbd34847386d47708faa7dd2051a2521 : SUCCESS in 12m 29s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/0f53af70ad6f4679bef54f601c65aa25 : SUCCESS in 1h 02m 13s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/af2b3c0cc82a4b05a61494968412fade : SUCCESS in 54m 24s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/866fdcca5c6f4bba9d73935e9a468135 : SUCCESS in 1h 19m 30s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/4183ab95c07343ba80c60c05026a8424 : SUCCESS in 51m 56s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/d1ed648efa3b4b63a85db21630336061 : SUCCESS in 54m 38s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0cafc56908f348438bc52eb38ccd35e4 : SUCCESS in 55m 10s\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b92cf5e2e0e4dff8b015a55b2c932e6 : SUCCESS in 34m 34s\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/3190c427edb04367a35afa2bfb635e8b : FAILURE in 46m 30s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/60af68c0c058425984db562d82fefdd8 : SUCCESS in 22m 16s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/90a7914460a042f8a7d4f186b20fa811 : FAILURE in 10m 42s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/48c6e9b287cd4843a0ede53d7dd58b24 : SUCCESS in 1h 44m 17s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b5b693a6afd46d9a991267e22f0c2dc : SUCCESS in 4m 16s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6cb76cf2494042c2888850154df116c1 : SUCCESS in 1h 25m 47s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e7d2a038cef0c099ea785e17d3955d8cefa6cf38","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-10-23 19:12:42.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"64940d10f4bd998b023bfa1bf28f56bea6e4f25c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-23 20:47:05.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/22/686822/7/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/c813f42/ : FAILURE in 40m 18s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/22/686822/7/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/37959b0/ : SUCCESS in 40m 04s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/22/686822/7/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/da59835/ : FAILURE in 47m 49s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/22/686822/7/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/874de8c/ : SUCCESS in 33m 43s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7bea860874301b0014bc828e5f0e85669fe56df3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 21:02:19.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e10453a0e87548ec9e1f1d2820175e0e : SUCCESS in 4m 36s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ad857eea1e504b73864f1a982ab8e71f : SUCCESS in 3m 48s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aec3c3d8fcbd49eca84399182ff07ee0 : SUCCESS in 3m 50s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/fed07da3c9e74dcbb76c74c6fdc54b87 : SUCCESS in 20m 03s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d062fdbb4bad452f9fe552ce7557e9d4 : SUCCESS in 10m 27s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a75d76d5b4314c99bd840fe92ac6e084 : SUCCESS in 10m 14s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/dc6c6aa6a6384dafa4aefa07a1b39ebe : SUCCESS in 54m 55s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/559b349c1df0441a904c416b87edfad3 : SUCCESS in 1h 05m 27s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/611686c62d28451481d70b7cba0b2f16 : SUCCESS in 1h 19m 56s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/33025472eeeb46fbbcdf8fec1e2732c1 : SUCCESS in 59m 10s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/783c6374bbca4ce1b2dd42148273c985 : SUCCESS in 50m 35s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/46275ac5fc3d4688b7f808a1d2016a12 : SUCCESS in 56m 15s\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5383a8672a974f908edc1413c43ad0d5 : SUCCESS in 36m 45s\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/dd6039d61bcf4bba8d7c2a9bd3520623 : SUCCESS in 34m 14s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/6aab760557f14b7ab6f086b96a60e04e : SUCCESS in 18m 04s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/76ffd78bdc6a46038b926416f549a5c8 : FAILURE in 8m 52s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/23f05dc905f0463abc6e8652aba28d56 : SUCCESS in 1h 42m 42s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41e6f21787eb47088d012b62bd9eda0f : SUCCESS in 3m 58s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b360f4629e2a46a9b68c29628b537415 : SUCCESS in 1h 37m 07s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"45cce662f91642fb04fa9696e9b7726558b3bef7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 21:02:31.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"de9787ce5a0bf96e1d49e5f67d431babd9d8a83f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 23:16:38.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a7ad288f569c45dca9b9d359b664fdb8 : SUCCESS in 7m 10s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f78d8f4e699e4e7fa74bfbb5fe43d5bc : SUCCESS in 4m 29s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/683e01a4721d47a986c0704f225b749b : SUCCESS in 4m 28s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/e71e7e9c318e4b40a51c23888ebd4eed : SUCCESS in 22m 02s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4943a962921b4489b9d31f6320b22614 : SUCCESS in 15m 11s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/1e7d6eb1e9934ef08e2322906d5e4372 : SUCCESS in 55m 10s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/88d69162b84a4b0d9be66561995e2c37 : SUCCESS in 1h 02m 16s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/f3e73ea508c348b0b9960a3864366969 : SUCCESS in 1h 07m 10s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/89d7bce83b8c42c5a111f078968990e4 : SUCCESS in 52m 02s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9e59f1b12c684da78f4f76488c6f814e : SUCCESS in 22m 32s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/ad0b986b687d49b5a6508fddf474c726 : SUCCESS in 1h 48m 10s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c5c9507c0c74175a8fdb741cc1363f9 : SUCCESS in 3m 44s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f19b73d29b5a450a8605581520855a69 : SUCCESS in 2h 07m 36s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"706eddc409129b86ef4602e4f03e0f0ac9245092","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 23:16:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"acd3dfb04d2c7caa3dc30fd094e3715aa6d1820d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 23:17:50.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ad42f6273df4d45b352b07d9e70fb25 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"fd02761c6e634100ce4583c88a092f2958742ae7","revisions":{"895fdfd0f15ef4a35c418e287e0a91aa74197f50":{"kind":"REWORK","_number":1,"created":"2019-10-04 18:40:51.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/22/686822/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/22/686822/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/22/686822/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/22/686822/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/22/686822/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/22/686822/1"}}},"commit":{"parents":[{"commit":"dbcdeb33839ff15d2d6c7d8fd0eb9b74da5d61ea","subject":"Prepare Train RC1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/dbcdeb33839ff15d2d6c7d8fd0eb9b74da5d61ea"}]}],"author":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2019-10-04 18:27:58.000000000","tz":-420},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2019-10-04 18:39:50.000000000","tz":-420},"subject":"Update the method of configuring glance backends","message":"Update the method of configuring glance backends\n\nThis patch updates the method of configuring glance backends to use the\nmodern syntax associated with glance\u0027s Multi Store feature. Glance\u0027s\nglance_store/stores and glance_store/default_store parameters were\ndeprecated in Rocky, and are scheduled for removal in Ussuri.\n\nBackends are now specified using new glance::api::enabled_backends and\nglance::api::default_backend parameters. A series of\nglance::backend::multistore::* defined resources support configuring\nmultiple instances of each backend.\n\nThe legacy glance::api::stores and glance::api::default_store parameters,\nas well as the glance::backend::* class resources, are deprecated, but\ncontinue to function for backward compatibility.\n\nThe glance::backend::s3 class is completely deprecated. Glance removed\nsupport for the s3 backend in Newton.\n\nCloses-Bug: #1846808\nChange-Id: I4f3ec9ee7c4147a4e62de961a9532bdc45526f6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/895fdfd0f15ef4a35c418e287e0a91aa74197f50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/895fdfd0f15ef4a35c418e287e0a91aa74197f50"}]},"branch":"refs/heads/master"},"e6a65b100832820b04ec86c96f176c113ca84af6":{"kind":"REWORK","_number":2,"created":"2019-10-04 18:51:57.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/22/686822/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/22/686822/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/22/686822/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/22/686822/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/22/686822/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/22/686822/2"}}},"commit":{"parents":[{"commit":"dbcdeb33839ff15d2d6c7d8fd0eb9b74da5d61ea","subject":"Prepare Train RC1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/dbcdeb33839ff15d2d6c7d8fd0eb9b74da5d61ea"}]}],"author":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2019-10-04 18:27:58.000000000","tz":-420},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2019-10-04 18:51:50.000000000","tz":-420},"subject":"Update the method of configuring glance backends","message":"Update the method of configuring glance backends\n\nThis patch updates the method of configuring glance backends to use the\nmodern syntax associated with glance\u0027s Multi Store feature. Glance\u0027s\nglance_store/stores and glance_store/default_store parameters were\ndeprecated in Rocky, and are scheduled for removal in Ussuri.\n\nBackends are now specified using new glance::api::enabled_backends and\nglance::api::default_backend parameters. A series of\nglance::backend::multistore::* defined resources support configuring\nmultiple instances of each backend.\n\nThe legacy glance::api::stores and glance::api::default_store parameters,\nas well as the glance::backend::* class resources, are deprecated, but\ncontinue to function for backward compatibility.\n\nThe glance::backend::s3 class is completely deprecated. Glance removed\nsupport for the s3 backend in Newton.\n\nCloses-Bug: #1846808\nChange-Id: I4f3ec9ee7c4147a4e62de961a9532bdc45526f6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e6a65b100832820b04ec86c96f176c113ca84af6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e6a65b100832820b04ec86c96f176c113ca84af6"}]},"branch":"refs/heads/master"},"e524bfb11eee3d0e4f0ad807688b0207e33851f2":{"kind":"REWORK","_number":3,"created":"2019-10-11 12:59:07.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/22/686822/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/22/686822/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/22/686822/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/22/686822/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/22/686822/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/22/686822/3"}}},"commit":{"parents":[{"commit":"13653589788c6ebe4f0d129968161216fd53f161","subject":"Fix properties in glance_image provider for osc \u003e\u003d 4.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/13653589788c6ebe4f0d129968161216fd53f161"}]}],"author":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2019-10-11 12:58:47.000000000","tz":-420},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2019-10-11 12:58:47.000000000","tz":-420},"subject":"Update the method of configuring glance backends","message":"Update the method of configuring glance backends\n\nThis patch updates the method of configuring glance backends to use the\nmodern syntax associated with glance\u0027s Multi Store feature. Glance\u0027s\nglance_store/stores and glance_store/default_store parameters were\ndeprecated in Rocky, and are scheduled for removal in Ussuri.\n\nBackends are now specified using new glance::api::enabled_backends and\nglance::api::default_backend parameters. A series of\nglance::backend::multistore::* defined resources support configuring\nmultiple instances of each backend.\n\nThe legacy glance::api::stores and glance::api::default_store parameters,\nas well as the glance::backend::* class resources, are deprecated, but\ncontinue to function for backward compatibility.\n\nThe glance::backend::s3 class is completely deprecated. Glance removed\nsupport for the s3 backend in Newton.\n\nCloses-Bug: #1846808\nChange-Id: I4f3ec9ee7c4147a4e62de961a9532bdc45526f6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e524bfb11eee3d0e4f0ad807688b0207e33851f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e524bfb11eee3d0e4f0ad807688b0207e33851f2"}]},"branch":"refs/heads/master"},"4e69f53c7cb4d908c62a205cf7d6e0b86fadaf7d":{"kind":"REWORK","_number":4,"created":"2019-10-11 15:33:34.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/22/686822/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/22/686822/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/22/686822/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/22/686822/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/22/686822/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/22/686822/4"}}},"commit":{"parents":[{"commit":"13653589788c6ebe4f0d129968161216fd53f161","subject":"Fix properties in glance_image provider for osc \u003e\u003d 4.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/13653589788c6ebe4f0d129968161216fd53f161"}]}],"author":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2019-10-11 12:58:47.000000000","tz":-420},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2019-10-11 15:33:28.000000000","tz":-420},"subject":"Update the method of configuring glance backends","message":"Update the method of configuring glance backends\n\nThis patch updates the method of configuring glance backends to use the\nmodern syntax associated with glance\u0027s Multi Store feature. Glance\u0027s\nglance_store/stores and glance_store/default_store parameters were\ndeprecated in Rocky, and are scheduled for removal in Ussuri.\n\nBackends are now specified using new glance::api::enabled_backends and\nglance::api::default_backend parameters. A series of\nglance::backend::multistore::* defined resources support configuring\nmultiple instances of each backend.\n\nThe legacy glance::api::stores and glance::api::default_store parameters,\nas well as the glance::backend::* class resources, are deprecated, but\ncontinue to function for backward compatibility.\n\nThe glance::backend::s3 class is completely deprecated. Glance removed\nsupport for the s3 backend in Newton.\n\nCloses-Bug: #1846808\nChange-Id: I4f3ec9ee7c4147a4e62de961a9532bdc45526f6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/4e69f53c7cb4d908c62a205cf7d6e0b86fadaf7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/4e69f53c7cb4d908c62a205cf7d6e0b86fadaf7d"}]},"branch":"refs/heads/master"},"96a6bf3464538a1c3ad0557c9d76acc6dc7df7a0":{"kind":"REWORK","_number":5,"created":"2019-10-15 15:14:24.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/22/686822/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/22/686822/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/22/686822/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/22/686822/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/22/686822/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/22/686822/5"}}},"commit":{"parents":[{"commit":"13653589788c6ebe4f0d129968161216fd53f161","subject":"Fix properties in glance_image provider for osc \u003e\u003d 4.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/13653589788c6ebe4f0d129968161216fd53f161"}]}],"author":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2019-10-11 12:58:47.000000000","tz":-420},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2019-10-15 15:11:08.000000000","tz":-420},"subject":"Update the method of configuring glance backends","message":"Update the method of configuring glance backends\n\nThis patch updates the method of configuring glance backends to use the\nmodern syntax associated with glance\u0027s Multi Store feature. Glance\u0027s\nglance_store/stores and glance_store/default_store parameters were\ndeprecated in Rocky, and are scheduled for removal in Ussuri.\n\nBackends are now specified using new glance::api::enabled_backends and\nglance::api::default_backend parameters. A series of\nglance::backend::multistore::* defined resources support configuring\nmultiple instances of each backend.\n\nThe legacy glance::api::stores and glance::api::default_store parameters,\nas well as the glance::backend::* class resources, are deprecated, but\ncontinue to function for backward compatibility.\n\nThe glance::backend::s3 class is completely deprecated. Glance removed\nsupport for the s3 backend in Newton.\n\nCloses-Bug: #1846808\nChange-Id: I4f3ec9ee7c4147a4e62de961a9532bdc45526f6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/96a6bf3464538a1c3ad0557c9d76acc6dc7df7a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/96a6bf3464538a1c3ad0557c9d76acc6dc7df7a0"}]},"branch":"refs/heads/master"},"498407cc6bb6ce4c82eba63387fef4e42e555c90":{"kind":"REWORK","_number":6,"created":"2019-10-15 17:27:25.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/22/686822/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/22/686822/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/22/686822/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/22/686822/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/22/686822/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/22/686822/6"}}},"commit":{"parents":[{"commit":"13653589788c6ebe4f0d129968161216fd53f161","subject":"Fix properties in glance_image provider for osc \u003e\u003d 4.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/13653589788c6ebe4f0d129968161216fd53f161"}]}],"author":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2019-10-11 12:58:47.000000000","tz":-420},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2019-10-15 17:27:19.000000000","tz":-420},"subject":"Update the method of configuring glance backends","message":"Update the method of configuring glance backends\n\nThis patch updates the method of configuring glance backends to use the\nmodern syntax associated with glance\u0027s Multi Store feature. Glance\u0027s\nglance_store/stores and glance_store/default_store parameters were\ndeprecated in Rocky, and are scheduled for removal in Ussuri.\n\nBackends are now specified using new glance::api::enabled_backends and\nglance::api::default_backend parameters. A series of\nglance::backend::multistore::* defined resources support configuring\nmultiple instances of each backend.\n\nThe legacy glance::api::stores and glance::api::default_store parameters,\nas well as the glance::backend::* class resources, are deprecated, but\ncontinue to function for backward compatibility.\n\nThe glance::backend::s3 class is completely deprecated. Glance removed\nsupport for the s3 backend in Newton.\n\nCloses-Bug: #1846808\nChange-Id: I4f3ec9ee7c4147a4e62de961a9532bdc45526f6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/498407cc6bb6ce4c82eba63387fef4e42e555c90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/498407cc6bb6ce4c82eba63387fef4e42e555c90"}]},"branch":"refs/heads/master"},"fd02761c6e634100ce4583c88a092f2958742ae7":{"kind":"REWORK","_number":7,"created":"2019-10-16 21:46:46.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/22/686822/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/22/686822/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/22/686822/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/22/686822/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/22/686822/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/22/686822/7"}}},"commit":{"parents":[{"commit":"13653589788c6ebe4f0d129968161216fd53f161","subject":"Fix properties in glance_image provider for osc \u003e\u003d 4.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/13653589788c6ebe4f0d129968161216fd53f161"}]}],"author":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2019-10-11 12:58:47.000000000","tz":-420},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2019-10-16 21:46:39.000000000","tz":-420},"subject":"Update the method of configuring glance backends","message":"Update the method of configuring glance backends\n\nThis patch updates the method of configuring glance backends to use the\nmodern syntax associated with glance\u0027s Multi Store feature. Glance\u0027s\nglance_store/stores and glance_store/default_store parameters were\ndeprecated in Rocky, and are scheduled for removal in Ussuri.\n\nBackends are now specified using new glance::api::enabled_backends and\nglance::api::default_backend parameters. A series of\nglance::backend::multistore::* defined resources support configuring\nmultiple instances of each backend.\n\nThe legacy glance::api::stores and glance::api::default_store parameters,\nas well as the glance::backend::* class resources, are deprecated, but\ncontinue to function for backward compatibility.\n\nThe glance::backend::s3 class is completely deprecated. Glance removed\nsupport for the s3 backend in Newton.\n\nCloses-Bug: #1846808\nChange-Id: I4f3ec9ee7c4147a4e62de961a9532bdc45526f6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/fd02761c6e634100ce4583c88a092f2958742ae7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/fd02761c6e634100ce4583c88a092f2958742ae7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
