)]}'
{"id":"openstack%2Fpuppet-glance~707163","triplet_id":"openstack%2Fpuppet-glance~master~Ifb3bbd3a07ccddafa757cfbc1bec358f49a95c65","project":"openstack/puppet-glance","branch":"master","topic":"registry","hashtags":[],"change_id":"Ifb3bbd3a07ccddafa757cfbc1bec358f49a95c65","subject":"Deprecate glance::registry","status":"MERGED","created":"2020-02-11 14:45:06.000000000","updated":"2020-02-16 22:34:38.000000000","submitted":"2020-02-16 22:34:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707163-1581892478363-b4850ae3","meta_rev_id":"ba372dbe3dbf6bbf002419784acf0600f46f85ed","_number":707163,"virtual_id_number":707163,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-02-12 03:55:39.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-16 22:34:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-02-13 17:02:59.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2020-02-14 18:30:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2020-02-16 20:35:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2020-02-16 20:35:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-13 17:02:59.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-02-14 18:30:33.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2020-02-16 20:35:48.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-02-16 22:34:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"da20c3de61adf9cd64f901816571a5f1c14aa517","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-02-11 14:45:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"727b355778306ebf4c1f2c8d8149f8df7f682dd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 15:52:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/99bfada5dd7e403198cde75c5a11bdc6 : SUCCESS in 6m 34s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f4fe96c161e54eeca21187d74944db27 : SUCCESS in 5m 47s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bdc293a905984c8d808e95d0a3a9c324 : SUCCESS in 4m 49s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/ab07f1e281834f61acc0b5cd095a1c6f : RETRY_LIMIT in 6m 07s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1a52f396084d41fc95d7acdad9d9aef0 : SUCCESS in 19m 05s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c20a4b5de344479ab49f3c498b55788c : SUCCESS in 19m 10s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/5962ea3260534ff5b0408533532547e5 : RETRY_LIMIT in 10m 08s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/aafd8840e3574716b6b7103ac21f66d9 : RETRY_LIMIT in 9m 54s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/f1551f679d2f415f9504574754e2105f : RETRY_LIMIT in 8m 54s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/8523c3d5f62d4d73b9e2951a8fef2fbc : RETRY_LIMIT in 7m 21s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/f6d2676597274f2d9c4e59240ea094bc : SUCCESS in 47m 22s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e064ae674a3b4cfcb3be3ed10cc416bc : SUCCESS in 50m 09s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0fc296d08b274cab803ce7c64e1cc683 : SUCCESS in 41m 06s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/374eac3f4616471e832ea4369c7861a3 : SUCCESS in 34m 55s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/91a4cb60b8a64c33b2705617a225ddbf : SUCCESS in 52m 16s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/32e2dbfc346948c0a4879c00b44a6672 : SUCCESS in 1h 01m 15s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2a1f5079af6c4beca464c30c979b0bfa : SUCCESS in 1h 00m 05s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/550ce45441784ac797a1e94c42cb4927 : SUCCESS in 46m 51s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/80ad09b029b1400fa31a1f5084faf658 : RETRY_LIMIT in 8m 24s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6d29410c2128428eaf28b057e66dadc1 : FAILURE in 14m 44s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/c8e8948aeebc454b939a5b0f50d11230 : FAILURE in 9m 58s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3df04f001cdc4984b8b54c8372203df6 : FAILURE in 9m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"662b1e333533ad91ff2f5ab66037781dbb64201a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 19:38:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/311faea4960a4443b0aa19c30d666d05 : SUCCESS in 7m 24s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c811b033748e4d4abff5bf7cac4c9a5a : SUCCESS in 6m 11s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b47b21c3b20487995ae826b81360221 : SUCCESS in 4m 59s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/f9697c05c562440fadee3d0a40a94cff : SUCCESS in 19m 57s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c074a598ccc427e913d793ca3e85ebb : SUCCESS in 12m 24s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/30c04259a3d7477e900ea0435ed0c8a9 : SUCCESS in 12m 43s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/ab7d1b7c5afa457f9a688a8447225807 : SUCCESS in 1h 05m 58s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/fbc1a1b03c624f409e3b6e2f90029231 : SUCCESS in 1h 02m 42s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/b91f84033c2c4ecb8767667a0dc31192 : SUCCESS in 1h 18m 14s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/a74d16a1e1b34bbd9f4ec397967e82dc : SUCCESS in 59m 39s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/0e041df0ebf64891b8e5b6ef9e3533cc : SUCCESS in 1h 02m 53s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8a447dddd35944faa7b631e8321c3bbd : RETRY_LIMIT in 2m 02s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0745ce24abd948a2baaff40d988b91b0 : RETRY_LIMIT in 2m 40s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/7f5393fc9e18494fbbfb079370a9b321 : RETRY_LIMIT in 2m 24s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0c184eaf818d496aa639ee115b9fa30a : SUCCESS in 1h 01m 51s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e2743fe9a6894f4ebb72f85f1e1d9a73 : FAILURE in 1h 17m 16s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f60d3f498770419caca5821a76275d85 : SUCCESS in 59m 59s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d45082c0e86343988c5f9400f5e4af71 : SUCCESS in 45m 31s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ef494cbcfa844fc082fc34f5fd02a243 : SUCCESS in 36m 36s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dcaa4e4c9d9645bbb9272258774f1061 : FAILURE in 18m 14s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/7742d18c5ce94261948aa9c28561ec85 : FAILURE in 13m 47s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8aea3240158941fe81084908ca0afe27 : FAILURE in 13m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3410251bd5d4a71730441ac0c0170069c37739e9","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-02-12 02:47:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"405e68cdf23697cb16eca9814019f3b944b72e27","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-02-12 03:55:23.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7505f9dbcae80917d96696f8c0d111286e8bedbd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-02-12 03:55:39.000000000","message":"Patch Set 2:\n\nupdated commit message to fix typo.","accounts_in_message":[],"_revision_number":2},{"id":"6f06cb7457980eebb32b65d544a837edf6c14e4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 06:04:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cb0b9222e0174c569469d636d738f8ed : SUCCESS in 6m 28s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/51e31f8a14f84ca3a9b2acef0c7ccb51 : SUCCESS in 7m 02s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/567640265ca44618881af9ad07f815cc : SUCCESS in 6m 31s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/bcc01c57d4344b55bd28329179dc057c : SUCCESS in 18m 22s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9cddb625b0794b2db1098ecf0c6c3eea : SUCCESS in 15m 04s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/34235ad50eb2462aae569ce390aa4fd2 : SUCCESS in 14m 44s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/c72f4cdd23424ca38643eb8758cff5de : SUCCESS in 1h 22m 00s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/4b2fd9c3e1484483acb0fb03dd2d69c6 : SUCCESS in 1h 04m 38s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/6803b30cfdf6455bb50cd21c5c9bd2f8 : SUCCESS in 1h 26m 58s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/3b94fe711992432aa6fcf1b945f281dc : SUCCESS in 58m 32s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/61b7dbb981f84361877337cce3bc093c : SUCCESS in 1h 16m 01s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/212f26b641df49bb99a7a345f5370fe4 : SUCCESS in 1h 11m 22s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/261ade36f9dd4c6bb64ba0a7c1de34fc : SUCCESS in 1h 07m 28s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/0c9304db13074f37851ff297a9e12189 : SUCCESS in 58m 03s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6a05ecdd59c44cc7b7da624b6ec62bf6 : FAILURE in 49m 46s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6d8e5019253c4b26b03ff43114149fb0 : FAILURE in 38m 37s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ba503a123279464d9ddd271496a180ab : SUCCESS in 1h 13m 18s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b8092195848342caa30880792cc2aedc : SUCCESS in 1h 09m 32s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/f72783bc04424e51840d2035977c2ce0 : SUCCESS in 31m 27s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/84532baa5e954bd3bef1371e19a5a911 : FAILURE in 24m 43s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/becb492e6dd94c72987b729bcf40f8be : SUCCESS in 2h 03m 35s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/51a9ac78f29140b1b1a7cbea1a511012 : SUCCESS in 1h 46m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"26bfc28765f075314fdef35999c097a5a1427c86","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-02-13 17:02:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"011f3a9839d493c80d30bea1bfbb482372bbcb89","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2020-02-14 18:30:33.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c326652a9eb7b85ba288b4a54616cc7d711dd07b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-02-16 20:35:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d064c1644aeaab8566fc31762510ed57c66e9b59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 20:36:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"abe156b987a2ac70f98a2fe1711106e6f6560ea8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 22:34:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6fb50806916f4688afa6a316979ecd3f : SUCCESS in 6m 02s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35466ee8d93b475db2d283afc801c463 : SUCCESS in 6m 08s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5f150978a5cf49aaa8a60a8de7005729 : SUCCESS in 6m 27s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/6576c649b16f4096b9b304b288efda3b : SUCCESS in 20m 44s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c88d3f07137b48ff8eeaaa2afcc8aaec : SUCCESS in 9m 54s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/5131a381cce54735838bcebd3e2e310b : SUCCESS in 57m 17s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/4965050c324e4ff3a7f1a6d6b766a815 : SUCCESS in 1h 03m 12s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/eac694eceb544b2d86bce025b6581be8 : SUCCESS in 1h 04m 49s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/1ed20d7d98a144b7ad522c5a34233446 : SUCCESS in 47m 35s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/b8a3b995965d41aa820b409790084c43 : SUCCESS in 18m 05s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/c1bc7edd2b124d2e9531f9226e2a20f5 : SUCCESS in 1h 35m 47s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a412620ecf0c4574a824a9df7f7b65d7 : SUCCESS in 1h 54m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1c55d237b783f88093923eb25bd8a9d1cd1eacde","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 22:34:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2c0a176b3541d0c4e1b833fc8ed69018c47043bd","revisions":{"d98b68e612e9dd6c6fdfeacb1cec042b988e906f":{"kind":"REWORK","_number":1,"created":"2020-02-11 14:45:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/63/707163/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/63/707163/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/63/707163/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/63/707163/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/63/707163/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/63/707163/1"}}},"commit":{"parents":[{"commit":"c97e42f67d1bff59237fe76f4dbb4d29aa0bc874","subject":"Deprecate glance::api::show_multiple_locations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/c97e42f67d1bff59237fe76f4dbb4d29aa0bc874"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-02-11 14:44:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-02-11 14:44:07.000000000","tz":540},"subject":"Deprecate glance::registry","message":"Deprecate glance::registry\n\nBecause glance-registry was deprecated in Queens release[1], deprecate\nthe claa, glance::registry, to configure the service.\n\n[1] http://specs.openstack.org/openstack/glance-specs/specs/queens/approved/glance/deprecate-registry.html\n\nChange-Id: Ifb3bbd3a07ccddafa757cfbc1bec358f49a95c65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/d98b68e612e9dd6c6fdfeacb1cec042b988e906f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/d98b68e612e9dd6c6fdfeacb1cec042b988e906f"}]},"branch":"refs/heads/master"},"2c0a176b3541d0c4e1b833fc8ed69018c47043bd":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-02-12 03:55:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/63/707163/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/63/707163/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/63/707163/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/63/707163/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/63/707163/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/63/707163/2"}}},"commit":{"parents":[{"commit":"c97e42f67d1bff59237fe76f4dbb4d29aa0bc874","subject":"Deprecate glance::api::show_multiple_locations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/c97e42f67d1bff59237fe76f4dbb4d29aa0bc874"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-02-11 14:44:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-02-12 03:55:13.000000000","tz":0},"subject":"Deprecate glance::registry","message":"Deprecate glance::registry\n\nBecause glance-registry was deprecated in Queens release[1], deprecate\nglance::registry class to configure the service.\n\n[1] http://specs.openstack.org/openstack/glance-specs/specs/queens/approved/glance/deprecate-registry.html\n\nChange-Id: Ifb3bbd3a07ccddafa757cfbc1bec358f49a95c65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/2c0a176b3541d0c4e1b833fc8ed69018c47043bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/2c0a176b3541d0c4e1b833fc8ed69018c47043bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
