)]}'
{"id":"openstack%2Fpuppet-glance~714877","triplet_id":"openstack%2Fpuppet-glance~master~I67e5c96a70ae4ebb6934129286516a24c5a22fbf","project":"openstack/puppet-glance","branch":"master","topic":"bug/1868511","hashtags":[],"change_id":"I67e5c96a70ae4ebb6934129286516a24c5a22fbf","subject":"Deprecate min_pool_size option","status":"MERGED","created":"2020-03-25 08:23:18.000000000","updated":"2020-03-25 21:17:11.000000000","submitted":"2020-03-25 21:15:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714877-1585170901775-28344ecc","meta_rev_id":"6ce956a3e2f92d3da5d15876369a2c01298d9439","_number":714877,"virtual_id_number":714877,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-25 21:15:01.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2020-03-25 17:29:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2020-03-25 19:03:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-25 17:29:38.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-03-25 19:03:10.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-03-25 21:15:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"932c2854c9120a0d5f480061e5f3661d0bed0afb","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-03-25 08:23:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"00178a928df13defa8a6d7745c4f1a7bb5f8dfa5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 10:28:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/72d7a35059ce4f728e4852ee0dee2590 : SUCCESS in 6m 31s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81f0edc9d2404262bbabac7e246772be : SUCCESS in 6m 30s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5b89309c669e46f49993130ee7607a7b : SUCCESS in 6m 39s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/450834be1e6c4caca44301413324f936 : SUCCESS in 21m 01s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/6628cb1a555843d7b3090e8ce9b844df : SUCCESS in 18m 40s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/75b03757d8944c71803ef18e70306a0a : SUCCESS in 17m 56s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/314e2fb8f2bc41e298cf437b5f90a595 : SUCCESS in 13m 22s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0cc4778d3b914a23acd1fd44408ffea0 : SUCCESS in 15m 52s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/11ab9b25bed24c0dbe3a120825e6e1e7 : SUCCESS in 47m 03s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/47b79526baaf4675b6825e5def2b8230 : SUCCESS in 49m 50s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f0a417f90236490d9b2a133e6a79bf9d : SUCCESS in 39m 18s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/8138496b88554d7da737ab7b674e7ce0 : SUCCESS in 39m 47s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/12a8e50f0d3543f789d632b6c9ab2907 : SUCCESS in 55m 43s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b93829e4984f44edac6d92ce0d557eb0 : SUCCESS in 1h 07m 32s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/bab788ed78bb49aeb9626b4bfe7c720c : SUCCESS in 1h 05m 56s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c3581df04fd94ce1afdb019bda88323a : SUCCESS in 45m 06s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e45f6265e7774c3db20ae184aeed8ce8 : SUCCESS in 25m 02s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e70ea661255345e0a59480933edcfe30 : FAILURE in 11m 50s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9df44712816471887fc9db88f854b3c : SUCCESS in 5m 36s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a01a0ebce15e40fdae78b9b9db7fea9b : SUCCESS in 1h 55m 54s","accounts_in_message":[],"_revision_number":1},{"id":"e6f1a0d2a4d050425fc774db8eced4c93d189183","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-25 17:29:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"44d1bf0151a28f301fccea106b5c19f27485f65f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-03-25 19:03:10.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7c37ef6e6662b05df9e3a45dc6c5c598afc24a93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 19:03:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bd0a24b8d4b041ffe3e065bf9bd847878b15760f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 21:15:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c1d107efbaf478da429bd3864bb081b : SUCCESS in 6m 14s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/27a7c7b8992d464faae0e3e179e90f3a : SUCCESS in 5m 43s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a29878a2b1fa45ddbc57868e57715a80 : SUCCESS in 7m 37s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/277fc97f8bba465da5b47c3060d5b06d : SUCCESS in 21m 14s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b610ac6d3d0844ee9bbf4fd5d8eefb8a : SUCCESS in 13m 16s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8a5d76393d06462f825bdc8aac79852e : SUCCESS in 1h 07m 28s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/406025e65fe7434abc60fb6d7139d920 : SUCCESS in 1h 04m 13s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/eceebd4750e64e28887feb5f5f4fe75a : SUCCESS in 1h 15m 51s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/618907d6231342a8b2631012538fdcb1 : SUCCESS in 44m 25s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/8f1e727e3fe749e6b0c8adcb00019563 : SUCCESS in 25m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/991de32999ba43a2b5577d48a446dc7a : SUCCESS in 5m 07s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8ddc4c17bfa24e7eac17cb761d7abdeb : SUCCESS in 2h 04m 45s","accounts_in_message":[],"_revision_number":1},{"id":"95f20e7ce1ccf44cfd435dcf633bee5e4b6e217c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 21:15:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"98a119a736e0c19933c49c762d7b3323cfa5a3d3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 21:17:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06d1cfaa430e4c88bb182c77156fd1b1 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ca939e97322c0ea9bdf5d10566f18308fcb2334a","revisions":{"ca939e97322c0ea9bdf5d10566f18308fcb2334a":{"kind":"REWORK","_number":1,"created":"2020-03-25 08:23:18.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/77/714877/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/77/714877/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/77/714877/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/77/714877/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/77/714877/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/77/714877/1"}}},"commit":{"parents":[{"commit":"16d279e21e66b842d58a05a138ee9a4c48688f59","subject":"Remove idle_timeout option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/16d279e21e66b842d58a05a138ee9a4c48688f59"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-03-25 06:53:40.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-03-25 06:53:40.000000000","tz":480},"subject":"Deprecate min_pool_size option","message":"Deprecate min_pool_size option\n\nmin_pool_size option is not used，see:\n\nhttps://review.opendev.org/#/c/565090/\n\nChange-Id: I67e5c96a70ae4ebb6934129286516a24c5a22fbf\nCloses-Bug: #1868511\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/ca939e97322c0ea9bdf5d10566f18308fcb2334a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/ca939e97322c0ea9bdf5d10566f18308fcb2334a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
