)]}'
{"id":"openstack%2Fpuppet-glance~716181","triplet_id":"openstack%2Fpuppet-glance~master~I3692a2015ccaa79acd69679e0c749d9d6fa78bcb","project":"openstack/puppet-glance","branch":"master","topic":"centos8-client","hashtags":[],"change_id":"I3692a2015ccaa79acd69679e0c749d9d6fa78bcb","subject":"Expect python3 client package in CentOS8","status":"MERGED","created":"2020-03-31 07:31:39.000000000","updated":"2020-04-01 15:40:14.000000000","submitted":"2020-04-01 15:40:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716181-1585755614953-e8299aed","meta_rev_id":"2afb6787af9f23d6e7e2f490119f3f1750e9ea6b","_number":716181,"virtual_id_number":716181,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-01 15:40:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2020-04-01 06:35:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-04-01 13:36:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-31 14:48:09.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-04-01 06:35:16.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-04-01 13:36:01.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-04-01 15:40:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"080a7e33b4360bc14fc6b50b51ffe1a58307f470","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-31 07:31:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e961fa3ce9a9b3e396d7d61e1121cf9cfcc1b138","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 10:04:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c528b358ce92457189c597be98e7720c : SUCCESS in 7m 34s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/028af66ba4774e74809599c62bfa74c1 : SUCCESS in 5m 44s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cdb35e9de8f741e0831bc3d9239436e1 : SUCCESS in 5m 37s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/404ef3a17d634adc88cdd69396ae568d : SUCCESS in 17m 13s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/71fca4826deb4b6190776435470041c2 : SUCCESS in 22m 57s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/129239f86db543429c659be1d590f4b7 : SUCCESS in 16m 22s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/888ec30e6e8843d9a3be0b0da80fa367 : SUCCESS in 13m 21s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/28e9c8a7959d402d926e5aaf984c5d7b : SUCCESS in 12m 45s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/36a7b5ee956b48cc889b68dbd12ca6ed : SUCCESS in 2h 01m 15s","accounts_in_message":[],"_revision_number":1},{"id":"3c75615e1e65b8364a3d5eb8ea0c64dab5945d8d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-31 14:31:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"958d4bbbf28948554c7607c300f7744ca1c3f35e","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-03-31 14:48:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"79d7085448fb77cdc6fd5dc101b9b975b203679f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-03-31 21:47:57.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cd09d4e2f980c2dbfb55fcd6debe65b576a9126d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 21:48:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"503529d1c09c93431809efca793128dbc2c6596a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-31 22:41:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"471e80aef8d6eaf45ee8d1990521ee612819c302","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-31 22:42:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"119bebcf7e3fd513911cebc83bd345f887d636dc","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-03-31 23:07:05.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ac37ede164feda02936a04b3abcbbd059062b1cb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-31 23:25:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"97ab6e7f539c2c2c68518b95be9aad2b576b5280","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-31 23:26:23.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e44ce7f2de31e65d8c56b35a2e7b45f7295c7265","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 01:14:53.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ad32e06f60a482bbdc410d19836763d : SUCCESS in 5m 02s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/939aa58e26b041c9a5e896c59681f524 : SUCCESS in 3m 52s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa01f524390f43d0a7cf70d5fd83263d : SUCCESS in 5m 30s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/74d7d7985e7149e5a456b16783aa35e8 : FAILURE in 10m 39s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/8c20013b94cc4cd28c2521bc8284e318 : FAILURE in 10m 24s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b3a32bb0edf744998ec20cf3dc08b1c4 : FAILURE in 8m 30s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/320dce420dc54b0d81b8190ca04f079f : FAILURE in 10m 47s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a154847ea6fc443f9c3e5c4444df005d : SUCCESS in 1h 44m 05s","accounts_in_message":[],"_revision_number":5},{"id":"a080264dda37776c47f0675f48d9db0ea6ec4153","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-01 01:28:37.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e23e3e56461875481b929a75c66765f0d50a939e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 03:46:47.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb42b32219d2403e9d78f50e3d8f5178 : SUCCESS in 4m 45s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8888e93c0b4842289cf9a4209217da87 : SUCCESS in 5m 16s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e5f39f9da14a412790e846e00eeb4e3c : SUCCESS in 4m 24s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/135d27ad207c4908bd47245e8262b8ea : SUCCESS in 11m 47s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/d462f1f8ec694c1f87b29d37bdbc1fa4 : SUCCESS in 12m 22s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1770006932284f7ba1518eca8b4c6894 : SUCCESS in 10m 13s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d94f00512e654686b3811c3947be2305 : SUCCESS in 10m 56s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/cf0b5a5d49bd44b98664caddf94a6379 : SUCCESS in 2h 13m 00s","accounts_in_message":[],"_revision_number":6},{"id":"1cda20bac7d8358c84ced506f3840ad5aa93763b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-04-01 06:35:16.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"a43f0ecf57be550d957a73823a35ea22a15070a6","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-04-01 13:36:01.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"20588642db515b5cbdb7a815d516ee74afa552de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 13:36:14.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"9be9a41a208701b5dd381a45233168276a6e36dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 15:40:14.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bbfb91016e91491e900146af5e7772ac : SUCCESS in 5m 56s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/71d8ae0fd2664779a5bece0ffba44888 : SUCCESS in 6m 12s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f20eb98138914885b09f3a59d4f6dad6 : SUCCESS in 5m 56s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/b31648eb0c444241b3139d80329dc74d : SUCCESS in 18m 39s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/e2b379dc1ad7494b8f0a13d12174eb77 : SUCCESS in 13m 54s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b06e7d6b801f47cebfb9649d1b3e81ca : SUCCESS in 13m 19s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/908038ba67b044e59ea5d3b1b4fcced4 : SUCCESS in 1h 58m 05s","accounts_in_message":[],"_revision_number":6},{"id":"5bed9d0e9cd9c7a865bd5075bf7b7efe01982f59","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 15:40:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"5bd1ee6bc749a0fd804f34fcedcd3d354b5ef2f1","revisions":{"cd6dee1e0ff7c9c0ef2b59e906b7081a61874253":{"kind":"REWORK","_number":1,"created":"2020-03-31 07:31:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/81/716181/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/81/716181/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/716181/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/716181/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/716181/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/81/716181/1"}}},"commit":{"parents":[{"commit":"16d279e21e66b842d58a05a138ee9a4c48688f59","subject":"Remove idle_timeout option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/16d279e21e66b842d58a05a138ee9a4c48688f59"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-31 07:31:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-31 07:31:37.000000000","tz":540},"subject":"Expect python3 client package in CentOS8","message":"Expect python3 client package in CentOS8\n\nIn CentOS, we expect to have python3 client package in 8.x while we\nexpect to have python2 in 7.x .\nFix unit tests to expect the correct version according to os major\nversion.\n\nChange-Id: I3692a2015ccaa79acd69679e0c749d9d6fa78bcb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/cd6dee1e0ff7c9c0ef2b59e906b7081a61874253"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/cd6dee1e0ff7c9c0ef2b59e906b7081a61874253"}]},"branch":"refs/heads/master"},"1f1b58a62ec1ee1d870e77fc14fc2086aa03d0f2":{"kind":"REWORK","_number":2,"created":"2020-03-31 22:41:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/81/716181/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/81/716181/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/716181/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/716181/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/716181/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/81/716181/2"}}},"commit":{"parents":[{"commit":"16d279e21e66b842d58a05a138ee9a4c48688f59","subject":"Remove idle_timeout option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/16d279e21e66b842d58a05a138ee9a4c48688f59"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-31 07:31:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-31 22:40:56.000000000","tz":540},"subject":"Expect python3 client package in CentOS8","message":"Expect python3 client package in CentOS8\n\nIn CentOS, we expect to have python3 client package in 8.x while we\nexpect to have python2 in 7.x .\nFix unit tests to expect the correct version according to os major\nversion.\n\nChange-Id: I3692a2015ccaa79acd69679e0c749d9d6fa78bcb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/1f1b58a62ec1ee1d870e77fc14fc2086aa03d0f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/1f1b58a62ec1ee1d870e77fc14fc2086aa03d0f2"}]},"branch":"refs/heads/master"},"a7058c29216873875a8a61ca20b74fd893acf667":{"kind":"REWORK","_number":3,"created":"2020-03-31 22:42:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/81/716181/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/81/716181/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/716181/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/716181/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/716181/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/81/716181/3"}}},"commit":{"parents":[{"commit":"16d279e21e66b842d58a05a138ee9a4c48688f59","subject":"Remove idle_timeout option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/16d279e21e66b842d58a05a138ee9a4c48688f59"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-31 07:31:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-31 22:42:39.000000000","tz":540},"subject":"Expect python3 client package in CentOS8","message":"Expect python3 client package in CentOS8\n\nIn CentOS, we expect to have python3 client package in 8.x while we\nexpect to have python2 in 7.x .\nFix unit tests to expect the correct version according to os major\nversion.\n\nChange-Id: I3692a2015ccaa79acd69679e0c749d9d6fa78bcb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/a7058c29216873875a8a61ca20b74fd893acf667"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/a7058c29216873875a8a61ca20b74fd893acf667"}]},"branch":"refs/heads/master"},"01cac65a6800f75b02ef1927ca057e7489765d5c":{"kind":"REWORK","_number":4,"created":"2020-03-31 23:25:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/81/716181/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/81/716181/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/716181/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/716181/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/716181/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/81/716181/4"}}},"commit":{"parents":[{"commit":"16d279e21e66b842d58a05a138ee9a4c48688f59","subject":"Remove idle_timeout option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/16d279e21e66b842d58a05a138ee9a4c48688f59"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-31 07:31:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-31 23:25:19.000000000","tz":540},"subject":"Expect python3 client package in CentOS8","message":"Expect python3 client package in CentOS8\n\nIn CentOS, we expect to have python3 client package in 8.x while we\nexpect to have python2 in 7.x .\nFix unit tests to expect the correct version according to os major\nversion.\n\nNote that this patch also removes broken unit tests which overrides\nos release information to test el6 and el7.\n\nChange-Id: I3692a2015ccaa79acd69679e0c749d9d6fa78bcb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/01cac65a6800f75b02ef1927ca057e7489765d5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/01cac65a6800f75b02ef1927ca057e7489765d5c"}]},"branch":"refs/heads/master"},"8fb1cdacefd52e28c889cfe43b40327aebbbc21d":{"kind":"REWORK","_number":5,"created":"2020-03-31 23:26:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/81/716181/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/81/716181/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/716181/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/716181/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/716181/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/81/716181/5"}}},"commit":{"parents":[{"commit":"16d279e21e66b842d58a05a138ee9a4c48688f59","subject":"Remove idle_timeout option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/16d279e21e66b842d58a05a138ee9a4c48688f59"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-31 07:31:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-31 23:26:15.000000000","tz":540},"subject":"Expect python3 client package in CentOS8","message":"Expect python3 client package in CentOS8\n\nIn CentOS, we expect to have python3 client package in 8.x while we\nexpect to have python2 in 7.x .\nFix unit tests to expect the correct version according to os major\nversion.\n\nNote that this patch also removes broken unit tests which overrides\nos release information to test el6 and el7.\n\nChange-Id: I3692a2015ccaa79acd69679e0c749d9d6fa78bcb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/8fb1cdacefd52e28c889cfe43b40327aebbbc21d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/8fb1cdacefd52e28c889cfe43b40327aebbbc21d"}]},"branch":"refs/heads/master"},"5bd1ee6bc749a0fd804f34fcedcd3d354b5ef2f1":{"kind":"REWORK","_number":6,"created":"2020-04-01 01:28:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/81/716181/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/81/716181/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/716181/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/716181/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/716181/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/81/716181/6"}}},"commit":{"parents":[{"commit":"16d279e21e66b842d58a05a138ee9a4c48688f59","subject":"Remove idle_timeout option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/16d279e21e66b842d58a05a138ee9a4c48688f59"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-31 07:31:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-01 01:28:15.000000000","tz":540},"subject":"Expect python3 client package in CentOS8","message":"Expect python3 client package in CentOS8\n\nIn CentOS, we expect to have python3 client package in 8.x while we\nexpect to have python2 in 7.x .\nFix unit tests to expect the correct version according to os major\nversion.\n\nNote that this patch also removes broken unit tests which overrides\nos release information to test el6 and el7.\n\nChange-Id: I3692a2015ccaa79acd69679e0c749d9d6fa78bcb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/5bd1ee6bc749a0fd804f34fcedcd3d354b5ef2f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/5bd1ee6bc749a0fd804f34fcedcd3d354b5ef2f1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
