)]}'
{"id":"openstack%2Fpuppet-glance~716455","triplet_id":"openstack%2Fpuppet-glance~master~Ic13ad21831e937a981cd176e11f733bec9920289","project":"openstack/puppet-glance","branch":"master","topic":"el6","hashtags":[],"change_id":"Ic13ad21831e937a981cd176e11f733bec9920289","subject":"Drop compatibility for el6","status":"MERGED","created":"2020-03-31 23:27:51.000000000","updated":"2020-04-03 07:16:45.000000000","submitted":"2020-04-03 07:13:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716455-1585898025321-dc9227fa","meta_rev_id":"d9f3c218232db33a449963c0c07a23167b93a5af","_number":716455,"virtual_id_number":716455,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-04-03 03:09:12.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-03 07:13:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-04-02 14:00:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2020-04-02 16:36:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2020-04-02 16:36:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-02 14:00:13.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-04-02 16:36:46.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-04-03 07:13:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8a8fb1057a97502d0f920a858ae61e6205719b3a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-31 23:27:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4616c52d5b82d57a2e1fd4be0a57d25494f94aa","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-31 23:31:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f0e4cafb8b455a2228eedbe97b635731446e8b18","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-01 01:28:37.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"81caca05de1cc972578f32f34812d7ee9d79dc61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 03:26:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bfd71ecbbf754b398e2c4ce2be9a64d0 : SUCCESS in 4m 30s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe22e9c8798845dda7de36bf8a2cd2d3 : SUCCESS in 4m 15s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/87517ae6bdd9467d823eaca27ba9cfd3 : SUCCESS in 4m 19s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/7c8f6f768b714177bc35c0c57c8b22d8 : SUCCESS in 11m 57s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/7b8d293de2874daf940fed6b7548bd99 : SUCCESS in 12m 03s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/494ceebc525a4bd299bac55a46e1237f : SUCCESS in 12m 45s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fcfbf6bcbb23485eb71ea3c523a01809 : SUCCESS in 10m 32s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/8f8f7e986221484e846a25f061d55a54 : SUCCESS in 45m 11s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/819ec02d3eb54c16b4e1044231b12249 : SUCCESS in 47m 46s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c4fba14ad70b43ad9a75838ade10002f : SUCCESS in 40m 10s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/d341638d7c4448c8bab9a72b50f3ed49 : SUCCESS in 35m 53s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6ab9a0947e66459a80fabfbf0010a60a : SUCCESS in 1h 10m 56s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c5d9ec5f6dc9443eb32aa21d70f77d01 : FAILURE in 1h 00m 52s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4b558b09fa6b4ccf8166d38841de49d6 : SUCCESS in 1h 00m 08s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cf87bb122f294d3b9a873e5aa4f93727 : SUCCESS in 42m 48s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/af12f63488e34355969c99d60331c516 : SUCCESS in 19m 40s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/17c8b55bde914bd896111ad2bfe75d64 : FAILURE in 5m 33s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0e501680f67e4fce8f856e7d0eae4b8e : SUCCESS in 1h 53m 10s","accounts_in_message":[],"_revision_number":3},{"id":"57d021d8b03522671321d7bb3d7c8ffd4eaff1ff","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-01 03:56:20.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"3e762750e79f13588f7368dfe311db9e1abad1b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 05:40:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4a293916e25e4869857ca14a3b1e83b2 : SUCCESS in 6m 46s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/12513b2268c24fee880b35e92391c9ea : SUCCESS in 5m 07s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/272b457381f047d98d3cac4c10a147f5 : SUCCESS in 5m 55s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/a769dfe0ec9046758e5881947efce9e9 : SUCCESS in 12m 23s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/d695d171a8474d2899fefd70add75df3 : SUCCESS in 13m 06s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/55424a02021b4287beb644f5efcc512f : SUCCESS in 13m 35s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e8a5def4319a42be8ae8c9978b60ac0e : SUCCESS in 11m 29s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/6bb57a631afe4488aa2dc4dfa1eb406e : SUCCESS in 47m 02s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c8932c00692a48069a2a2875a1dbd9a1 : SUCCESS in 48m 35s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2d3033819afa4f27aaa1f2d0389fad99 : SUCCESS in 39m 12s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/2dc4193c189a4226915105ebb1e281cb : SUCCESS in 37m 22s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e7d8dd4272a547148b0947d46ff9ee50 : SUCCESS in 55m 21s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ae0d44ad82c0461093d6908020b6e15a : SUCCESS in 57m 42s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f642130a1ad14ed39b3ef5aba9299fee : SUCCESS in 56m 13s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/35604c7548f0414f9f27444c3eaf702b : SUCCESS in 48m 15s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/d938cc2e761c4b2a8857c6a57cacd8f0 : SUCCESS in 18m 05s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c8a971a385204abc92ff7be41fc993be : FAILURE in 6m 22s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9a585ca51e4d4f46b071e98f6074bc34 : SUCCESS in 1h 42m 19s","accounts_in_message":[],"_revision_number":4},{"id":"465c1daa927bfdbbac2dea2587ea70248060b1b1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-02 01:01:12.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"81e63fe9250e4cd29c43261eadb6e79ff8f6cc32","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-02 01:01:55.000000000","message":"Topic changed from centos8-client to el6","accounts_in_message":[],"_revision_number":5},{"id":"ca476602baba62622f0ff1baa91017dc7bf78b06","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-02 01:02:36.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"9bb2d7a564e73f1cf2d2a7cfbc39d4672b4e9ce5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 03:20:17.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6bf8a15abf7f4bf1931a30b68191f27b : SUCCESS in 5m 19s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c0e1f85b09e94e05a152aaea9cc7ac07 : SUCCESS in 4m 26s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9d32712c71ab41848c0ea0fb6e956a75 : SUCCESS in 5m 09s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/c2ae53c6fa45422f92683037c060d29a : SUCCESS in 18m 37s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/52a1ca370b4c4b9bb89cda9b02027e0f : SUCCESS in 13m 17s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/725a4168b800429fbe3f11809292206c : SUCCESS in 14m 25s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e244aa6bb0eb4312b298f7cd7b6c03d3 : SUCCESS in 11m 23s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/c2821767672e49d0806307d7d7fe9df2 : SUCCESS in 1h 05m 06s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f554f1fec4134b69b4ac9bbdc259741c : SUCCESS in 1h 05m 53s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d9dd31c78a941da882c68f406b42bbc : SUCCESS in 58m 24s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/8d8c0c6dd15c41a3a51f7191da9c10ed : SUCCESS in 42m 56s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c807b367d7cf4f7f9dd3269c0e63b2e5 : SUCCESS in 54m 36s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e58b0534433e450fbdca9b38920a6bc0 : SUCCESS in 1h 03m 27s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b4058658a0c84b8ba81f3b4a63ddfdf4 : SUCCESS in 1h 13m 58s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f3e0b13ede7b42f89361d7940149d3b1 : SUCCESS in 47m 18s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/c53e41aefa4f4924b55ce81e46494046 : SUCCESS in 29m 34s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/256efd45c1684f979ed5c605bf73709f : FAILURE in 8m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e000ea2856040f889f62086cc934dbb : SUCCESS in 4m 49s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/242b9ea899b346e3893d7069780540f4 : SUCCESS in 1h 41m 09s","accounts_in_message":[],"_revision_number":6},{"id":"026cce29be72ec084b3e0e4959ecdc26ef27782e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-04-02 14:00:13.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"7aa9ddc72a5d73e4aeb6de825799c6c6799421ee","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-04-02 16:36:46.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"b94c9f577465cd5114aeb1f41947844a37eaffdb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 17:11:41.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"3d89239ac4ca62577370e5c559333a6c90f5d52a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 19:30:09.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/87a8c02591de43b6a7af704cdb1dbf54 : SUCCESS in 5m 53s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6028f4a990a347e49d403d7cbee7fb86 : SUCCESS in 5m 54s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/50a7e38efce34d02a34965e506629ab1 : SUCCESS in 6m 13s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/8141c7aa63c34366b1b481d1099f9212 : SUCCESS in 18m 43s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/8e3c03e53fd049c1b499b1c3bd0d0abe : SUCCESS in 15m 37s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/92048e2055fe49a18c4fb663c39f9649 : SUCCESS in 16m 18s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4bc7b55ddfa9461eb0442aff9716bc2d : SUCCESS in 1h 00m 37s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/56a81d0e6a1341d2ab52c9b9e204d0ba : FAILURE in 1h 03m 17s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8c3cfcac0ebd43d6889a2d59fe51f810 : SUCCESS in 1h 13m 00s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c84641d2f8ad4b86a80e5df7e237edb8 : SUCCESS in 49m 32s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/6026c9ee3c22437abe95df39fabf7430 : SUCCESS in 21m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/022969658b8a45cda1c56affad291aca : SUCCESS in 5m 12s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/97b4420cbf6046ee9b523286f744261a : SUCCESS in 2h 00m 00s","accounts_in_message":[],"_revision_number":6},{"id":"597c26d63a65de8fe45fff257cc92885b4e20b19","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-02 23:04:01.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"c8ca2d717697cb2e768635397f81f1890e5c0059","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 01:05:16.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/298e054258a14728bffc819f7820c0e7 : SUCCESS in 6m 47s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eef61672344e40d58c1dfc67c197e70a : SUCCESS in 6m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c4fa37b4b4e34a34abaddb7b742ad9aa : SUCCESS in 6m 55s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/5859fdea785245288bf63e73dbbb4642 : SUCCESS in 15m 40s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/a34506c91cf740b4a5ca7f3f694991c2 : SUCCESS in 17m 54s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aaa5922da7794960bd77ad1051542ff0 : SUCCESS in 17m 19s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ef8e581e6954466af440753356a70a3 : SUCCESS in 13m 53s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/a164d948ceaa45a6a93068bb0a26ae66 : SUCCESS in 1h 02m 02s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b994eae32f24e13be89b1260e730195 : SUCCESS in 50m 55s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f5bd09223e7c466ca1d5569045fdccaf : SUCCESS in 40m 28s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/82182760894e40039a0f0b324b92995e : SUCCESS in 39m 13s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c9feb24154454b7b85daeb690e5d643b : SUCCESS in 58m 29s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3fd559e867734d539de717a76d46999c : SUCCESS in 1h 09m 53s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cbf08d848fea48a2a6bfe7a641fa0df3 : SUCCESS in 59m 35s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/21e3e8ffa7184e2aaf0213e15e6786cc : SUCCESS in 49m 24s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/037d20d446a046758a8635daa2eae825 : SUCCESS in 26m 19s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f3b5f0b9b07945b89576fa953b27ec6d : FAILURE in 9m 01s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6b523dd641b4ac2b3f90d8f49fc0f05 : SUCCESS in 4m 58s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f8af9e92457a484f9eb263444ffa7ded : SUCCESS in 1h 53m 59s","accounts_in_message":[],"_revision_number":6},{"id":"fa897084be02f182cfe12b073f0e4ce2917cbdcd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 01:05:27.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"0756ac1a7bf4d35bd515e167fb4c4226842332c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 03:06:10.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/45adf660914c44e88ba1eeb0937eef89 : SUCCESS in 5m 24s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db1f1eb7a5464e81b0a3dcadb668c26e : SUCCESS in 4m 57s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d4e6cb738a884fdcb32ff8c790ea1bb1 : SUCCESS in 4m 40s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/6bb0f0f226734383b0d9bfcec62a9f02 : SUCCESS in 18m 12s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/2aad033eee6d4aa8ab88050ff6aab14b : SUCCESS in 12m 54s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c6f1488b26d24eb19527e79a5dab1a47 : SUCCESS in 15m 17s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d82d0dffc63941a29cb4beffdf6b5bf6 : SUCCESS in 1h 07m 18s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d42a2069424647e19a8fabc085f2db63 : FAILURE in 1h 04m 54s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e6419f9942d34b66911c739cbae35d66 : SUCCESS in 1h 13m 27s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/44108ab463fe46738c2e1b67a1c0bf27 : SUCCESS in 47m 54s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/f4fab44e6ecc4aea8517ca5bb53942d2 : SUCCESS in 21m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b67abd3c45947d4a38e830c50de64f7 : SUCCESS in 4m 35s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7c36cf031fb24b6cb03b13498fd22806 : SUCCESS in 1h 58m 50s","accounts_in_message":[],"_revision_number":6},{"id":"0df6bfa4387257d53d764c563917014ff69e691e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-03 03:09:12.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"86daca521de678a8021e334601f1533d6f294bb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 05:17:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/63d383ba9b884db58cace58b5ff48c41 : SUCCESS in 4m 41s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca84027acf124c478391535e602f27a7 : SUCCESS in 4m 26s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce46161c34044f2c9a45033b7156418d : SUCCESS in 4m 40s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/f8b822efff754e28b9a20755f9ba9812 : SUCCESS in 21m 52s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/34ee3423690344b686b48cb2aad02cbb : SUCCESS in 17m 31s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4cc674d8b5e0455c9f509c0845f169da : SUCCESS in 10m 58s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ab89134baa24c15a13ddb4935a9a92e : SUCCESS in 13m 11s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/e4d6a01fd0bd40cb9bd06a5a474f10f0 : SUCCESS in 1h 09m 27s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25ec23dcaba9496a8471eae2e16e7790 : SUCCESS in 48m 31s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc9e17b867014cb48f4c99d911d40b16 : SUCCESS in 39m 54s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/d43b7dc66be344a9b58d76b57b745ece : SUCCESS in 45m 51s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/12c55629c6f84ccb9c52021543b58a21 : SUCCESS in 50m 36s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2c34f32d4ca44e2681e6a37014cc788a : SUCCESS in 1h 28m 08s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/56d679e0941d4d61958860d8ac854de7 : SUCCESS in 55m 25s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ad3e2e863c574b5ab86675005264ad34 : SUCCESS in 39m 31s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e86d9d1950d844c18439b09e5de175ef : SUCCESS in 33m 04s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7cf3841286d04da4a76b9d85f33fc33c : FAILURE in 7m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2a31b17bd5041b3bede0e6d1c3cab33 : SUCCESS in 4m 33s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2bb04e65fc0142239b6da02374cb93ca : SUCCESS in 2h 06m 12s","accounts_in_message":[],"_revision_number":6},{"id":"a1a8a815649a8b5d57fff61ff24d9a539d96121d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 05:17:22.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"9ab037a1efde71ec7761ffd695ed0357512878fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 07:13:45.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5872efbba1b3428e867634e77ef1019a : SUCCESS in 5m 07s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/52042c169e164871b37a2c02192f8f4a : SUCCESS in 4m 28s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/29fb133fa9b64076b01a05fcdc0960bb : SUCCESS in 5m 07s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/68f34d6822994436b66bc846f1412697 : SUCCESS in 13m 57s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/a1c352da92e34c6f9740a0f4fb951cbd : SUCCESS in 14m 06s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/352a22847e6c42febf98b063a0c55642 : SUCCESS in 13m 23s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2daedc8bfc2f4a23bcc86400dc01b5c2 : SUCCESS in 55m 10s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6a22dc010e1b4deeb470595054b258af : SUCCESS in 1h 10m 18s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e399dd1e10b54d51a989448b5892a361 : SUCCESS in 1h 00m 14s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ce140e8d31d545da8b08f53621d02958 : SUCCESS in 47m 08s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e2ee509da4ec46c0b8173058dbc5bd41 : SUCCESS in 18m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9a884a369d04f94b8783160af523a28 : SUCCESS in 3m 41s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/71438d11f2b9474a8cd562e12b9d9c94 : SUCCESS in 1h 55m 15s","accounts_in_message":[],"_revision_number":6},{"id":"4675e1850cb1a09046ae7986a6f09ee12f467ca7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 07:13:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"d9f3c218232db33a449963c0c07a23167b93a5af","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 07:16:45.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/401816acb2d546a4b74038b527e71bc5 : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"e6d89a6355dbdd51b6400cf98748d2e07b7a77fc","revisions":{"09639120173ae0d0c015575f83c9b983178e3007":{"kind":"REWORK","_number":1,"created":"2020-03-31 23:27:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/716455/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/55/716455/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/55/716455/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/55/716455/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/55/716455/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/55/716455/1"}}},"commit":{"parents":[{"commit":"8fb1cdacefd52e28c889cfe43b40327aebbbc21d","subject":"Expect python3 client package in CentOS8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/8fb1cdacefd52e28c889cfe43b40327aebbbc21d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-31 23:26:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-31 23:27:32.000000000","tz":540},"subject":"Drop compability for el6","message":"Drop compability for el6\n\n... because it\u0027s now really old, and we don\u0027t expect any users\nuse it to run OpenStack.\n\nChange-Id: Ic13ad21831e937a981cd176e11f733bec9920289\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/09639120173ae0d0c015575f83c9b983178e3007"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/09639120173ae0d0c015575f83c9b983178e3007"}]},"branch":"refs/heads/master"},"94813e88471e53d145c1d7d0f107a0ff501f9df8":{"kind":"REWORK","_number":2,"created":"2020-03-31 23:31:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/716455/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/55/716455/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/55/716455/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/55/716455/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/55/716455/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/55/716455/2"}}},"commit":{"parents":[{"commit":"8fb1cdacefd52e28c889cfe43b40327aebbbc21d","subject":"Expect python3 client package in CentOS8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/8fb1cdacefd52e28c889cfe43b40327aebbbc21d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-31 23:26:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-31 23:29:06.000000000","tz":540},"subject":"Drop compability for el6","message":"Drop compability for el6\n\n... because it\u0027s now really old, and we don\u0027t expect any users\nuse it to run OpenStack.\n\nChange-Id: Ic13ad21831e937a981cd176e11f733bec9920289\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/94813e88471e53d145c1d7d0f107a0ff501f9df8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/94813e88471e53d145c1d7d0f107a0ff501f9df8"}]},"branch":"refs/heads/master"},"11546d33fd7bbf952bf010069308430caf1bb7bd":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-04-01 01:28:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/716455/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/55/716455/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/55/716455/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/55/716455/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/55/716455/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/55/716455/3"}}},"commit":{"parents":[{"commit":"5bd1ee6bc749a0fd804f34fcedcd3d354b5ef2f1","subject":"Expect python3 client package in CentOS8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/5bd1ee6bc749a0fd804f34fcedcd3d354b5ef2f1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-31 23:26:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-01 01:28:15.000000000","tz":540},"subject":"Drop compability for el6","message":"Drop compability for el6\n\n... because it\u0027s now really old, and we don\u0027t expect any users\nuse it to run OpenStack.\n\nChange-Id: Ic13ad21831e937a981cd176e11f733bec9920289\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/11546d33fd7bbf952bf010069308430caf1bb7bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/11546d33fd7bbf952bf010069308430caf1bb7bd"}]},"branch":"refs/heads/master"},"e1c746fbed8e51502f9700af3baddfeb59abfacd":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-04-01 03:56:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/716455/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/55/716455/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/55/716455/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/55/716455/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/55/716455/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/55/716455/4"}}},"commit":{"parents":[{"commit":"5bd1ee6bc749a0fd804f34fcedcd3d354b5ef2f1","subject":"Expect python3 client package in CentOS8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/5bd1ee6bc749a0fd804f34fcedcd3d354b5ef2f1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-31 23:26:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-01 03:55:59.000000000","tz":540},"subject":"Drop compatibility for el6","message":"Drop compatibility for el6\n\n... because it\u0027s now really old, and we don\u0027t expect any users\nuse it to run OpenStack.\n\nChange-Id: Ic13ad21831e937a981cd176e11f733bec9920289\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e1c746fbed8e51502f9700af3baddfeb59abfacd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e1c746fbed8e51502f9700af3baddfeb59abfacd"}]},"branch":"refs/heads/master"},"10286d612cdbd0ac77cd78870069e2bae2120c0f":{"kind":"REWORK","_number":5,"created":"2020-04-02 01:01:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/716455/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/55/716455/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/55/716455/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/55/716455/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/55/716455/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/55/716455/5"}}},"commit":{"parents":[{"commit":"5bd1ee6bc749a0fd804f34fcedcd3d354b5ef2f1","subject":"Expect python3 client package in CentOS8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/5bd1ee6bc749a0fd804f34fcedcd3d354b5ef2f1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-31 23:26:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-02 01:01:04.000000000","tz":540},"subject":"Drop compatibility for el6","message":"Drop compatibility for el6\n\n... because it\u0027s now really old, and we don\u0027t expect any users\nuse it to run OpenStack.\n\nChange-Id: Ic13ad21831e937a981cd176e11f733bec9920289\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/10286d612cdbd0ac77cd78870069e2bae2120c0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/10286d612cdbd0ac77cd78870069e2bae2120c0f"}]},"branch":"refs/heads/master"},"e6d89a6355dbdd51b6400cf98748d2e07b7a77fc":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-04-02 01:02:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/716455/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/55/716455/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/55/716455/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/55/716455/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/55/716455/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/55/716455/6"}}},"commit":{"parents":[{"commit":"b3a0da5ba0cb4c9f3af34a43fae1421c31a3b814","subject":"Merge \"Expect python3 client package in CentOS8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/b3a0da5ba0cb4c9f3af34a43fae1421c31a3b814"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-31 23:26:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-02 01:02:27.000000000","tz":540},"subject":"Drop compatibility for el6","message":"Drop compatibility for el6\n\n... because it\u0027s now really old, and we don\u0027t expect any users\nuse it to run OpenStack.\n\nChange-Id: Ic13ad21831e937a981cd176e11f733bec9920289\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e6d89a6355dbdd51b6400cf98748d2e07b7a77fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e6d89a6355dbdd51b6400cf98748d2e07b7a77fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
