)]}'
{"id":"openstack%2Fpuppet-glance~719338","triplet_id":"openstack%2Fpuppet-glance~master~I4a7c9ac97ff2f1c04b651ecbc89b3cb1c583f2f9","project":"openstack/puppet-glance","branch":"master","topic":"glance-os_region_name","hashtags":[],"change_id":"I4a7c9ac97ff2f1c04b651ecbc89b3cb1c583f2f9","subject":"Add cinder_os_region_name parameters for cinder backend","status":"MERGED","created":"2020-04-12 14:45:01.000000000","updated":"2020-04-14 08:00:35.000000000","submitted":"2020-04-13 05:25:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"719338-1586755558555-354c625b","meta_rev_id":"66a95fdbea3c33e43b9dc9271cb397b5155e948d","_number":719338,"virtual_id_number":719338,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-13 05:25:58.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-04-13 02:47:23.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-04-13 03:47:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-04-13 03:47:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-13 02:47:23.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-04-13 03:47:01.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-04-13 05:25:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4cf69cf79242b7917be273ba8d2ba8009ebed6f2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-12 14:45:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"585bfbd75f527c1602e7fb752284737c586ffb7e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-12 14:45:56.000000000","message":"Topic set to glance-os_region_name","accounts_in_message":[],"_revision_number":1},{"id":"f662d36d3dcca624fb52d775af703a7a95350b6b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-12 14:55:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0ac7b55f6f7df4b5a8499e4fc7104c5dd5e68585","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-12 15:05:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3f2b9787d73ab4b59c29d1018fc8487b3d902c5c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-12 15:20:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f60c6c1fa0e2116fe91d043676c3cb0185dba388","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-12 16:53:35.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b47404968fb4f6c8a97d80c9040efd5 : SUCCESS in 3m 49s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6e38a9480bf4472fab3c2cc4d436cbbf : SUCCESS in 4m 11s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9da925a7aa9b45ef91d1bc42db919349 : SUCCESS in 3m 44s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/b48608e865a1430fa0b5121ee54d5401 : FAILURE in 11m 32s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/03cc6238b2e34b20828966e67a9f6721 : FAILURE in 11m 12s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1049deef476d4da6af6166ee429c749c : FAILURE in 8m 21s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/37dccd8dcd504e5180858740daf54996 : FAILURE in 10m 00s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/c0349f7949424bd392c81df6e9c6c76c : SUCCESS in 44m 17s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b6bf694b737846b6b084b6935a1dd529 : SUCCESS in 54m 54s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2379e66324fc47e486699daa3e17533f : SUCCESS in 56m 13s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/99152b81bfaf49cd866f017dea3a81cf : SUCCESS in 40m 58s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e62b91af9c884c0fb48f74a7a566e68a : SUCCESS in 55m 58s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/193d6be491ae40f4921a35a21a850ec0 : FAILURE in 1h 11m 57s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1b299186e10248ea9f6464332c36e45a : SUCCESS in 1h 01m 39s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/60e9b9e7b0b4421f879cd822362ed2c3 : SUCCESS in 45m 11s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9aa8562ec2254c6a8e64843341711e2b : SUCCESS in 19m 07s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4bb0828dac834b65b7f0414f6a800cbe : FAILURE in 6m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a430d59db4a409184dadf18def8c9d8 : SUCCESS in 2m 54s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/487716c606814768a57a43b12fa88a98 : SUCCESS in 1h 32m 08s","accounts_in_message":[],"_revision_number":4},{"id":"ee1e0026335a2c9ab7266027a28daba51000d07f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-12 23:07:37.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c3ffcda0a4ce99a83127c4b06a729f7e8664aa31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 00:32:26.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/82f0cbe49e9147c39bda7617e99f2a61 : SUCCESS in 5m 08s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bacd6dabcf73418ba1d9da200799b8f5 : SUCCESS in 4m 51s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/91df5f01a76d4c0abd0fdb0821ca26ed : SUCCESS in 4m 31s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/9c72e7645e9c404dbf4b33648a6d8f74 : SUCCESS in 13m 14s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/b7819b37e2b4421ea79af6ba674bbf98 : SUCCESS in 13m 02s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/06b74991f5d64d5bb1c6715f204d59f6 : SUCCESS in 10m 52s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/06b5d1ee566e45fcaa329849d0829170 : SUCCESS in 11m 39s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/3a895b2d7e98448a8f1741b8b1677c50 : SUCCESS in 1h 11m 28s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d6b84a86fd04466aa5019ab052d4dd3 : SUCCESS in 1h 15m 07s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f6d9511665af40d39876f57a775e5ecb : SUCCESS in 43m 31s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/6f6d25297de843f182dffb4489031291 : SUCCESS in 36m 21s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/35b430deead7426faf9ec1b31489db11 : SUCCESS in 55m 39s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a86f1b0186a1473f9f0d70c1ddf447c8 : SUCCESS in 59m 54s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9187717a99e74a7992d4edbe755f359e : SUCCESS in 59m 50s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9351bc1445124ba99851e495b14bead4 : SUCCESS in 41m 42s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/b78745da0e7b42d7853fdced4f697714 : SUCCESS in 21m 02s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a0550a7d48044189a2be095db510733 : FAILURE in 6m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfd0036ff8744414b54cfa5a8a947718 : SUCCESS in 3m 27s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/910273c766cc4a09b42d566111f03d5b : SUCCESS in 1h 23m 20s","accounts_in_message":[],"_revision_number":5},{"id":"d132e9f4de758857faa7c47dd6c490546af88762","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-04-13 02:47:23.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"6857ccfc312b701ad0761c536b3e1eac3fd8d2f6","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-04-13 03:47:01.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"0845a73cc9bb48214c5eb554ddd18af814b88616","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 03:47:12.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"9d81881c2d560d06dff05c285f0fe56f7e797bd2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 05:25:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"d2ff89b479d9efc1c0ad967e6ad0504d6e00cd07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 05:25:58.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/afbbfc132d4c4bbd9537a142c66977a5 : SUCCESS in 4m 16s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/226228e7a22844508853711d7c497fc7 : SUCCESS in 3m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/902f93cc323247c49fb8bb1d2067b96e : SUCCESS in 4m 06s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/b4051b1e40414fc1b4e4c0ed7fff8e3b : SUCCESS in 15m 11s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/5b6f8cd1d19541b0a7bfd2b81ea812c0 : SUCCESS in 14m 53s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5b6fb84fe1d2469f888d1d7ac4fb322c : SUCCESS in 10m 01s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fbf58255fb4e411e9d3c91d283966dcb : SUCCESS in 50m 53s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e62075f6dc0148e5aba367250e275da3 : SUCCESS in 1h 00m 20s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4952f9c37efe409c8fd3720389eff7c3 : SUCCESS in 1h 01m 00s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/06ae1805d95a480aa6e02d67ed0d348b : SUCCESS in 46m 38s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/bc579138aed84dfe98fbc24061cee5da : SUCCESS in 18m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/565fe9c80fdc481fa1893f5d0ea931da : SUCCESS in 3m 35s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a55300e22db045c489ac8ff709b82519 : SUCCESS in 1h 34m 09s","accounts_in_message":[],"_revision_number":5},{"id":"9bd25c35ce1aa1e09fe0c0624ddf150da7625d70","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 05:27:05.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47799ab11f7a455297f49cf3d0aff457 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":5},{"id":"66a95fdbea3c33e43b9dc9271cb397b5155e948d","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-04-14 08:00:35.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"9e82f598ad86d99a5bc356ba5c118a52a1d5d9bc","revisions":{"5f201ba03356c3db02acca9c6297da8cb41590bc":{"kind":"REWORK","_number":1,"created":"2020-04-12 14:45:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/719338/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/38/719338/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/38/719338/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/38/719338/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/38/719338/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/38/719338/1"}}},"commit":{"parents":[{"commit":"5d2acead72c28910c7712b49bb90ff5f441a79e9","subject":"Add parameters for multipath connection in cinder backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/5d2acead72c28910c7712b49bb90ff5f441a79e9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-12 14:41:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-12 14:44:35.000000000","tz":540},"subject":"Add cinder_os_region_name parameters for cinder backend","message":"Add cinder_os_region_name parameters for cinder backend\n\nThis patch introduces cinder_os_region_name parameters for cinder\nbackend, because global os_region_name parameter was replaced by this\ncinder specific parameters[1].\n\n[1] 410747822c9433c83fd4aec547d3e4c288502f90\n\nChange-Id: I4a7c9ac97ff2f1c04b651ecbc89b3cb1c583f2f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/5f201ba03356c3db02acca9c6297da8cb41590bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/5f201ba03356c3db02acca9c6297da8cb41590bc"}]},"branch":"refs/heads/master"},"2239aadd75d91280081833fbb4c3297d71af9cc7":{"kind":"REWORK","_number":2,"created":"2020-04-12 14:55:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/719338/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/38/719338/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/38/719338/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/38/719338/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/38/719338/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/38/719338/2"}}},"commit":{"parents":[{"commit":"5d2acead72c28910c7712b49bb90ff5f441a79e9","subject":"Add parameters for multipath connection in cinder backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/5d2acead72c28910c7712b49bb90ff5f441a79e9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-12 14:41:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-12 14:54:55.000000000","tz":540},"subject":"Add cinder_os_region_name parameters for cinder backend","message":"Add cinder_os_region_name parameters for cinder backend\n\nThis patch introduces cinder_os_region_name parameters for cinder\nbackend, because global os_region_name parameter was replaced by this\ncinder specific parameters[1].\n\n[1] 410747822c9433c83fd4aec547d3e4c288502f90\n\nChange-Id: I4a7c9ac97ff2f1c04b651ecbc89b3cb1c583f2f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/2239aadd75d91280081833fbb4c3297d71af9cc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/2239aadd75d91280081833fbb4c3297d71af9cc7"}]},"branch":"refs/heads/master"},"9b13e41d0c6fe2f135028ed5f97287fbb8dd54c2":{"kind":"REWORK","_number":3,"created":"2020-04-12 15:05:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/719338/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/38/719338/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/38/719338/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/38/719338/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/38/719338/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/38/719338/3"}}},"commit":{"parents":[{"commit":"5d2acead72c28910c7712b49bb90ff5f441a79e9","subject":"Add parameters for multipath connection in cinder backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/5d2acead72c28910c7712b49bb90ff5f441a79e9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-12 14:41:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-12 15:05:08.000000000","tz":540},"subject":"Add cinder_os_region_name parameters for cinder backend","message":"Add cinder_os_region_name parameters for cinder backend\n\nThis patch introduces cinder_os_region_name parameters for cinder\nbackend, because global os_region_name parameter was replaced by this\ncinder specific parameters[1].\n\n[1] 410747822c9433c83fd4aec547d3e4c288502f90\n\nChange-Id: I4a7c9ac97ff2f1c04b651ecbc89b3cb1c583f2f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9b13e41d0c6fe2f135028ed5f97287fbb8dd54c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9b13e41d0c6fe2f135028ed5f97287fbb8dd54c2"}]},"branch":"refs/heads/master"},"0c75d208b59bb7f0f57fba1d95c108b703210b5d":{"kind":"REWORK","_number":4,"created":"2020-04-12 15:20:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/719338/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/38/719338/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/38/719338/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/38/719338/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/38/719338/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/38/719338/4"}}},"commit":{"parents":[{"commit":"5d2acead72c28910c7712b49bb90ff5f441a79e9","subject":"Add parameters for multipath connection in cinder backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/5d2acead72c28910c7712b49bb90ff5f441a79e9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-12 14:41:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-12 15:20:17.000000000","tz":540},"subject":"Add cinder_os_region_name parameters for cinder backend","message":"Add cinder_os_region_name parameters for cinder backend\n\nThis patch introduces cinder_os_region_name parameters for cinder\nbackend, because global os_region_name parameter was replaced by this\ncinder specific parameters[1].\n\n[1] 410747822c9433c83fd4aec547d3e4c288502f90\n\nChange-Id: I4a7c9ac97ff2f1c04b651ecbc89b3cb1c583f2f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/0c75d208b59bb7f0f57fba1d95c108b703210b5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/0c75d208b59bb7f0f57fba1d95c108b703210b5d"}]},"branch":"refs/heads/master"},"9e82f598ad86d99a5bc356ba5c118a52a1d5d9bc":{"kind":"REWORK","_number":5,"created":"2020-04-12 23:07:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/719338/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/38/719338/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/38/719338/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/38/719338/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/38/719338/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/38/719338/5"}}},"commit":{"parents":[{"commit":"5d2acead72c28910c7712b49bb90ff5f441a79e9","subject":"Add parameters for multipath connection in cinder backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/5d2acead72c28910c7712b49bb90ff5f441a79e9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-12 14:41:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-12 23:07:25.000000000","tz":540},"subject":"Add cinder_os_region_name parameters for cinder backend","message":"Add cinder_os_region_name parameters for cinder backend\n\nThis patch introduces cinder_os_region_name parameters for cinder\nbackend, because global os_region_name parameter was replaced by this\ncinder specific parameters[1].\n\n[1] 410747822c9433c83fd4aec547d3e4c288502f90\n\nChange-Id: I4a7c9ac97ff2f1c04b651ecbc89b3cb1c583f2f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9e82f598ad86d99a5bc356ba5c118a52a1d5d9bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9e82f598ad86d99a5bc356ba5c118a52a1d5d9bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
