)]}'
{"id":"openstack%2Fpuppet-glance~720618","triplet_id":"openstack%2Fpuppet-glance~stable%2Ftrain~I1dfcb311c10c633788c24484c21277255b60e4e5","project":"openstack/puppet-glance","branch":"stable/train","topic":"bug/1872691","hashtags":[],"change_id":"I1dfcb311c10c633788c24484c21277255b60e4e5","subject":"Fix outdated get_glance_crednetials in Puppet::Provider::Glance","status":"MERGED","created":"2020-04-16 23:39:38.000000000","updated":"2020-04-17 21:12:23.000000000","submitted":"2020-04-17 21:12:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720618-1587157943009-5f04ccdf","meta_rev_id":"666a64af2133972b135f344a285a23382ba1ede5","_number":720618,"virtual_id_number":720618,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-17 21:12:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-04-17 18:51:31.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-04-17 13:43:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-04-17 18:51:31.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-17 13:43:05.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-04-17 18:51:31.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-04-17 21:12:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c529e40272189b0cb229a7314df529bd21826409","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-16 23:39:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b7b1631a55ca48734d6e7554b91a182504622262","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 01:38:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce0dac921edf45678c06e210a1c9c6af : SUCCESS in 4m 09s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/69c046d2c13541d7ae24934d21f0720e : SUCCESS in 4m 07s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/64b8c741dca340f3be7c7cd9fcc60f25 : SUCCESS in 4m 09s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/abc91fab068d4a49bf80de61b9809f55 : SUCCESS in 22m 46s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/591ebf3fecde4c219d3777753e56796e : SUCCESS in 16m 21s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/32d298c9cef046a594aaf07589f4aa6d : SUCCESS in 15m 42s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d13394eff041421da137dec554beb0bc : SUCCESS in 10m 42s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b022ec46a8d44f1a9e3a4b6533b2f76 : SUCCESS in 11m 51s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/b2b2ddb8e829498e9eb6c7e9f679a757 : SUCCESS in 56m 21s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/f94ac4ecf58a4a9fa2b6dfb02b376b8b : SUCCESS in 1h 05m 35s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/4901562da018458eb5154b0493d2632c : SUCCESS in 1h 01m 59s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/22c5c5ecdd5d4764a4399c97df15b793 : SUCCESS in 51m 46s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/2ca57c65a43c49bf9faf4d6145818e74 : SUCCESS in 53m 48s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ed26a832a184a67bd3d35c490031fc0 : SUCCESS in 54m 34s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c1d75ab2c801409e92cb28550bd59666 : SUCCESS in 39m 46s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/8e307dbd5d6440c0b9f778b2fe86de0f : SUCCESS in 43m 07s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e4652391050549319b37377a53d77437 : SUCCESS in 55m 45s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/39bf227ae84643dc80053ce0729f7b7f : SUCCESS in 1h 00m 56s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5e370e6baddf4024a79d32e6e21c6246 : SUCCESS in 54m 45s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8ecfdea2f96e46488170288bbffbdf8f : SUCCESS in 44m 06s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1a41b333b0464c47b6c07de6cf1c3f67 : SUCCESS in 22m 11s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/76b0dd80691a47d2b043d41d0049b049 : FAILURE in 20m 05s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/7c3fa161cd844eddab7b20a6b6ab9515 : SUCCESS in 1h 53m 50s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/83178fa0aa2b46a0bb8814a67ffcdd6d : SUCCESS in 1h 50m 44s","accounts_in_message":[],"_revision_number":1},{"id":"48a16f3cff9cbd9ce21226995a9a65764d569621","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-04-17 09:26:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"32459c5374e60f31c5e1abbc9427408b3efa2388","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-17 09:44:04.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"71aab85e8acbd009cde868461eca81934ecc9f4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 11:50:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/74bced0ca9424b1fa96a1479141a9680 : SUCCESS in 5m 57s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/120f64a5fd444201b5bfdcd7ee9f0fca : SUCCESS in 5m 16s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/344e4e0ffbf7483187b8b4642db21347 : SUCCESS in 5m 07s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/20f7fffd532046e89355e3406e3140a8 : SUCCESS in 13m 47s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/b7a713c7e26c47619ab174cdc79ac314 : SUCCESS in 16m 10s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/ecc390ffbd464a11a085b5b082530d1c : SUCCESS in 13m 25s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e736211f178045f98022fa95819b3ae7 : SUCCESS in 12m 31s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/49d48dde89ad4ed6ab73defd6baf4004 : SUCCESS in 12m 59s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/dac2c0e05c3047d9b9719c997e929cef : SUCCESS in 55m 20s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/bef803c2f2f1491a9954e2f4c578a0a3 : SUCCESS in 1h 02m 34s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/3541577da01048419997ccbba3ee5d32 : SUCCESS in 1h 04m 51s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/c0c950496d484a0c862f553caa81e70e : SUCCESS in 50m 01s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/76a5d876a9b14269916e836837b24269 : SUCCESS in 43m 37s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b061c4e98f8b4c0da5eb879bdad89a0b : SUCCESS in 45m 14s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/578d07e0fb9646a0971e3e9b6bdaa7c8 : SUCCESS in 35m 18s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/230941ee441146d4a96fd80d3cedcdfd : SUCCESS in 59m 59s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e4611742f411464c8816fb7081acfd7f : SUCCESS in 1h 02m 55s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/17e9c4bb03544401844937b068103371 : FAILURE in 1h 05m 58s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b5830f2a9cca44459709869e358913c1 : SUCCESS in 1h 16m 39s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c6ef4a0b8fd048368ef564fc26a5d00c : SUCCESS in 42m 08s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a23f33f5103e45478b4838690ecd2094 : SUCCESS in 21m 16s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/17ffcdcc905b45d580b795e5d8a4816c : FAILURE in 18m 46s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/e7d7bb23c9f347c888afabd70edd7862 : SUCCESS in 1h 57m 28s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/af21e69f0ceb41a4b625c34d9b70c034 : SUCCESS in 2h 00m 05s","accounts_in_message":[],"_revision_number":2},{"id":"ddd9e6b4393aa698d76cc3a0890bed6d7e43d9ef","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-04-17 13:43:05.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c26b9b7f03b06c1c20ae5551ce815fd38e506175","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-04-17 18:51:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0a86a6d7610cb75ba0f11b0a6395f04a38be0c8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 18:51:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"86ea83469460706c6724a0dfae9a5545dcb2cd6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 21:12:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9040c64c454b42eb9aa7508267ccc817 : SUCCESS in 5m 50s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba7151adea5f4e7786c0002a9f4ee954 : SUCCESS in 5m 56s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aed87320879a448da26ce8cbf7952c70 : SUCCESS in 4m 51s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/de1dbf5097f24d7aaf0a0bd420720242 : SUCCESS in 13m 03s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/72cfc33565264622bc551d929c761c58 : SUCCESS in 11m 21s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/ee5f5b68beee459685381f5ebf3fa646 : SUCCESS in 1h 13m 52s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/e941935d46cf4c39b264f8886cfc185a : SUCCESS in 54m 02s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/d2677142d52f46b1bf539dbcf4a24570 : SUCCESS in 56m 45s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/e655c51e08204a349ea06da367df9131 : SUCCESS in 45m 18s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/4022279dcbc54e3f8b9f04e3da6fe20f : SUCCESS in 18m 58s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/2ca02578f6f840dcabb86feec424fdc1 : SUCCESS in 2h 18m 29s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/dad0f8c6b1644e55a2fce0f4f56e6fff : SUCCESS in 1h 48m 25s","accounts_in_message":[],"_revision_number":2},{"id":"1df9c3dfe1a34b80f3cd5cf9e29fab27a1773d75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 21:12:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"43f3d5eab2b282d5baa6449f775ed64ee8eb59d6","revisions":{"35bf678c2d8bc04be0735fdf38f60fbe3613560c":{"kind":"REWORK","_number":1,"created":"2020-04-16 23:39:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/720618/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/18/720618/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/18/720618/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/18/720618/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/18/720618/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/18/720618/1"}}},"commit":{"parents":[{"commit":"a9b1a1284eff7a77e5d6c9fd3b3fa776e01a304f","subject":"Fix rabbitmq dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/a9b1a1284eff7a77e5d6c9fd3b3fa776e01a304f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-14 08:20:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-16 23:37:28.000000000","tz":540},"subject":"Fix outdated get_clance_crednetials in Puppet::Provider::Glance","message":"Fix outdated get_clance_crednetials in Puppet::Provider::Glance\n\nThis patch is a cherry pick of the change in master[2] to update\noutdated provider implementation, and introduce the implementetion\nconsisntent with other modules like puppet-nova or puppet-neutron.\n\n[1] 066f45758d974bb6343c0b92583c8185ef588823\n\nCloses-Bug: #1872691\nChange-Id: I1dfcb311c10c633788c24484c21277255b60e4e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/35bf678c2d8bc04be0735fdf38f60fbe3613560c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/35bf678c2d8bc04be0735fdf38f60fbe3613560c"}]},"branch":"refs/heads/stable/train"},"43f3d5eab2b282d5baa6449f775ed64ee8eb59d6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-04-17 09:44:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/720618/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/18/720618/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/18/720618/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/18/720618/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/18/720618/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/18/720618/2"}}},"commit":{"parents":[{"commit":"a9b1a1284eff7a77e5d6c9fd3b3fa776e01a304f","subject":"Fix rabbitmq dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/a9b1a1284eff7a77e5d6c9fd3b3fa776e01a304f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-14 08:20:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-17 09:43:58.000000000","tz":0},"subject":"Fix outdated get_glance_crednetials in Puppet::Provider::Glance","message":"Fix outdated get_glance_crednetials in Puppet::Provider::Glance\n\nThis patch is a cherry pick of the change in master[2] to update\noutdated provider implementation, and introduce the implementetion\nconsisntent with other modules like puppet-nova or puppet-neutron.\n\n[1] 066f45758d974bb6343c0b92583c8185ef588823\n\nCloses-Bug: #1872691\nChange-Id: I1dfcb311c10c633788c24484c21277255b60e4e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/43f3d5eab2b282d5baa6449f775ed64ee8eb59d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/43f3d5eab2b282d5baa6449f775ed64ee8eb59d6"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
