)]}'
{"id":"openstack%2Fpuppet-glance~723954","triplet_id":"openstack%2Fpuppet-glance~stable%2Fussuri~I6b12f2c0b4592e1faba93c2f47f77627db848831","project":"openstack/puppet-glance","branch":"stable/ussuri","topic":"create-ussuri","hashtags":[],"change_id":"I6b12f2c0b4592e1faba93c2f47f77627db848831","subject":"Update .gitreview for stable/ussuri","status":"MERGED","created":"2020-04-28 14:51:36.000000000","updated":"2020-05-09 11:55:12.000000000","submitted":"2020-05-09 11:55:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723954-1589025312012-1d1e633c","meta_rev_id":"6ec7b0ff4969523d2eeccf2d9e52e9514cb3b96b","_number":723954,"virtual_id_number":723954,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-05-09 09:26:31.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"date":"2020-05-09 06:39:41.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-09 11:55:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-04-29 02:59:01.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-04-28 16:45:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-04-29 02:59:01.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-28 16:45:05.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-04-29 02:59:01.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-05-09 06:39:41.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-05-09 09:26:31.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-05-09 11:55:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e656ba1c06a0afccf53e28d79fdbdf452c24b009","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-28 14:51:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"97485881c73776e0a1983dba1b788cd7992e66a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 16:18:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/31070a2393354d2dbcdce808c77f7145 : SUCCESS in 5m 26s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a0f7f1e81974494b80889508431dc4f0 : SUCCESS in 8m 05s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/60990f4e9d6e4436a8695b54b3350e59 : SUCCESS in 7m 51s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/4b8f9cc2ef3d4aba94897df2a9ca38d3 : SUCCESS in 15m 19s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/a020b2e9f16946ef83133ed4b907d10e : SUCCESS in 16m 00s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/26aa90c8421b46d5ab10121f696278ef : SUCCESS in 13m 46s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/13aa8bfe7ae94c649d9249e8049d1b6c : SUCCESS in 15m 07s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/5bae03e6216141f89f7c4c8e4f005b9c : SUCCESS in 51m 12s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/edbbe5c456e34dc9bcfd315bc905501a : SUCCESS in 51m 06s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dc1ee98216e84d6c9daf8bae30bee2f9 : SUCCESS in 45m 13s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/42acafd4afd246eebe9fedf631ebe4a1 : SUCCESS in 57m 42s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/83e18d17dcde4a569c0e13237825a780 : SUCCESS in 58m 46s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/db3650309165447080c7def3f8f65ff0 : SUCCESS in 1h 09m 39s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9d3a725617da44b3a3bfa701caef616d : SUCCESS in 1h 06m 51s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/34ba85a03e9c4754bc22976d6e9058a3 : SUCCESS in 44m 35s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/336b966d337f4bb299a1ddf13dda38a6 : SUCCESS in 21m 47s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/cb4d2bac5b12476f85f4b9b209de0f61 : FAILURE in 10m 56s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6e2364030eaa406596d9bbcc5bfc2687 : FAILURE in 8m 37s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/cb5cd5229f6442aaa0b84a20a602717b : FAILURE in 12m 13s","accounts_in_message":[],"_revision_number":1},{"id":"3e3aac4e12d706d5f037f965e1e8d0ab145fb318","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-04-28 16:45:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5280ad26fcf2d477e647fc64b32777370a415c52","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-04-29 02:59:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3df95a621aa051fcc8d09a2f8fa7e0a77495476a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 05:51:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ad8eab79679a4e39a3d8fa22c1400431 : SUCCESS in 4m 07s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/930bc6491a9541a691a569eb126e1678 : SUCCESS in 7m 32s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fd55022855a44bbd870a29ccdaef544f : SUCCESS in 4m 22s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/a8166fa785e643feb9d4da2dbd33b460 : SUCCESS in 22m 35s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/e57ad8a87b2d4fd58b658cf455f47017 : SUCCESS in 13m 57s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/96b46b48d9b54d1194557cf794fe17be : SUCCESS in 17m 41s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1b71e5fbce8f4a0fa7d326ce17e5306d : SUCCESS in 13m 30s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/8b431a67dcdf452f9d4eb675d6e25ab4 : SUCCESS in 1h 20m 28s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c65ce7e58734d1db00100b06052debf : SUCCESS in 50m 35s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe3c16fbdbe34d9c96ed0f5affd9dd7f : SUCCESS in 1h 15m 24s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/4242706515ad43d48370f75bb90a0794 : SUCCESS in 37m 54s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a1a5de73216d4166916f40f58f89a507 : SUCCESS in 1h 26m 20s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c8930fd2bcc84f1990eb20fa99957890 : FAILURE in 1h 04m 19s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/86faa1e20d604a6db5c8f3c41829ea2f : SUCCESS in 1h 36m 41s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fdfe731f785f4fdd8992ae2d81d3144e : SUCCESS in 44m 47s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/5b397d6327d54e8f98eed0a50d1dfcb0 : SUCCESS in 18m 27s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/dd4a222522fe4a50b2c1602c8f6895cd : FAILURE in 12m 14s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b5841723388c416fbbee845cbf106d89 : FAILURE in 7m 38s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/804a7fc86df3462ba81f7f0aa35071a1 : FAILURE in 11m 48s","accounts_in_message":[],"_revision_number":1},{"id":"919dc3a9fa629de766646e3f72b81e982b92e4a3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-29 15:12:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cee95e6d33af5b0a5a468cfd62b1390ce9c929df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 16:24:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4dd2595f00cb426db2a39c6a7b861a52 : SUCCESS in 4m 51s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5520588248f048c59f84d34c3e457550 : SUCCESS in 4m 37s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4e173b583e954cc78f2bf145d3d86e55 : SUCCESS in 4m 33s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/24941edc356a4d7e94afbae608b03fd0 : SUCCESS in 12m 48s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/43115b85873849b89426e8413ee13e17 : SUCCESS in 15m 48s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da9c920228744e0081cea67505485e1f : SUCCESS in 12m 05s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/10a1cc0fba884a82a214c969ba88b7ab : SUCCESS in 11m 50s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/f3950c83428a43d19b7fb9915f8824b7 : SUCCESS in 45m 35s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f517292fa97847edbcdb9f8a3c4b434d : SUCCESS in 47m 10s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fcb68b18e68b4919927f9a54c447a536 : SUCCESS in 50m 06s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/544b1230c2f04d59b828ff287c4b7034 : SUCCESS in 40m 31s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/91e864899a1642199ecb7ef115cad445 : SUCCESS in 59m 55s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6105ac08c73a4f4696d72209461bcae2 : FAILURE in 37m 48s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/55e2ae5c6b02478aa1e5828396afb7e9 : SUCCESS in 1h 05m 12s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/18c337e22b444d9c977b00a861411ce0 : SUCCESS in 44m 04s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/fb41592509ff432dad11e8b88f3cd364 : SUCCESS in 20m 07s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/72047adfe81c4a09aa204c75b646e0c2 : FAILURE in 9m 32s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/489d6c6fba0e4d329b60e3730f35d324 : FAILURE in 8m 16s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9fa7ac2726d64caf8fa33288810a750f : FAILURE in 10m 18s","accounts_in_message":[],"_revision_number":1},{"id":"70fdf4f2b0eead9d4cd83f332d8fbcc95b7ecd65","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-29 16:25:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"76a83472265294994b7f234da8f0afb65a8e2a85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 17:45:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b2e86eedf3947128b64f0c9e536a59a : SUCCESS in 4m 59s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7aeb93bee9b04945bee0e3ba2ef955c2 : SUCCESS in 5m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/07fa8ad8e75047da8846b46f7d464aa7 : SUCCESS in 6m 19s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/045ba3c6a84440ccab03f4bf7664b527 : SUCCESS in 15m 56s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/fe0c72add2c04f998cd3edf28f68d234 : SUCCESS in 14m 26s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e68a484e678b442d994a1157bc8bbf08 : SUCCESS in 14m 15s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f4576487a47141b69ed025b8663b23fe : SUCCESS in 16m 31s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/e3010678fdc04b549900767797af40c3 : SUCCESS in 48m 15s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/abea3b77c59c415e8aa8bdc4eb8d79d2 : SUCCESS in 48m 25s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6ba43a46fb3f4098aa82ddc20e38974b : SUCCESS in 45m 15s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/e510da8d2c7a4465997afa18a1665967 : SUCCESS in 37m 01s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/94b5f86645cb4c63a7fe1dab63294f1c : SUCCESS in 1h 04m 19s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b3e03ccf6d8d4eb9bb253bd9dfb5bef6 : FAILURE in 45m 49s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c99734d9d2e443e4b97ef721f402cdc7 : SUCCESS in 1h 16m 35s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8733f3041eed43619ba92b42a4de225c : SUCCESS in 1h 00m 58s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a833956bf456481dbbdd6754ac1f5319 : SUCCESS in 21m 56s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/dd63e17b4e284288b49f4bd95195163e : FAILURE in 9m 42s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ab9f71dc4b8a4ce49e894392e00c0b64 : FAILURE in 6m 48s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3bb9276ff6784fa18f367285059891d1 : FAILURE in 12m 31s","accounts_in_message":[],"_revision_number":1},{"id":"94aa7d3c3d8152ba0e0be45398fd2dc65754c32b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-05 16:05:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"da4eb77fad6e0e53d0c0e25a3692d937efeec36a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 17:43:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8918694f2b014c40955c14c6240790bb : SUCCESS in 5m 31s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/811f660f55de4e99a24d5b3aff5f1ed4 : SUCCESS in 7m 47s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c1a55a9d5e884d2399f1a67dcf765654 : SUCCESS in 7m 55s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/7500cd85d8ad4767848bef14fa551d1a : SUCCESS in 18m 33s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/94f2f1f926134014bab36a9846d93937 : SUCCESS in 16m 28s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d90c6d24901d406ab0552da6ebdb1984 : SUCCESS in 11m 57s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25f6f914af284681925b08ffdb7c03b1 : SUCCESS in 14m 01s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/c417865f30fe44ad8f2181b7cd6cbe84 : SUCCESS in 46m 41s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/73766f1c23bc4905bf64d7f8335c13f0 : SUCCESS in 48m 03s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/de67f48a7d894a608222781cd6693e87 : SUCCESS in 43m 20s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/129f5cae41a2461aa1d235f7b66f48b2 : SUCCESS in 36m 27s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5229aa385cdf4e00b95a324b7e42db6c : SUCCESS in 1h 07m 51s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/61e77924df3f4715bec085fd860584b7 : FAILURE in 1h 19m 11s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5adbeaa441d2484c9d92421d955db63f : SUCCESS in 1h 25m 07s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d8f3a312971b4b92b6bd62d586e62d8c : SUCCESS in 44m 18s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/19571e5a7b2b4bd2826e77f9f1531bb1 : SUCCESS in 22m 58s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/211bcec85da04a4db2e897acce70a792 : FAILURE in 8m 49s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/56fb8da846d747f79357d07d4f91b046 : FAILURE in 8m 39s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0b89de8e98a14f6f85206197d4aaaffe : FAILURE in 32m 51s","accounts_in_message":[],"_revision_number":1},{"id":"0b9b4c1d3db90fc7e3c38166eb1967b1663bafc5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-09 02:06:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c519d2d60117b3cfe35c7f2250208563d54479f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 03:22:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9d1729370a9147659bc8ad25bcdbc24b : SUCCESS in 5m 52s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c5190a039f4549848e0d30fd25ec7fc4 : SUCCESS in 4m 15s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/23a4d1f286fc4d7b91f50ec2d5d01acf : SUCCESS in 5m 47s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/f0315105da5242a9b4f6167502fe68e3 : SUCCESS in 12m 08s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/1c281c424b274068a1c49d6c9a45df6f : SUCCESS in 12m 34s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/488ee8e4d8d04bd4af7bee8f26ef237a : SUCCESS in 13m 02s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e2370f93e874e5994898c43ad747dc3 : SUCCESS in 11m 12s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/181ce26fbb7a46829352403a382072e2 : SUCCESS in 43m 56s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/728a6b9812e34536adc8d1cc70fcea40 : SUCCESS in 44m 51s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/148afa3dd2ac4f699e0b8a57b0ec2f9b : SUCCESS in 46m 56s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/7fbe9993015f4210b9194410111a0cf2 : FAILURE in 36m 32s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e9d7c872574f460aadbcc6aa1e0a0371 : SUCCESS in 59m 24s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/315cd17c6ed04378871aff8577b4badb : FAILURE in 1h 03m 17s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8e69b102b9b742b5bafecae4f132e67e : SUCCESS in 1h 14m 42s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9bdab9fe8a0e4f2eb6bc2e78d4860c29 : SUCCESS in 47m 17s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ff0a2750e55c461a9a740b5982086a8e : SUCCESS in 24m 33s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/3fa399525345426cacbec3c2abbd9bb6 : FAILURE in 10m 10s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c6450fd2e6644a63a24adcbb55981424 : FAILURE in 7m 04s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/070c1acbc114448591ab954e7ae42cbb : FAILURE in 29m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1d50944eebdd76bf836a16725705b33b9469b827","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-09 03:46:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"847bf2ec1283f66692f157fe86b778e74613a5af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 04:56:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/106a4d11eb6e4b90bd47972f5c89646b : SUCCESS in 4m 33s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2eb6c8f6320944a1b7451194730086c6 : SUCCESS in 4m 30s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c7c599dcc9a94636b8a43e93785b3bda : SUCCESS in 4m 24s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/dc09d18d1b9945c4a1f260700b79e003 : SUCCESS in 15m 48s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/897847142baf48a8bc7c2852482ffdd5 : SUCCESS in 12m 24s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b7dad6d778949679671abe6102a9af7 : SUCCESS in 8m 50s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/67cb594cb06a4e03971cb4f39598af63 : SUCCESS in 13m 15s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/cd67f592f3924cbf9ddc0b28714ede45 : SUCCESS in 42m 43s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/944ef595d7ef4c4383f9cfaff2adedb4 : SUCCESS in 48m 38s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7dd8754431684aa38896ba6e692a2075 : SUCCESS in 42m 31s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/b281db54f6784abaad94552f4a70e54d : SUCCESS in 46m 32s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6826d83ddc45410ea6e02d11bf19a447 : SUCCESS in 58m 25s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e626c3c4a2f546809384b600e38334fe : SUCCESS in 1h 08m 00s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e3c3b13432e34924aefc374dd44f3ef4 : SUCCESS in 1h 09m 16s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5049a02c676d488e87dc36c97c743708 : SUCCESS in 45m 00s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/6a0537f161bf4f28bbf63435783d0771 : SUCCESS in 17m 27s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/427dda3e32e94216b33a493c8eaf0965 : FAILURE in 21m 40s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a99b62c5f954b88b9824230254a652d : FAILURE in 7m 16s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b6c175b936a54cc2bae1aea787afb6d3 : FAILURE in 30m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"17bc0b911aa9273709c02d64774595f7a328e68b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 04:57:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"687a5d10c05ca5d0ea61caecfbcc9efd00cb3412","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 06:08:49.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/59c81f9bfe234b4080656fd7593056c0 : SUCCESS in 4m 20s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a0861a8cd457472a84b0e470117b27f2 : SUCCESS in 4m 26s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db66671e435c46f8970840208453b748 : SUCCESS in 4m 32s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/a593182bec054e14b00dd32eeb90697f : SUCCESS in 15m 38s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/221b6208d06d451c9f3d4e7a2fa6ab9b : SUCCESS in 15m 32s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bdc8c243a1034cea82bf59474e1d4dd4 : SUCCESS in 12m 45s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2d6810951b4d4c0b8f6faadfd8949cff : SUCCESS in 56m 25s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/be77df7b7bb443d6bd6de0d4179f6cb8 : FAILURE in 12m 05s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f5d709b75cc748b9b48dc3e4c40efb45 : SUCCESS in 1h 07m 10s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/13b093e2cbbf470a88fdfda17de59b36 : SUCCESS in 42m 13s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/766c41297edb4ad2824dc67749fe2efd : SUCCESS in 24m 01s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1e2b27a1049343f39d77583dfd1ce21b : FAILURE in 29m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f165561572d7cf5585508a4c271693ffca0026b0","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-09 06:39:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f362142186b177752d00d3e1abfd3131a9dae535","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 07:59:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/727f2811f4f7448398e93281359c31df : SUCCESS in 5m 57s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6913aba9a9a6430f9a5215f2a33b84be : SUCCESS in 5m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b75f697725b4000a12d82f096b03898 : SUCCESS in 8m 03s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/d9f37fc714b940a0b950b08d81648c4e : SUCCESS in 17m 45s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/4a04e420dff646e6aa55cf7905df34f6 : SUCCESS in 18m 27s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81ef9c6127334f72a9cd70db4b20f551 : SUCCESS in 16m 17s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/26dfed09339449828a2725aa23c57656 : SUCCESS in 14m 56s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/bc1ceedca8bc4b1ba27a5b80b2239919 : FAILURE in 52m 36s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c0b1fdab45a4dbd8e3c4398b51ded83 : SUCCESS in 54m 13s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2398cdad1f744252a72f78c30c2d3db9 : SUCCESS in 44m 54s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/8a28009575b149d9b1a0fa2abe42aa33 : SUCCESS in 41m 14s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ed200a414d6845a39417d801cf396c0a : FAILURE in 59m 20s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4d01a9f7d4a44c548b001af341217e55 : SUCCESS in 1h 00m 41s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ba12f93bcd0c4bd784db8da014bac1ab : SUCCESS in 1h 08m 46s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/942291d0b2914491a80efc58f96a5e24 : SUCCESS in 47m 25s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/cbdf2c8f0f77495d90755c93017abd7f : SUCCESS in 21m 20s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/3a5f0f48226f41b0a56b26f301fedde5 : FAILURE in 12m 49s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d10ad341f02a41f1b16408f64521f779 : FAILURE in 10m 15s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5f2fb451f5e74bf5abc4173f471ddc0d : FAILURE in 33m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"87d51c394e169980ce44c8383c37d3fa46bcede5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-09 09:26:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"52a9df33e2874c4384dbc54bc05c9099a5b48274","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 10:49:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa73f266c5064f2d9842b20ddd38b3b6 : SUCCESS in 4m 36s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/85b46167081e407f83710297d2153fe7 : SUCCESS in 4m 39s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/89ac465b4bfa4c57ad9768183b8775f0 : SUCCESS in 4m 28s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/7ed90179182c43df92aba718600acb9e : SUCCESS in 15m 09s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/375a8b377d2a467aaf36a2de73af3497 : SUCCESS in 15m 07s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/beb7698e1baa4db0bf1556b4c4c6f534 : SUCCESS in 9m 58s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f48ded6416d2484bac1418a90f97ee75 : SUCCESS in 11m 05s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/68dde8112f6f46f6b6e3ecefee9a83df : SUCCESS in 39m 18s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b88026fbde244ab69fd180250eee5136 : SUCCESS in 49m 14s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/669acee821dc4668b5df02bd34924db6 : SUCCESS in 42m 25s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/b671eb7b4e97481bbadc8edd0397c854 : SUCCESS in 34m 13s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/18c874a52cef4473a90b1dbc4c2ce613 : SUCCESS in 55m 54s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0b0feadd363947179af81c4d77b436f0 : SUCCESS in 57m 56s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/82865179e95d44a0b6f1dfd05e24ea4d : SUCCESS in 1h 21m 11s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/63f8cb3a920946a0a1ab0674db66dec3 : SUCCESS in 43m 30s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/0aa1cb769842449fa4f34f5d7ade0be5 : SUCCESS in 19m 06s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/c264d64c43dd43a899dd3a967c78a21f : FAILURE in 9m 32s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/077f3297eb9a40598225118545d92759 : FAILURE in 6m 45s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c03bbc752f824567bb76e1c972bc9826 : FAILURE in 31m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d5e428d8a1932d7a6c6ebff76747ac88255a5a37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 10:49:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2abdfff1a32d27fe5bc6df9b5c5d91e494845e16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 11:55:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/069c28c7a4764d1eae926d1781cb17bd : SUCCESS in 4m 07s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f2c5918ed3a546caa34ba2c20109fcf4 : SUCCESS in 4m 13s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/637c1dc216af450caba232bc4a396616 : SUCCESS in 4m 11s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/fda0f0a85aac4d9a88db50b68a335b24 : SUCCESS in 15m 26s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/84e136d5f720426da05797ed10a88628 : SUCCESS in 15m 41s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c8dc18365b4045a9a291f02344801752 : SUCCESS in 13m 47s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e5eb27bd48d846539be841d9c4b2e3e3 : SUCCESS in 57m 32s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b21bad485e9740c9b890cd1efb1069d9 : SUCCESS in 1h 01m 03s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/85394a58ffe746d4bf9ea9157b9fe7e2 : SUCCESS in 1h 00m 01s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/40c748161ec94223964697164dfa1900 : SUCCESS in 42m 10s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/33910ad11168414f815f9d6fd6bd4642 : SUCCESS in 17m 50s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1b1b5fa001ba4f1390bdff21d531455a : FAILURE in 32m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3805786f16f5ba3f2ffa5a673ed78736c7cea661","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 11:55:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ff89537c7af902731c48b83cf2279903c74c755e","revisions":{"ff89537c7af902731c48b83cf2279903c74c755e":{"kind":"REWORK","_number":1,"created":"2020-04-28 14:51:36.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/54/723954/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/54/723954/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/54/723954/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/54/723954/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/54/723954/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/54/723954/1"}}},"commit":{"parents":[{"commit":"e9f63508551b1e1bc07702be481e3a9ded9483b9","subject":"Prepare Ussuri RC1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e9f63508551b1e1bc07702be481e3a9ded9483b9"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-28 14:51:34.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-28 14:51:34.000000000","tz":0},"subject":"Update .gitreview for stable/ussuri","message":"Update .gitreview for stable/ussuri\n\nChange-Id: I6b12f2c0b4592e1faba93c2f47f77627db848831\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/ff89537c7af902731c48b83cf2279903c74c755e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/ff89537c7af902731c48b83cf2279903c74c755e"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
