)]}'
{"id":"openstack%2Fpuppet-glance~725127","triplet_id":"openstack%2Fpuppet-glance~master~I03016890692b406ca86750d7ec8b3b3e1d276b1a","project":"openstack/puppet-glance","branch":"master","topic":"autorequire","hashtags":[],"change_id":"I03016890692b406ca86750d7ec8b3b3e1d276b1a","subject":"Use anchor to require necessary packages/services","status":"MERGED","created":"2020-05-03 16:41:22.000000000","updated":"2020-05-04 02:05:32.000000000","submitted":"2020-05-04 02:05:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725127-1588557932619-56a7dbf5","meta_rev_id":"54db85e4d00f823d7b85b8dbb68d018b653f8f8c","_number":725127,"virtual_id_number":725127,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-04 02:05:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-05-04 00:29:35.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-05-03 20:01:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-05-04 00:29:35.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-03 20:01:56.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-05-04 00:29:35.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-05-04 02:05:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"938d1be1f8f8d264b31603e132f6816bd85d398e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-03 16:41:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30fca64355b6c40d3be820724d1c12005f99bb7e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-03 16:59:48.000000000","message":"Topic set to autorequire","accounts_in_message":[],"_revision_number":1},{"id":"2661379cd79f34e2fa73184d24231dc255686356","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-03 17:16:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"139a5d2f1cd9bc5e3196896db3f348ca5e2bf026","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 18:32:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8609eb3e0f9943b48ec5eee3e71e10da : SUCCESS in 5m 31s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35fb01a7800e4bfa8a2f1bad56c7f1cd : SUCCESS in 5m 25s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/650c09fa36d943dfa32cab4b2d3fc2f9 : SUCCESS in 4m 36s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/a0df570891bd4fbabd651cf7754745bc : SUCCESS in 11m 25s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/e933d0eaca0d4989ac0bc510d0b2ac04 : SUCCESS in 13m 23s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/391326dc73a245d986c842a407fb99de : SUCCESS in 11m 33s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0d91551aceaf43b3a424111dda5e5c82 : SUCCESS in 12m 32s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/4fbd5a4524ce4de7b72755a29bdb49d2 : SUCCESS in 54m 14s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eee395b89de641d093dabcccf1ff86ac : SUCCESS in 46m 04s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c4c49899d85442993c35106f8a873cb : SUCCESS in 41m 06s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/eb4335c37f93458584ab22e6ff40a231 : SUCCESS in 39m 14s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a7b0e9094ba94471b158147e0bcdfa72 : SUCCESS in 54m 02s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0e36b375430f4fe6ad1335385e5abbc4 : SUCCESS in 1h 08m 33s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c90787124ae34307972a036f747e674e : SUCCESS in 1h 01m 42s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b2ab80e91f12424a8d80c8b232dab284 : SUCCESS in 42m 14s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/4acdea7b38474b18bcf459f047496a97 : SUCCESS in 17m 19s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/34161da3d0044ff9978d6032d9db58b2 : FAILURE in 9m 06s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/710a96c6a7e64a6a8b70792d2aee2c2e : FAILURE in 6m 00s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4ee45e67234a408d9be7008b6997ec21 : SUCCESS in 1h 14m 56s","accounts_in_message":[],"_revision_number":2},{"id":"1aa9e93de6aff7be35c39080d38c467d76d1fd20","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-03 20:01:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3b51a373e8773d13a84992958ddb15b4c0d750dc","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-05-04 00:29:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"954730c7f067bd98d70a839ab9616eeee5c16137","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 00:29:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b0832d71112781174135db4dc2d299f304e8bf5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 02:05:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/94c7959d5d2344fda3e5af8734709e56 : SUCCESS in 4m 16s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a4f8a1ef98ed4c2d8f65f8a95a4287d1 : SUCCESS in 3m 53s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c2b7a473ea52484fb0e81c4b730fb0af : SUCCESS in 4m 17s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/56f0d91d6cfd474c9c1d138006efe8ec : SUCCESS in 16m 37s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/73532d9927814bfe98f4739f884f5f4f : SUCCESS in 12m 45s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/24bb6675ee4046ad864562236cf5b6a6 : SUCCESS in 10m 57s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7ee6aecec77f4245bdcb7393fb45e71c : SUCCESS in 56m 56s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a59c6df7dd7e4d119b2a09106be63ad5 : SUCCESS in 1h 01m 21s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d803a2d880444864b9b7b80a10ecb07f : SUCCESS in 1h 02m 49s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/bfe0a4becc33456e920da06579d2a8ac : SUCCESS in 43m 21s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9e22d7bed22641cdab9d7c94bfa9e4ef : SUCCESS in 17m 55s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4836e9fcdd8148faa4c683baa641bf1b : SUCCESS in 1h 28m 55s","accounts_in_message":[],"_revision_number":2},{"id":"54db85e4d00f823d7b85b8dbb68d018b653f8f8c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 02:05:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"10a9cbe4d757519b76b0af8a320f24b1f4b29784","revisions":{"3f6671ada36e65749c08a3bade049d5f46f35fbd":{"kind":"REWORK","_number":1,"created":"2020-05-03 16:41:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/725127/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/27/725127/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/27/725127/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/27/725127/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/27/725127/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/27/725127/1"}}},"commit":{"parents":[{"commit":"56fd00fe5afb338d49cdbbea1f5332a41483b77b","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/56fd00fe5afb338d49cdbbea1f5332a41483b77b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-03 16:40:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-03 16:40:47.000000000","tz":540},"subject":"Use anchor to require necessary packages/services","message":"Use anchor to require necessary packages/services\n\n... so that correct packages and services are required according\nwithout re-defining them in resource implementations.\n\nChange-Id: I03016890692b406ca86750d7ec8b3b3e1d276b1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/3f6671ada36e65749c08a3bade049d5f46f35fbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/3f6671ada36e65749c08a3bade049d5f46f35fbd"}]},"branch":"refs/heads/master"},"10a9cbe4d757519b76b0af8a320f24b1f4b29784":{"kind":"REWORK","_number":2,"created":"2020-05-03 17:16:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/725127/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/27/725127/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/27/725127/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/27/725127/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/27/725127/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/27/725127/2"}}},"commit":{"parents":[{"commit":"56fd00fe5afb338d49cdbbea1f5332a41483b77b","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/56fd00fe5afb338d49cdbbea1f5332a41483b77b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-03 16:40:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-03 17:15:52.000000000","tz":540},"subject":"Use anchor to require necessary packages/services","message":"Use anchor to require necessary packages/services\n\n... so that correct packages and services are required according\nwithout re-defining them in resource implementations.\n\nChange-Id: I03016890692b406ca86750d7ec8b3b3e1d276b1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/10a9cbe4d757519b76b0af8a320f24b1f4b29784"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/10a9cbe4d757519b76b0af8a320f24b1f4b29784"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
