)]}'
{"id":"openstack%2Fpuppet-glance~729202","triplet_id":"openstack%2Fpuppet-glance~master~I2908540605a824acdc48b478f76da17c5dd08a39","project":"openstack/puppet-glance","branch":"master","topic":"db-password","hashtags":[],"change_id":"I2908540605a824acdc48b478f76da17c5dd08a39","subject":"Remove password hash generation in each puppet modules","status":"MERGED","created":"2020-05-19 10:50:38.000000000","updated":"2020-05-20 05:31:04.000000000","submitted":"2020-05-20 05:31:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729202-1589952664964-278ba6fa","meta_rev_id":"8650a3a2bbc9438f683832e4e40fc283dd7bd302","_number":729202,"virtual_id_number":729202,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-05-20 02:33:58.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-20 05:31:04.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-05-19 12:14:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2020-05-19 21:45:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2020-05-19 21:45:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-19 12:14:20.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-05-19 21:45:51.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-05-20 05:31:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"46324e7b98c76df83f718001b99fa9146d109ca4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-19 10:50:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"037ddeaea326de19d2d6366e0a0c8653f9d3c648","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-19 10:51:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ffe2774c42d555f41fdb1414732e47364e7db5e4","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-19 12:14:20.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3969408264f56b29f5a6439d7bbaca12cc4f9c31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 12:30:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/4f5d385f95b24dcfa9340453188f3df1 : SUCCESS in 2m 36s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4fb59e66eae648fdad111f0b1937e88f : SUCCESS in 4m 02s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1690f42d211c48bb8e87f643d334d245 : SUCCESS in 6m 28s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a3c0ebfc83fa480db92d2c9c70f17bdc : SUCCESS in 5m 30s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/cb48906cbcec46f19a38c95fff0eea8b : SUCCESS in 16m 47s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/3b9c1416207041c09041cac9aa522c55 : SUCCESS in 15m 10s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e6a4977f0c047c1a2455219b509f6ff : SUCCESS in 14m 38s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f9ca03945e304cafb6ec660c2417fbc1 : SUCCESS in 14m 36s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/2b8ce287d3284fa191adde7a16fa3fd0 : FAILURE in 38m 32s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1424028507f34b80a88df3edfc968797 : FAILURE in 40m 36s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/174361010abd408eb7d5a45619e3577d : FAILURE in 32m 05s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/6f013ba55ee04282a6c9c34e9b97c3e6 : FAILURE in 29m 16s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ecdaa889182243838d66e4c72ec10491 : SUCCESS in 52m 55s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7e7bf5b093fa4d5e8d5be24cdc2417ec : SUCCESS in 1h 07m 04s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4002c063f3304f25b71acf6121cc87aa : SUCCESS in 1h 12m 27s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/14f72498ad1e434896a93e69f7874f0c : SUCCESS in 49m 01s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9f9142374e9f48e8b5baefb44369ff70 : SUCCESS in 20m 46s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/64f1a25fd6e141a691537bda42f8e4fc : FAILURE in 12m 13s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/739ccfbbdafd40edb72b80e73fce7b5b : FAILURE in 8m 26s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4ce8d95702c94ebbace09a434a1d983f : SUCCESS in 1h 34m 07s","accounts_in_message":[],"_revision_number":2},{"id":"c0e3c35787cda6224c824de5c3e4b58e951b5231","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-05-19 21:45:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b84265ea95356fd9e6844aba4d74b159586c15d5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-20 02:33:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"be084e0eadb15bbce4429abca0b4ee3f7de844b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 03:58:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/b3b9ba0fbc6b46efb71514294c7629c5 : SUCCESS in 3m 43s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc70dd0b09cb48a88955a67ddb166de3 : SUCCESS in 7m 04s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/27585b8055504dd1b15c879b569ae3d9 : SUCCESS in 6m 13s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1248062121c347498750388d2d67a2ce : SUCCESS in 6m 29s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/391593e328254bed8dde3b5de72ba2cf : SUCCESS in 15m 24s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/b7b035adbd3246b6b624a9f5590e94d4 : SUCCESS in 16m 24s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7dd2c89aacb14748a0039026b40a615c : SUCCESS in 14m 26s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c14235510cfa4b2981a7ab67874c11cf : SUCCESS in 13m 37s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/7045077785e942f6bdecb578969db7a5 : FAILURE in 35m 48s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6906fcc60eb74c40a2925b66dc8a23b3 : FAILURE in 37m 05s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d24f848df45c46588bd4d10b29b13b7d : FAILURE in 35m 49s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/f0ca8bf22fa84c71a0d99c78da17b05e : FAILURE in 37m 09s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c0b58f1d44594c6597396cb63a4bcb31 : SUCCESS in 54m 20s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/210c3199272448f286223807116fe1dd : SUCCESS in 52m 32s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2151bfece89a451da73748c2f6690178 : SUCCESS in 1h 09m 46s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ed3bcb1c4b144d578cc28d1260663227 : SUCCESS in 44m 14s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/27919af8071048f293ad7ad676049d7d : SUCCESS in 17m 49s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/78b3e7902379410481d1d68acc10168a : FAILURE in 12m 06s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/034b6bfeb9734d829460c213420a467d : FAILURE in 8m 34s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0d6f7b270e22443ca835c8c37f55225a : SUCCESS in 1h 22m 56s","accounts_in_message":[],"_revision_number":2},{"id":"d2c9e5dca6704a9433d6959f7f7915f590048ced","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 03:58:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2f9aff981ac9c8e22f90a56d1baad938b531a7f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 05:31:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"fa364e55ac998d6107c982951aa81269f78d03a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 05:31:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/c7890c0061034acd9b51d083cc81a55b : SUCCESS in 4m 34s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce6693c7873144ef83d5c3eac30bdda0 : SUCCESS in 8m 24s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ea28f340308443c1bc9b70a1c6ed71b3 : SUCCESS in 5m 36s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/61f256df6cf34a1b9ff8faf37e24cb88 : SUCCESS in 4m 27s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/785a4cbfb05a42c8a3094f974ee8717c : SUCCESS in 14m 17s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/3093baf977c4426ab16c5053db9cda58 : SUCCESS in 13m 58s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b775b6e4b4694e70a8b9465eff89e56b : SUCCESS in 12m 08s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/388ab1f9f1c84a7f990486b98cdff34b : SUCCESS in 54m 42s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5f46d1a537444609b5fd27961a77212e : SUCCESS in 1h 08m 18s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1a65ecaf70324265a3062c7e296ff572 : SUCCESS in 1h 07m 23s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7f25b73072dd48b09411e609ea84594c : SUCCESS in 44m 09s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/4d67c08513e34e509d754e758faf16ac : SUCCESS in 18m 50s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/aacca7e8f25942b393ab84a766e63d0c : SUCCESS in 1h 31m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1015639de821e544be80f4543921be542a7528e8","revisions":{"afc8ce28a5c28cd6af9568c1085a892278e753da":{"kind":"REWORK","_number":1,"created":"2020-05-19 10:50:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/729202/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/02/729202/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/02/729202/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/02/729202/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/02/729202/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/02/729202/1"}}},"commit":{"parents":[{"commit":"0ca3551c1c28ecd7d862f42c3bfa63513911bf9c","subject":"Merge \"[ussuri][goal] Add CONTRIBUTING.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/0ca3551c1c28ecd7d862f42c3bfa63513911bf9c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-19 10:50:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-19 10:50:28.000000000","tz":540},"subject":"Remove password hash generation in each puppet modules","message":"Remove password hash generation in each puppet modules\n\n... and migrate it to openstacklib so that all logics about database\nconfiguration are implemented in one common place.\n\nDepends-on: https://review.opendev.org/#/c/728595/\nChange-Id: I2908540605a824acdc48b478f76da17c5dd08a39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/afc8ce28a5c28cd6af9568c1085a892278e753da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/afc8ce28a5c28cd6af9568c1085a892278e753da"}]},"branch":"refs/heads/master"},"1015639de821e544be80f4543921be542a7528e8":{"kind":"REWORK","_number":2,"created":"2020-05-19 10:51:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/729202/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/02/729202/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/02/729202/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/02/729202/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/02/729202/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/02/729202/2"}}},"commit":{"parents":[{"commit":"0ca3551c1c28ecd7d862f42c3bfa63513911bf9c","subject":"Merge \"[ussuri][goal] Add CONTRIBUTING.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/0ca3551c1c28ecd7d862f42c3bfa63513911bf9c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-19 10:50:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-19 10:51:09.000000000","tz":540},"subject":"Remove password hash generation in each puppet modules","message":"Remove password hash generation in each puppet modules\n\n... and migrate it to openstacklib so that all logics about database\nconfiguration are implemented in one common place.\n\nDepends-on: https://review.opendev.org/#/c/728595/\nChange-Id: I2908540605a824acdc48b478f76da17c5dd08a39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/1015639de821e544be80f4543921be542a7528e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/1015639de821e544be80f4543921be542a7528e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
