)]}'
{"id":"openstack%2Fpuppet-glance~729836","triplet_id":"openstack%2Fpuppet-glance~master~I8ece28a9b0c6f104130d817490e9ea4c2b4e7808","project":"openstack/puppet-glance","branch":"master","topic":"cache-prefetcher","hashtags":[],"change_id":"I8ece28a9b0c6f104130d817490e9ea4c2b4e7808","subject":"Add cache prefetcher interval configuration option","status":"MERGED","created":"2020-05-21 05:14:49.000000000","updated":"2020-05-26 06:34:50.000000000","submitted":"2020-05-22 20:52:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729836-1590180761797-f4975ef0","meta_rev_id":"31bd3ef8b2945c6256f06d84143f6a7562bee0e3","_number":729836,"virtual_id_number":729836,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-05-22 05:52:38.000000000","_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-22 20:52:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-05-22 19:17:01.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-05-22 14:33:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2020-05-22 15:07:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-05-22 19:17:01.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-21 12:43:16.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-05-22 05:35:49.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-05-22 14:33:46.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-05-22 15:07:58.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2020-05-22 19:17:01.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-05-22 20:52:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"445ac2911768d4e5366a5aa9bb6c74c995e6a02f","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2020-05-21 05:14:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3afeab6f95ad1e98aa4dca6838576725c03ff25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 06:48:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/89b92514b8614f2dafc2381a64d9ffff : SUCCESS in 4m 42s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/525e3add65664408bded50727ff3edd2 : SUCCESS in 4m 07s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/54948f73e7c146ae82391399be542d7c : SUCCESS in 4m 05s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/069184afd3bf4ae8a664b2bedb6acafa : SUCCESS in 16m 05s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/d181347e28ef4989ab2bcabe24d48737 : SUCCESS in 14m 30s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9e00ef8a6f6d41b581dcc18ae78dd575 : SUCCESS in 12m 16s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/03fcdeba60134734a8e8e5f11bccd684 : SUCCESS in 10m 49s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/259021dfe7d04b289231ff7ca9dd93c9 : FAILURE in 33m 41s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f4be0b9bffe94cff8f7db6eb93fbe1d3 : FAILURE in 33m 02s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e48194cf3b464a14a5bee2f953dd1dd8 : FAILURE in 56m 38s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/035f5e65b8a04cf09d9709c4bf53000e : FAILURE in 20m 26s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/46c93396bae8431a8e670c986774e469 : SUCCESS in 57m 35s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/af140dd49b89405abdb7301e55082755 : SUCCESS in 1h 01m 19s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7de85d01de984097a53880d92f8b615e : SUCCESS in 1h 04m 56s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/653291965d2249dbade3de94df74baec : SUCCESS in 44m 57s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1864a6ca1e234527abae0cb8166410fa : SUCCESS in 19m 36s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/71e4f57ca0d44b50a95cffc4f6f8c3d0 : FAILURE in 8m 17s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a87703e67aec4ce49adfc98f67d2f269 : FAILURE in 6m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d47a8c391e3c4928861b66d76a12c4fa : SUCCESS in 3m 11s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/13aa912409c443bc8e28df113ad926ef : SUCCESS in 1h 29m 38s","accounts_in_message":[],"_revision_number":1},{"id":"fdb71b86255ed56d1df87c1b00bd5f1866ac9684","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-21 12:43:16.000000000","message":"Patch Set 1: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e1b1509ed8f27aa4ae7b56fa6dbb878e10d7896f","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2020-05-21 17:47:57.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2b82c2233a42760de96b969447357bb034763991","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2020-05-21 17:50:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ba6828c7e1bdc05936843f0f53e0a7d6b22e7b43","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-05-22 05:35:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1224cd097b81cecad2e7c0a74391ae425abfb976","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2020-05-22 05:44:24.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"af1fdbd35f0509baed68a26bcbea0d937d0db9b5","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2020-05-22 05:52:38.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7c33cb96df16743c5cfc18d1d27c0c93d6c831e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 07:16:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/046c4a47cf514604a26eea1f3be60e6d : SUCCESS in 3m 57s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5eb6811c12a84d8e97585416691958da : SUCCESS in 3m 59s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b28c9556704c45c7957981c17c20a9ae : SUCCESS in 4m 15s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/4efb3a37b3144ccf8496063720eb69ab : SUCCESS in 11m 49s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/5a2ed50c385d41fb8e8b417dd8ad6bd1 : SUCCESS in 13m 08s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a5b0905cd01e4fcc954718be61a5b71f : SUCCESS in 10m 06s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ea5e8fcb8f64f8f8c8a0a08bef5b8f3 : SUCCESS in 11m 25s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/f781e7e874064841bc55c376eb730bd4 : FAILURE in 34m 37s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c3492d9b36db4120b6551e374a9df584 : FAILURE in 33m 30s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88f4a7366d1e4836ba3848af7bfe5c9c : FAILURE in 53m 42s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/6a9cae4735914f92854bf616d17b2be7 : FAILURE in 29m 39s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/742353ca563041c7abd4d2591874ae23 : SUCCESS in 1h 01m 35s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/71727e67497541dd9c897f2de3b1e2f2 : SUCCESS in 1h 10m 06s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4b033e7ea0f546919a970a0556153fd1 : SUCCESS in 1h 11m 51s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/74778d52367e49b39cdc314cd3de2610 : SUCCESS in 40m 15s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/6a85cf1249e141c59e76a8c013462393 : SUCCESS in 18m 09s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/748c256fc7b54d36a0bbe6fe8834cdf5 : FAILURE in 8m 09s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf7a9f2424514f9e974c81009affe8f1 : FAILURE in 10m 36s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b085f2d0a44645aab9a9f1dacac2fedf : SUCCESS in 3m 17s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/24f235091f494046b483255567aae939 : SUCCESS in 1h 27m 54s","accounts_in_message":[],"_revision_number":2},{"id":"d4856dae8299b4c5300f369db13077130001e1db","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-05-22 14:33:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4327f24a2fc3dea4080ae2a589ce8e859e8360ac","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2020-05-22 15:07:58.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c16f5e35d41f0d7c475dd44e866caf439cf95265","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-05-22 19:17:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d58eea0c3b16fe80695c100e3e409100b2812fd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 19:17:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"52a433b0dca606ddd899df02e98eb72e4242ef21","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 20:52:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f3f1c6af185b0b4e3b2c21e85891df111065838b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 20:52:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d900cf0e0b974951a0e943bd2e908232 : SUCCESS in 5m 36s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/661d0394d0b94e80b91e97dd86ab0222 : SUCCESS in 5m 20s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/03c03972f1f4481f8d6140a2932c4857 : SUCCESS in 4m 54s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/4d3f8170121d4e9daabe83a5f3b6089c : SUCCESS in 14m 03s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/00ea64e8fbff4def9333b74b4a2a8c87 : SUCCESS in 14m 35s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5efcc17d28064b2a8b24e662d6d7ce92 : SUCCESS in 14m 34s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8a7107c2b59d40378c5386f51debaa9b : SUCCESS in 1h 04m 50s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f62798c87ff14ee6b5380f4e60c6973b : SUCCESS in 1h 17m 24s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/15ecd8ea91324551b61608c184f3303b : SUCCESS in 1h 08m 53s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/461aeff7cff54d80954e538b3b866227 : SUCCESS in 49m 10s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/5bacdb87d33d4f75b9f44e6fdc7b053f : SUCCESS in 23m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69ac335bb9a7492fb103ef3937bb77f9 : SUCCESS in 4m 56s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/54369a3f5a2c40f39cf9c053eb3cfd69 : SUCCESS in 1h 34m 35s","accounts_in_message":[],"_revision_number":2},{"id":"005945b13ecf6c355d197d74ba090d84b3c0fdeb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 20:54:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4aa7cd71101248d49767f664cbd1a0a7 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2},{"id":"7e186f3f893a37f711dce3b9054c2d1a6bf2e95c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-24 01:20:37.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nI know this patch was already merged, but we need some follow up about configuration parameters about cache config.","accounts_in_message":[],"_revision_number":1},{"id":"77686ac045a0917d83df2e3142e6ffcd5119c582","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-24 01:27:41.000000000","message":"Patch Set 2:\n\nSorry but I missed that image cache parameters were already included in glance-api conf. please ignore my previous comment. sorry again for confusing you.","accounts_in_message":[],"_revision_number":2},{"id":"c24fed527cea8a02ed972dd0e53f96449a2904bd","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2020-05-26 06:34:50.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 311cb2ade54812d3bec0c947d1453aad47c22c3a","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"24ee288b85d376c96746f1f0864bfb0344cf5b19","revisions":{"2036708a30128569c721e5d9468feac62a2f1cf9":{"kind":"REWORK","_number":1,"created":"2020-05-21 05:14:49.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/36/729836/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/36/729836/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/36/729836/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/36/729836/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/36/729836/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/36/729836/1"}}},"commit":{"parents":[{"commit":"8e4c57b6a30d0b8c40c117835c1cf62a7ecdf2a7","subject":"Merge \"Remove password hash generation in each puppet modules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/8e4c57b6a30d0b8c40c117835c1cf62a7ecdf2a7"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2020-05-21 05:02:40.000000000","tz":330},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2020-05-21 05:14:34.000000000","tz":330},"subject":"Add cache prefetcher interval configuration option","message":"Add cache prefetcher interval configuration option\n\nAs new periodic job added in glance which will run as per\ninterval set using \u0027cache_prefetcher_interval\u0027 configuration\noption and fetch images which are queued for caching in cache\ndirectory.\n\nEnabled \u0027cache_prefetcher_interval\u0027 configuration option\nin glance-api.conf.\n\nChange-Id: I8ece28a9b0c6f104130d817490e9ea4c2b4e7808\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/2036708a30128569c721e5d9468feac62a2f1cf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/2036708a30128569c721e5d9468feac62a2f1cf9"}]},"branch":"refs/heads/master"},"24ee288b85d376c96746f1f0864bfb0344cf5b19":{"kind":"REWORK","_number":2,"created":"2020-05-22 05:44:24.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/36/729836/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/36/729836/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/36/729836/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/36/729836/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/36/729836/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/36/729836/2"}}},"commit":{"parents":[{"commit":"8e4c57b6a30d0b8c40c117835c1cf62a7ecdf2a7","subject":"Merge \"Remove password hash generation in each puppet modules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/8e4c57b6a30d0b8c40c117835c1cf62a7ecdf2a7"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2020-05-21 05:02:40.000000000","tz":330},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2020-05-22 05:41:57.000000000","tz":0},"subject":"Add cache prefetcher interval configuration option","message":"Add cache prefetcher interval configuration option\n\nAs new periodic job added in glance which will run as per\ninterval set using \u0027cache_prefetcher_interval\u0027 configuration\noption and fetch images which are queued for caching in cache\ndirectory.\n\nEnabled \u0027cache_prefetcher_interval\u0027 configuration option\nin glance-api.conf.\n\nChange-Id: I8ece28a9b0c6f104130d817490e9ea4c2b4e7808\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/24ee288b85d376c96746f1f0864bfb0344cf5b19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/24ee288b85d376c96746f1f0864bfb0344cf5b19"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
