)]}'
{"id":"openstack%2Fpuppet-glance~730425","triplet_id":"openstack%2Fpuppet-glance~stable%2Fussuri~I18012986b816b52b4ac8b81354b3f30b0bc8c48c","project":"openstack/puppet-glance","branch":"stable/ussuri","topic":"cache-os_region_name","hashtags":[],"change_id":"I18012986b816b52b4ac8b81354b3f30b0bc8c48c","subject":"Remove remaining os_region_name","status":"MERGED","created":"2020-05-23 00:35:05.000000000","updated":"2020-05-25 18:42:11.000000000","submitted":"2020-05-25 18:42:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730425-1590432131527-7d9b3224","meta_rev_id":"36d70548f0de1b9e909a6b242454e9bd14f3b141","_number":730425,"virtual_id_number":730425,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-25 18:42:11.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-05-25 17:04:48.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-05-25 06:54:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-05-25 17:04:48.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-25 06:54:07.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-05-25 17:04:48.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-05-25 18:42:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1977fd0066584daf52b4b567977e3a6d19aab840","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-23 00:35:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"814f8db2ed1b291e9818022dfb8e3413678fa306","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-23 00:35:24.000000000","message":"Topic set to cache-os_region_name","accounts_in_message":[],"_revision_number":1},{"id":"82d1b77e72ca033fc81a901cddce0156edea1556","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-23 01:57:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/829496affd664d16aa7fec3b69ef5456 : SUCCESS in 4m 33s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/121370e65ba340798ec3c2098a120d2c : SUCCESS in 3m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bdacd2c620c84d4cb2090b55436346b5 : SUCCESS in 4m 10s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/30ed709b746140e8ba82d0c01b19f13b : SUCCESS in 11m 05s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/67d8e5d2f48c4814ae76b63538af73c2 : SUCCESS in 13m 28s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fbe54d91654e4731ab7871c5da800f85 : SUCCESS in 8m 47s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e452a7804c724d8d8dcc78f67270edb6 : SUCCESS in 12m 16s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/6a0d7cc803e6448c8d2de212515fa977 : FAILURE in 36m 31s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ffff96b792540ccaff104f0d9964bb4 : FAILURE in 36m 30s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b72d47a28bad48c9b20bd61196fbff77 : FAILURE in 33m 28s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/f32028833cdc4bef9e01d6a5e9471ef1 : FAILURE in 25m 19s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/73bba849593d42acaf8297dd75f2ceb8 : SUCCESS in 52m 21s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/27b43d0e1a7f462ab33bfdc16596ec75 : SUCCESS in 57m 19s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d8170939355a4c83923ee8ffff214d71 : SUCCESS in 1h 12m 49s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/743cc7c8b8be4e5ea3a24ef43a237bb9 : SUCCESS in 38m 28s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ddfd86ed11de48b8b842b669d4d11ebc : SUCCESS in 19m 57s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/c843e54b141b40cea05aa23d37f4f519 : FAILURE in 9m 05s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/593fbe82f42443ebbaa5451b82c68f38 : FAILURE in 7m 01s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/bda0fa4baf804eed8b3d3c7f76fce831 : SUCCESS in 1h 21m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f472d9a168c6f04b7558d1970c496f9fc46ae929","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-25 06:54:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"946858f9f6b46d34df283991a069c9a2ba7ab471","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-05-25 17:04:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0ef83d405ba9a8fad7bcd04c09fda23943a167a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 17:05:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0649ab2d9251174e888f79f7fdac768de6082051","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 18:42:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"36d70548f0de1b9e909a6b242454e9bd14f3b141","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 18:42:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e09ce12568f4d6496fd7e8d1427b2fe : SUCCESS in 5m 12s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a7a9d789cb464d0698d5e18752d447cc : SUCCESS in 4m 58s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/406c5ec23c5442bc8c22c080d5be9bd3 : SUCCESS in 4m 29s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/b45c369fccbf4b87b51e8d5042614dd2 : SUCCESS in 14m 44s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/6cc9c011606446799e83a46f0d59365c : SUCCESS in 17m 37s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/586fa8537f734cc2a66bc9e5a216f4ac : SUCCESS in 13m 42s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2673e2d85aae450ea4c0af5cf9db7950 : SUCCESS in 1h 00m 23s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b896009ed2764c5c83d6558adfd0d27e : SUCCESS in 1h 10m 07s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/76e3cec7de50469781cb95aeee4eb34f : SUCCESS in 1h 05m 14s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d2cfce97ba4a4464a7c231272206aecd : SUCCESS in 53m 21s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/38feda6bb542422791fdb4abd94ac2c4 : SUCCESS in 19m 53s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a5fe0d812d144d109a760ecf4c77764b : SUCCESS in 1h 32m 17s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2cc8a4b3b27d32d2338f892c56cca7de0c0c3b1b","revisions":{"2cc8a4b3b27d32d2338f892c56cca7de0c0c3b1b":{"kind":"REWORK","_number":1,"created":"2020-05-23 00:35:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/25/730425/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/25/730425/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/25/730425/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/25/730425/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/25/730425/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/25/730425/1"}}},"commit":{"parents":[{"commit":"a265ac1ee513c08a90bbc2f229ab35b1e2e0d1f7","subject":"Switch to Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/a265ac1ee513c08a90bbc2f229ab35b1e2e0d1f7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-21 12:46:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-23 00:34:57.000000000","tz":540},"subject":"Remove remaining os_region_name","message":"Remove remaining os_region_name\n\nThis patch removes glance_store/os_region_name from glance-cache.conf,\nbecause the parameter should have been removed when we replaced it by\ncinder_os_region_name[1]\n\n[1] 9e82f598ad86d99a5bc356ba5c118a52a1d5d9bc\n\nChange-Id: I18012986b816b52b4ac8b81354b3f30b0bc8c48c\n(cherry picked from commit ccadb53ef17489aec9f3a017026c8f94427c9ceb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/2cc8a4b3b27d32d2338f892c56cca7de0c0c3b1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/2cc8a4b3b27d32d2338f892c56cca7de0c0c3b1b"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
