)]}'
{"id":"openstack%2Fpuppet-glance~731727","triplet_id":"openstack%2Fpuppet-glance~stable%2Ftrain~I8ece28a9b0c6f104130d817490e9ea4c2b4e7808","project":"openstack/puppet-glance","branch":"stable/train","hashtags":[],"change_id":"I8ece28a9b0c6f104130d817490e9ea4c2b4e7808","subject":"Add cache prefetcher interval configuration option","status":"MERGED","created":"2020-05-29 11:01:42.000000000","updated":"2020-06-11 15:39:09.000000000","submitted":"2020-06-11 15:37:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731727-1591889823438-b1f3fad0","meta_rev_id":"d1ff589ff4ca5fde2dff2b74a86664dc8f94a9ef","_number":731727,"virtual_id_number":731727,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-06-09 11:39:11.000000000","_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-11 15:37:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-06-10 14:28:08.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":1,"date":"2020-06-11 13:21:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-10 14:28:08.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-06-11 13:21:10.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-06-11 15:37:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"11b8200be226aba2d728ec2b3b165efc954c038d","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2020-05-29 11:01:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66e2392508c3fe24d7aed995ef5f9a54b59a3aac","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2020-05-29 11:07:30.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"6335073daeee3f256e73fe941f7b5a88677e049a","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2020-05-29 11:24:06.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2},{"id":"ba3f524e7bd3bd0bed88d593d567460f75e40826","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2020-06-09 09:49:08.000000000","message":"Restored\n\nRestoring back as dependent THT patch in stable/ussuri is merged.","accounts_in_message":[],"_revision_number":2},{"id":"f47dbdbaf3e558428d04830f296d27db07060523","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 11:12:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0dd054bbab4e4e3889856cd3cb24558d : SUCCESS in 5m 10s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/def26490273748ecb12d6763793d2279 : SUCCESS in 5m 18s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c97731d4fcc8481e878569c74e1c614b : SUCCESS in 5m 41s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/bdfc26097f7b48ffbe3e41bc2cf2b202 : SUCCESS in 16m 12s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/9d63c71678934bb4a4a9b3bad023eaf0 : SUCCESS in 13m 03s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/0d433e77977c4087a67746ce620e9205 : SUCCESS in 13m 18s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ebf0f170ac0e46769b990c8cf34760a3 : SUCCESS in 13m 50s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e73c4b5771814639b885ea58d5535745 : SUCCESS in 15m 51s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/1be6b0d54b5044bbbae6781210588ccd : SUCCESS in 54m 13s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/44a8145923174cec918e45453ab113f9 : SUCCESS in 1h 15m 38s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/29673995d5e343e6ae598aab828831f3 : SUCCESS in 1h 02m 41s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/be8fe6682969473f88f6a393de043468 : SUCCESS in 49m 17s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/af1f87cd05f842f3bae21c5f2a0033ee : SUCCESS in 31m 46s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f5519d44201847b48da119057794b1d9 : SUCCESS in 50m 49s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1843562505e740e38bd366764e781553 : SUCCESS in 43m 09s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/5cb8bddf867b413c9b4f9d3e76d314d2 : SUCCESS in 55m 10s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/81a92a08bf2240b3b45a75adb5a301ab : SUCCESS in 48m 48s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b202b54a4d254ca68920f6d68fe91d19 : SUCCESS in 1h 03m 23s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/59187123af0b4c459ebe631ec63d88b9 : SUCCESS in 55m 50s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9b96cfa5df0c48a7850a2b0c94ad655e : SUCCESS in 45m 22s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ddcc66fbf9b34b64ba9fcacdc7ae46ea : SUCCESS in 22m 00s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e78a8adc73e24af2aba9657ca7cc100e : FAILURE in 18m 53s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/51e2c789de8b4051a7dde209ec33b9f8 : FAILURE in 57m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bacd992d6ed84cc2a909e43330e78488 : SUCCESS in 7m 39s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ee2b31b6ee5946b6810fd6140627b2a5 : SUCCESS in 1h 19m 10s","accounts_in_message":[],"_revision_number":2},{"id":"d6cdbc374246b2193e9bd7baf1098d63f7f594cc","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2020-06-09 11:39:11.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e9321044115bcf1bc9683587e9ede1b045252dfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 13:45:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/50f0623c244644f69ce704e5e7742ddf : SUCCESS in 4m 57s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f0c4cb604454b0895ea4519950ed062 : SUCCESS in 4m 11s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8df87c351d9a468da51fd09d5753fc04 : SUCCESS in 5m 24s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/111240b1e1f9449595230145ba1b4e76 : SUCCESS in 20m 16s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/d15329383680452781a29395ef35ccc8 : SUCCESS in 12m 45s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/45b2ba5476a04e6fac419f23baf66e7c : SUCCESS in 15m 56s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/904b5475b9eb4194989c752f30e1a547 : SUCCESS in 14m 32s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/05f5806f07ab42048980be353285a704 : SUCCESS in 14m 01s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/7fd9fae9bb6342eab43c6bdcc29a1eda : SUCCESS in 53m 13s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/25a02424652c413abdb40ea9f5277157 : SUCCESS in 1h 09m 10s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/b39e13c0edbc4cda90cf92e051140362 : SUCCESS in 1h 14m 09s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/a939575e761b4fa09ed4ca7cbb245a4a : SUCCESS in 54m 54s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/5004008a70ed4ed58daafd99490c759d : SUCCESS in 49m 13s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce6c219930ed4e438558970015a87473 : SUCCESS in 50m 11s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ddbdc747b38341348fc92ffb5e239177 : SUCCESS in 35m 53s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/fa566aee44c64fd5a041c37b44c35d1a : SUCCESS in 39m 04s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4815eccf98fd4cd583291c999692f41a : SUCCESS in 1h 01m 25s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4e041136de1f4c5dbd86a7c92a3d58b6 : SUCCESS in 1h 06m 02s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2ceac4b543974a73af7c802a76382043 : SUCCESS in 56m 20s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fb06b66c26674b39935f18db26bb11e3 : SUCCESS in 40m 07s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1b8dbed16c5746bd86ef3cac4b47926e : SUCCESS in 19m 04s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa801f2e51f447ac98b970104a7d03ce : FAILURE in 19m 10s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/6e7e8584e2674302b8c754f674ad60d4 : SUCCESS in 1h 59m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7619e7a47f0842f7aa4a1932aa0ea2da : SUCCESS in 4m 14s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a58d88fc1f00413a988baac2e02e3dc4 : SUCCESS in 1h 47m 58s","accounts_in_message":[],"_revision_number":2},{"id":"4c444132646501e05a1706b04ba1431622052de8","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-06-10 14:28:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0e0b8d964f7d9c230ac949f5977e460f4fb78895","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-06-11 13:21:10.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d4ca3f6589e48041623784fc6273273e76c00c40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 13:21:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"904622c0f828cebf8593ae0773af5b4c68cc502e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 15:37:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e1db3e86d61e49c1be6005806861856f : SUCCESS in 4m 35s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c6582e25de0471a8776b48f95cbe5ce : SUCCESS in 4m 25s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4b762e0a1a2d4bf5bd8fc061fcbf9f3b : SUCCESS in 4m 57s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/1cfe254b54b74cd89364463f9e084c84 : SUCCESS in 20m 40s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/635e8b304db54e738336d0d0b4a5ca88 : SUCCESS in 11m 13s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/c269ee7cacf643a797117772f39ae028 : SUCCESS in 57m 35s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/29c3c71959844f079d8ca917f8876b5a : SUCCESS in 1h 03m 28s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/da66918dcad244008d65b66b884fbc9e : SUCCESS in 1h 03m 20s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/68c987bb080a44f6a94a3b607cab6f41 : SUCCESS in 53m 42s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e9b05e13fa5347e28b0475631ae199ff : SUCCESS in 19m 49s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/0efd49f7612d45fa8ff6a6a3cb29b358 : SUCCESS in 2h 11m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/037fb042602049868bb2e65745422a46 : SUCCESS in 3m 46s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6472a62a1ea1448da66f5d178206f9d9 : SUCCESS in 2h 11m 07s","accounts_in_message":[],"_revision_number":2},{"id":"3850afb995d91339ffcb53382b5fa995b1e3b96f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 15:37:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d1ff589ff4ca5fde2dff2b74a86664dc8f94a9ef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 15:39:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3d3882b516a4167805efbe59944374d : SUCCESS in 1m 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8ac5b132e3b2ac4e71d2671de5751e349312dcaf","revisions":{"0a1afdf58aaf6355ac3056656ec2c4b1a292d869":{"kind":"REWORK","_number":1,"created":"2020-05-29 11:01:42.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/27/731727/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/27/731727/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/27/731727/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/27/731727/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/27/731727/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/27/731727/1"}}},"commit":{"parents":[{"commit":"43f3d5eab2b282d5baa6449f775ed64ee8eb59d6","subject":"Fix outdated get_glance_crednetials in Puppet::Provider::Glance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/43f3d5eab2b282d5baa6449f775ed64ee8eb59d6"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2020-05-21 05:02:40.000000000","tz":330},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2020-05-29 10:58:56.000000000","tz":330},"subject":"Add cache prefetcher interval configuration option","message":"Add cache prefetcher interval configuration option\n\nAs new periodic job added in glance which will run as per\ninterval set using \u0027cache_prefetcher_interval\u0027 configuration\noption and fetch images which are queued for caching in cache\ndirectory.\n\nEnabled \u0027cache_prefetcher_interval\u0027 configuration option\nin glance-api.conf.\n\nConflicts:\n        manifests/api.pp\n        spec/classes/glance_api_spec.rb\n\nChange-Id: I8ece28a9b0c6f104130d817490e9ea4c2b4e7808\n(cherry picked from commit 24ee288b85d376c96746f1f0864bfb0344cf5b19)\n(cherry picked from commit 311cb2ade54812d3bec0c947d1453aad47c22c3a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/0a1afdf58aaf6355ac3056656ec2c4b1a292d869"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/0a1afdf58aaf6355ac3056656ec2c4b1a292d869"}]},"branch":"refs/heads/stable/train"},"8ac5b132e3b2ac4e71d2671de5751e349312dcaf":{"kind":"REWORK","_number":2,"created":"2020-05-29 11:07:30.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/27/731727/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/27/731727/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/27/731727/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/27/731727/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/27/731727/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/27/731727/2"}}},"commit":{"parents":[{"commit":"43f3d5eab2b282d5baa6449f775ed64ee8eb59d6","subject":"Fix outdated get_glance_crednetials in Puppet::Provider::Glance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/43f3d5eab2b282d5baa6449f775ed64ee8eb59d6"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2020-05-21 05:02:40.000000000","tz":330},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2020-05-29 11:07:03.000000000","tz":0},"subject":"Add cache prefetcher interval configuration option","message":"Add cache prefetcher interval configuration option\n\nAs new periodic job added in glance which will run as per\ninterval set using \u0027cache_prefetcher_interval\u0027 configuration\noption and fetch images which are queued for caching in cache\ndirectory.\n\nEnabled \u0027cache_prefetcher_interval\u0027 configuration option\nin glance-api.conf.\n\nConflicts:\n        manifests/api.pp\n        spec/classes/glance_api_spec.rb\n\nChange-Id: I8ece28a9b0c6f104130d817490e9ea4c2b4e7808\n(cherry picked from commit 24ee288b85d376c96746f1f0864bfb0344cf5b19)\n(cherry picked from commit 311cb2ade54812d3bec0c947d1453aad47c22c3a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/8ac5b132e3b2ac4e71d2671de5751e349312dcaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/8ac5b132e3b2ac4e71d2671de5751e349312dcaf"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
