)]}'
{"id":"openstack%2Fpuppet-glance~740787","triplet_id":"openstack%2Fpuppet-glance~master~Ia74bb7c58f4f2c303e0226733b53503c93dea5c7","project":"openstack/puppet-glance","branch":"master","hashtags":[],"change_id":"Ia74bb7c58f4f2c303e0226733b53503c93dea5c7","subject":"Change to relative include","status":"MERGED","created":"2020-07-13 19:19:52.000000000","updated":"2020-07-14 09:40:14.000000000","submitted":"2020-07-14 09:40:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740787-1594719614728-1939c49e","meta_rev_id":"2bae073ce4c59d20618405bb57cd0ee03dbf5c2f","_number":740787,"virtual_id_number":740787,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-14 09:40:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2020-07-14 02:31:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2020-07-14 07:50:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-14 02:31:25.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-07-14 09:40:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e1a01956ff017f48319a7ebf186ef4ada8d2288b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-07-13 19:19:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1cf37234928c37392789e47147c935e521688f23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 21:06:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b1b1448cb0904d059ee257c6e7ed4738 : SUCCESS in 4m 52s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b964a15e60db4e248dd5ec0fa4a39804 : SUCCESS in 5m 01s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/64bc6955627040a99fe10227eef16d19 : SUCCESS in 5m 24s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/789ad899b2d0433ca53a19209d25c652 : SUCCESS in 18m 51s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/d78711915fcd4dd7927e6df2963e5884 : SUCCESS in 14m 32s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/94b50543f5e44cf4bd0dd48ad241610d : SUCCESS in 12m 26s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/97a4415383da4948b6a06df6e7ffc6a3 : SUCCESS in 11m 54s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/23964ff4873c4c6c9fe9436b9545a3d9 : SUCCESS in 49m 18s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e837c077d485479a8f7b3669b1dffca5 : FAILURE in 56m 46s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a6d58611cdb74d1f82fc5e54b42200ca : SUCCESS in 52m 14s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/d0b6e0acf12a454d8dfa22a019929c59 : SUCCESS in 45m 21s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/442da93735dd4cbcbd30ad97a44fde34 : SUCCESS in 1h 23m 38s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7fd3b64ef6c04860996734a1f747f6bf : SUCCESS in 1h 31m 08s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e4c30af029424a2cab89e1137b155f34 : SUCCESS in 1h 29m 21s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5ea37cc0e53c4523b206af43a3befcaa : SUCCESS in 1h 00m 19s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/8b4aa4a4854243b2a34bb8d1f49a0b5d : SUCCESS in 20m 22s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/bd9e6019306941d0b37f9b13d59a7064 : FAILURE in 11m 24s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c47d417688ea4c2a93b94dfe15716e96 : FAILURE in 8m 51s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2490463990654d078235d444e54611ae : SUCCESS in 1h 35m 21s","accounts_in_message":[],"_revision_number":1},{"id":"013b95cafbee3ff02a227476f2544363d7feb48b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-14 02:31:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c5ee8580edf2af4b3f6c87866e2e4a389552e95a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-07-14 07:50:16.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eba77e7696a318f3db0c42d52f51be4d63711f22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 07:50:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c166cf626b6f6682eadd5638331f7e234415ec90","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 09:40:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2bae073ce4c59d20618405bb57cd0ee03dbf5c2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 09:40:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/242b8c672d934fe595c0840d01295c6d : SUCCESS in 6m 56s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6db928daea29432d9449688db4f0ab23 : SUCCESS in 5m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/97a8bea042ae4f37a281a5f09f131106 : SUCCESS in 5m 59s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/a8e8d78fcc744e239f78b40a0be5d88a : SUCCESS in 14m 59s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/7b73b4262fc04bc7a58ddbcabef7cfae : SUCCESS in 17m 29s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ddf18b7e9fe493eacc6efc43c8f2087 : SUCCESS in 14m 33s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8b42f1817f234ab5b9f88cd7bb12b84e : SUCCESS in 1h 21m 53s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4815777cfe3c4f759b73115de77e7347 : SUCCESS in 1h 46m 14s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1cd93596370845f4949a295c5e2ef5ce : SUCCESS in 1h 36m 48s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e757a6dd0d404dccac7da91e3549cd81 : SUCCESS in 1h 01m 51s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/0bb9f5979f6643d6968991dfd9634257 : SUCCESS in 24m 17s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/bc56effb10874b2cab5259b8e7c0f0ba : SUCCESS in 1h 01m 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f919b6012f4a788dc19dba1a1adfa1672615a194","revisions":{"f919b6012f4a788dc19dba1a1adfa1672615a194":{"kind":"REWORK","_number":1,"created":"2020-07-13 19:19:52.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/87/740787/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/87/740787/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/87/740787/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/87/740787/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/87/740787/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/87/740787/1"}}},"commit":{"parents":[{"commit":"0b973b4648dd0f5cc74ae70588a1e37f7636a0af","subject":"Add support for the interface parameter in authtoken middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/0b973b4648dd0f5cc74ae70588a1e37f7636a0af"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-07-13 19:19:42.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-07-13 19:19:42.000000000","tz":120},"subject":"Change to relative include","message":"Change to relative include\n\nChange-Id: Ia74bb7c58f4f2c303e0226733b53503c93dea5c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/f919b6012f4a788dc19dba1a1adfa1672615a194"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/f919b6012f4a788dc19dba1a1adfa1672615a194"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
