)]}'
{"id":"openstack%2Fpuppet-glance~746452","triplet_id":"openstack%2Fpuppet-glance~master~I53f0d0b5cbbb2361d34927800b5b3dcf8a0dc326","project":"openstack/puppet-glance","branch":"master","topic":"db_purge","hashtags":[],"change_id":"I53f0d0b5cbbb2361d34927800b5b3dcf8a0dc326","subject":"Add support for db purge cron jobs","status":"MERGED","created":"2020-08-17 08:15:25.000000000","updated":"2020-08-18 08:34:48.000000000","submitted":"2020-08-18 08:32:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746452-1597739575946-44e8c23f","meta_rev_id":"7a32ba3134d6fd3205b6bd88e063cada6c8d4d0f","_number":746452,"virtual_id_number":746452,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-18 08:32:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-17 15:19:51.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-08-18 07:02:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-08-18 07:02:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-17 15:19:51.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-18 07:02:47.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-08-18 08:32:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c7b419e3d2da580663b2b7afac2148c2f9f9d7c7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-17 08:15:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24495f533808cd8300bf66b38cc900caa3650264","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-17 08:16:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d860965321d559d4de0d2d5aac64850bc95e4665","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-17 08:20:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6cdc3429bc1e0a25a1c845b72970cbec7b1b2c97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 09:58:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5a5a12d984834e5dbea3e0772b63c2cb : SUCCESS in 5m 40s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c6e52ae943a43228e2068a20995eaea : FAILURE in 4m 07s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf7b4b815a594a2d82773962ff6e1aca : FAILURE in 4m 07s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/2710b1143a4f4898947b147d27df639f : FAILURE in 15m 05s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/ebb5a0a73b6040c59e6e47cf9c87dcc2 : FAILURE in 13m 39s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6db3b71137824de899e472f9d6fa8230 : FAILURE in 9m 43s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/06819a4a5de64d54b041e50dae2de448 : FAILURE in 6m 19s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/4dc735dc321b47c6aa86720a49e2e0aa : SUCCESS in 51m 01s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2896e91ea7284c64b58d6043296cb1aa : FAILURE in 45m 10s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/249de465abbd4316a7802718da23ffd4 : SUCCESS in 40m 27s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/aeddf52ae9e2426a878b9070ae7a0e57 : SUCCESS in 37m 54s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e367599958c14e49adc77aef135c2ff5 : SUCCESS in 48m 03s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6b866af99aee4e5984da06ed4dcd0341 : SUCCESS in 1h 31m 33s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3dd72c47e3114cc4b21fbec538b22106 : SUCCESS in 1h 17m 15s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/49b14f5e43234c319bceb10a46786f04 : SUCCESS in 53m 51s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/99865a9602344c29b1141291685a2058 : SUCCESS in 17m 13s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/563c532f818a4c50b36347e4cc865b36 : FAILURE in 8m 36s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3ae344ec32b84e869f77f0c209c59386 : FAILURE in 6m 14s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25bd2ff063874b329943cd7882646f56 : SUCCESS in 3m 11s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ad20407aaec34fa58d051b9c17a4ce43 : SUCCESS in 1h 27m 24s","accounts_in_message":[],"_revision_number":2},{"id":"a41e5883c2984f47dd57a270a6134edc0ccbd5ce","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-17 10:40:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4f368b4a8ef436ba61a2964ec1940ae9049085df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 12:31:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7d4dcf955cae4a0bb9690effbc7ae35b : SUCCESS in 10m 19s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ad4ed475ac24beb81100539ceb77b8d : SUCCESS in 9m 51s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0e08aa6b25fe4f3b81e1da9e80133f59 : SUCCESS in 4m 34s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/eb3cf3b7f4894e7691531dacb26bea03 : SUCCESS in 16m 54s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/465d513e183f4e95876fe3d6984e453a : SUCCESS in 14m 50s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0361f4b2d9fb43ec92c55edcb2622d22 : SUCCESS in 12m 34s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9cc4f78ab3a34346adca6152e7df19ec : FAILURE in 7m 36s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/2d009c0802504600b41dbd8c9823ed91 : SUCCESS in 51m 57s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/df617cb7143c498f9bd38828a7d62a69 : FAILURE in 1h 16m 17s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a1d527d784434b9bb91a7cff26bc5eb6 : SUCCESS in 46m 34s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/fd76f6b4d3d84269a707737eb0cb8111 : SUCCESS in 1h 00m 09s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e49a368086d041b9b9da73d360e9c142 : SUCCESS in 1h 23m 53s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3b5d3800500644e487225b26e91b40a6 : SUCCESS in 1h 46m 53s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fc984e65bb6245d7aa1649f6c8ca14d3 : SUCCESS in 1h 24m 56s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1256a94211994d3797b5bcf769dab415 : SUCCESS in 1h 14m 05s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/d151977cdbfd440e98f6e6d5f5d06f5b : SUCCESS in 21m 34s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/93d784ec3edb46f6bb4c73347653f363 : FAILURE in 11m 06s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c2a0b86b735147e6af7790845c07eed9 : FAILURE in 8m 33s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1fca2a190a69443badaa192cb9f04768 : SUCCESS in 4m 20s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/69b8ba03d5a748f58a423945b5bfaf93 : SUCCESS in 1h 45m 42s","accounts_in_message":[],"_revision_number":3},{"id":"5d527b4117126436afde3421ae904a1f1bb0d241","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-17 15:19:51.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c59e55d35ce4b35fff434a87ecbc015d84836d60","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-18 07:02:47.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d633ecc392d55a0224855d3612ac57ddc98cf164","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 07:02:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3e9bda5db223dd60cf9de19f5cab66ae5a4fce94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 08:32:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"e3ed8fce3eba01799a258bf1206207199f9068fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 08:32:55.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e01ceced53a5429295a41d08401bf661 : SUCCESS in 5m 17s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7d2846f04c744a919141662cf3ff8156 : SUCCESS in 4m 32s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/443a30fd8ad94e8f84ce1d774b1e8369 : SUCCESS in 5m 27s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/03ecabaa36204b1ba12ac6078e59f717 : SUCCESS in 15m 21s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/21c9566330ff42769db2371e08392c18 : SUCCESS in 12m 43s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/67f70f817f904f43a465d8b19769a964 : SUCCESS in 10m 11s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6be04be0e6ca423fbab2a120da798bb3 : SUCCESS in 53m 33s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1102fb26fc214ec7bb93dcb4d2f3894d : SUCCESS in 56m 56s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b46d4f16d15344eeab7b715dfaea437c : SUCCESS in 1h 22m 33s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3d10d27b3847436c8fefd81b00ff9593 : SUCCESS in 54m 15s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/2ddc7728a8444903bb8c22e7553cefc9 : SUCCESS in 19m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8fa79337cf914d44abaf1cd714e9bbdd : SUCCESS in 5m 01s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/aa877c175b6d40fb8b717bba395f0207 : SUCCESS in 1h 20m 47s","accounts_in_message":[],"_revision_number":3},{"id":"4ef4ec0085c2bc000010000d1c7dd246fba7e779","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 08:34:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8e1c0e15f894f7f99a7e0289932d62b : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"43f08226c3917b6aa5244880546040c8dbf66ab3","revisions":{"de91bf0ae17e708512412f1d1b2ce023fe3a681a":{"kind":"REWORK","_number":1,"created":"2020-08-17 08:15:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/746452/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/52/746452/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/52/746452/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/52/746452/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/52/746452/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/52/746452/1"}}},"commit":{"parents":[{"commit":"789065b481dee2c058d1951a201c1680df563539","subject":"Merge \"Prepare Victoria M2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/789065b481dee2c058d1951a201c1680df563539"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-17 08:05:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-17 08:15:18.000000000","tz":540},"subject":"Add support for db purge cron jobs","message":"Add support for db purge cron jobs\n\nThis patch introduces the following 2 classes to manage cron jobs to\npurge deleted records from database.\n - glance::cron::db_purge\n - glance::cron::db_purge_images_table\n\nChange-Id: I53f0d0b5cbbb2361d34927800b5b3dcf8a0dc326\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/de91bf0ae17e708512412f1d1b2ce023fe3a681a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/de91bf0ae17e708512412f1d1b2ce023fe3a681a"}]},"branch":"refs/heads/master"},"cdef78faed828c9de6ab862ca2d7e7bd9f4910e4":{"kind":"REWORK","_number":2,"created":"2020-08-17 08:20:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/746452/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/52/746452/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/52/746452/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/52/746452/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/52/746452/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/52/746452/2"}}},"commit":{"parents":[{"commit":"789065b481dee2c058d1951a201c1680df563539","subject":"Merge \"Prepare Victoria M2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/789065b481dee2c058d1951a201c1680df563539"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-17 08:05:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-17 08:17:47.000000000","tz":540},"subject":"Add support for db purge cron jobs","message":"Add support for db purge cron jobs\n\nThis patch introduces the following 2 classes to manage cron jobs to\npurge deleted records from database.\n - glance::cron::db_purge\n - glance::cron::db_purge_images_table\n\nNote that purging the images table can cause a security risk. Users\nshould read and understand the implications of OSSN-0075[1] before they\nuse the glance::cron::db_purge_images_table class.\n [1] https://wiki.openstack.org/wiki/OSSN/OSSN-0075\n\nChange-Id: I53f0d0b5cbbb2361d34927800b5b3dcf8a0dc326\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/cdef78faed828c9de6ab862ca2d7e7bd9f4910e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/cdef78faed828c9de6ab862ca2d7e7bd9f4910e4"}]},"branch":"refs/heads/master"},"43f08226c3917b6aa5244880546040c8dbf66ab3":{"kind":"REWORK","_number":3,"created":"2020-08-17 10:40:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/746452/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/52/746452/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/52/746452/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/52/746452/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/52/746452/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/52/746452/3"}}},"commit":{"parents":[{"commit":"789065b481dee2c058d1951a201c1680df563539","subject":"Merge \"Prepare Victoria M2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/789065b481dee2c058d1951a201c1680df563539"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-17 08:05:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-17 10:40:48.000000000","tz":540},"subject":"Add support for db purge cron jobs","message":"Add support for db purge cron jobs\n\nThis patch introduces the following 2 classes to manage cron jobs to\npurge deleted records from database.\n - glance::cron::db_purge\n - glance::cron::db_purge_images_table\n\nNote that purging the images table can cause a security risk. Users\nshould read and understand the implications of OSSN-0075[1] before they\nuse the glance::cron::db_purge_images_table class.\n [1] https://wiki.openstack.org/wiki/OSSN/OSSN-0075\n\nChange-Id: I53f0d0b5cbbb2361d34927800b5b3dcf8a0dc326\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/43f08226c3917b6aa5244880546040c8dbf66ab3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/43f08226c3917b6aa5244880546040c8dbf66ab3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
