)]}'
{"id":"openstack%2Fpuppet-glance~748787","triplet_id":"openstack%2Fpuppet-glance~stable%2Fussuri~I0439509589a7b28daf96d5a88ae2f5e01a27cf40","project":"openstack/puppet-glance","branch":"stable/ussuri","topic":"centos8-beaker-stable/ussuri","hashtags":[],"change_id":"I0439509589a7b28daf96d5a88ae2f5e01a27cf40","subject":"Ignore onwer_specifed.* properties in images","status":"MERGED","created":"2020-08-29 04:10:25.000000000","updated":"2020-08-31 15:40:27.000000000","submitted":"2020-08-31 15:40:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748787-1598888427480-abf5dc2b","meta_rev_id":"b4cf380c67d4317d8c20d29175b705c89b374ef2","_number":748787,"virtual_id_number":748787,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-31 15:40:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-31 13:36:57.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-08-29 11:38:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2020-08-31 13:37:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-29 12:57:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-08-31 13:36:57.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2020-08-31 13:37:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-29 11:38:25.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-08-29 12:57:44.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2020-08-31 13:36:57.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-31 13:37:21.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-08-31 15:40:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bc045e361ca01053b7645ab9ad23ea6e3f007f9d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-29 04:10:25.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a032e112cf0cdeafc74fb423bf08698d33698f96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-29 05:48:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b2042f42a7374d2f9b59fadff099eab8 : SUCCESS in 4m 42s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a45c3ce951946a59cd8ea3f47b37207 : SUCCESS in 4m 45s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2bd4a33471724f55ba6cc7d24e3cfcba : SUCCESS in 4m 19s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/610329e4ad464d35ac0cffc384e40ff8 : SUCCESS in 14m 05s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/1973b8ba6d6c41f2ba69f7c69be66ee6 : SUCCESS in 12m 19s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/64c41e65eac74d8f96abd0eeb029fa46 : SUCCESS in 10m 02s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/94582ef8adcf43d983512b8c3af4e9bb : SUCCESS in 53m 13s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/381632d219234b3cb2a90801bd0eb05b : SUCCESS in 48m 53s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e8fa708b26604f1eabf38d0f400a2e2f : SUCCESS in 47m 46s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f5f74ecfe6a47c1af28183174fad6a6 : SUCCESS in 45m 03s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/cd1d9f4e6d7d4590b14d892499af5fe4 : SUCCESS in 42m 59s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/72b8f0260657420596619b65dc6993ea : SUCCESS in 59m 00s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3747740167274846bdde90dc80d90e0e : SUCCESS in 1h 05m 19s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d218d803535b42ab8feaa81532f7a607 : SUCCESS in 1h 05m 52s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8dd5b84f9f1a49f9916b699d51c724bc : SUCCESS in 48m 06s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/6ca6f752793d4057a77205b2d5e286bc : SUCCESS in 18m 31s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/f895d1e9f17b43ab81a3bd396f53201c : FAILURE in 9m 53s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/952b7832606e46fbb65c8cc35e5fcb6c : FAILURE in 5m 53s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/06e90251e9b048edbe7d32036b6b7989 : SUCCESS in 1h 33m 08s","accounts_in_message":[],"_revision_number":1},{"id":"5b043249b9ff5993b8e8a36f16f00a87cf33001e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-29 11:38:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"10ebfd71a5f7b8e52ed02a6224096b44f67e821e","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-08-29 12:57:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a278cece9efa8499cd82dd6be2cd8353a25ad40d","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-31 13:36:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b9cba5523c7c1b0b6026e0352d85d39f989a7f80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 13:37:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1f072797bc8df41e9962447520ba273cc34807b7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-31 13:37:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ab7f1bcf3c8c2e14aaa2041d2659b3dc2b8e97fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 15:40:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ff69e649222d1e8e63f4a1e100506137dfced4bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 15:40:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4f0a2db657d8428889d7df8f1e9a119f : SUCCESS in 5m 11s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca251abedee7487bbf96d83303851e3d : SUCCESS in 5m 07s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/302465947b784cd18e1e9734ed2d9cd7 : SUCCESS in 5m 21s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/01d357f142da4a81a1246b9e320f1166 : SUCCESS in 15m 11s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/1087a224208b462db5d59a84816168c8 : SUCCESS in 17m 23s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/05060a5a751b4d8d962a9fdb99161834 : SUCCESS in 10m 26s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5926b79fbd704ea8a53f7c5bf0519c6a : SUCCESS in 1h 00m 15s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/14106761b0804d82bcd7b20ad1e9fc54 : SUCCESS in 1h 06m 27s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e8b6014d1606499c9299a53038db455c : SUCCESS in 1h 06m 48s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4b5a709009d14d848e65f157b478f9d4 : SUCCESS in 36m 11s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/92c8391632ef4e8ba9124a7758779c6c : FAILURE in 12m 37s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1c2edf73f0ce451abe6be01a44717d16 : SUCCESS in 1h 28m 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1622ed1e703bf605a603a300866102805c188b59","revisions":{"1622ed1e703bf605a603a300866102805c188b59":{"kind":"REWORK","_number":1,"created":"2020-08-29 04:10:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/748787/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/87/748787/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/87/748787/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/87/748787/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/87/748787/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/87/748787/1"}}},"commit":{"parents":[{"commit":"6d91c6f253d613a16eaa5e540416b3094a83fadc","subject":"Add CentOS8 to nodesets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/6d91c6f253d613a16eaa5e540416b3094a83fadc"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-27 02:57:23.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-29 04:10:25.000000000","tz":0},"subject":"Ignore onwer_specifed.* properties in images","message":"Ignore onwer_specifed.* properties in images\n\nSince Ussuri release, openstackclient uses OpenStackSDK instead of\nglanceclient to create an image[1]. This switch made \"openstack image\ncreate\" command to create images with owner_specified.* properties.\n\nThis patch makes glance_image type filter out these properties to avoid\nfalse changes in image properties.\n\n[1] 60e7c51df4cf061ebbb435a959ad63c7d3a296bf\n\nChange-Id: I0439509589a7b28daf96d5a88ae2f5e01a27cf40\n(cherry picked from commit 15d4c5915351fdaa32aec0eb47616eaf824251f7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/1622ed1e703bf605a603a300866102805c188b59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/1622ed1e703bf605a603a300866102805c188b59"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
