)]}'
{"id":"openstack%2Fpuppet-glance~749108","triplet_id":"openstack%2Fpuppet-glance~master~Ib09e32db27e71384f122b54a6d1d269aa1e1a375","project":"openstack/puppet-glance","branch":"master","hashtags":[],"change_id":"Ib09e32db27e71384f122b54a6d1d269aa1e1a375","subject":"Add Puppet Litmus","status":"MERGED","created":"2020-08-31 19:21:00.000000000","updated":"2020-09-10 02:14:55.000000000","submitted":"2020-09-10 02:14:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749108-1599704095505-7e3b2230","meta_rev_id":"24a13bfdcc1feaebce8a6cf9f04f314757ade9dd","_number":749108,"virtual_id_number":749108,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"date":"2020-09-09 22:55:37.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-10 02:14:55.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-09-08 18:27:40.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-09-08 18:18:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-09-08 18:27:40.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-09-09 21:27:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-08 18:27:40.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-09-09 21:27:55.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-09-10 02:14:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c44df7bc63bc8529bbde89209b644a216ad7f8e1","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-31 19:21:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f41de25cb840f14a49950f06bcba75847d26b565","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 21:05:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/614f3912d4224f1f82cf834670dac8c2 : SUCCESS in 4m 34s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc7d91d67c8446fcbe5045ac346442e0 : SUCCESS in 3m 03s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/6d4b6e1c812f495e9925f0f70f77cd61 : SUCCESS in 11m 02s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/604a586a91394efd91173fc4e2a54690 : SUCCESS in 8m 05s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc103cbfc27d445ea0c9fb611bc8cea9 : SUCCESS in 9m 56s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/a25c26122e064122ab699b376172a77a : SUCCESS in 45m 22s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c94c8b38039a4342aa9b871e34a2a37a : FAILURE in 50m 02s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/66efefa35bad414ebd2e2bde8845b732 : SUCCESS in 46m 39s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/d645d4ae3553401face053db63eb0bc2 : SUCCESS in 35m 25s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/55feef7c30974a05a1292dd7b73b67d2 : SUCCESS in 1h 17m 57s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5b2500968b2d4ab4837d468e8c1dfe72 : SUCCESS in 1h 37m 20s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/04077ed63e72433b911107d4eb9b072c : SUCCESS in 1h 28m 24s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d7d130d8fb8146deaa3f1480af400037 : SUCCESS in 1h 00m 11s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/cc4c0aa0bc384597bc4b673bf9762cca : SUCCESS in 19m 44s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1a421a4aea7b4e4eb816ab28789870c9 : SUCCESS in 13m 22s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4d199608be854e409a50c53cf22188f3 : SUCCESS in 1h 25m 45s","accounts_in_message":[],"_revision_number":1},{"id":"04c07846213396b5ddaf5c52efcc23be4c7a0080","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-09-08 18:18:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f47493297e98220376698c06007eedc6742e219c","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-09-08 18:27:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2834bd0f4e870f2b28dbec935814504b44c44028","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-09-09 21:27:55.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d68ca7845e97924770233ac5312ffd7d9890c2ba","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-09 22:55:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b0bc25abc4e61079db6d876f68309e79f14a306f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 00:46:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dbe854aae0cb4669960b9b136c3d1f66 : SUCCESS in 3m 26s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/10718158355b4201aecb34f060ba82fb : SUCCESS in 3m 47s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/234c1b84bf6e4d0b9792e1f21cc614a8 : SUCCESS in 10m 22s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ab6d0c86b48c405599db7f6c64013dfe : SUCCESS in 8m 35s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff12ea45e5e24d619730e89040f2bca1 : SUCCESS in 10m 23s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/41cf43eb4ae84dae9b70027609a0663d : SUCCESS in 49m 59s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/713f4fea97bf41d79bb8a1f097c868e8 : SUCCESS in 53m 22s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b84e91b4628c42608bd0cdac541039e1 : SUCCESS in 42m 34s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/3ae2f77f719d448b91892fe08fe5134f : SUCCESS in 40m 25s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/de4e34ed19c24e6e9322c19071c85eea : SUCCESS in 1h 15m 59s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/aaa3d27cd99c4df6ada8681232213c5a : SUCCESS in 1h 29m 19s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/40343b1ecd7f4c0ba7af57b01c53e71a : SUCCESS in 1h 21m 03s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3174cab902284374b7c0fa89fa9d07bf : SUCCESS in 58m 56s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/b1a397acef7f4c4baf080a9a325c83f7 : SUCCESS in 14m 52s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c00519e22614c07b220efa7085d8ec6 : SUCCESS in 16m 17s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/495fd7839abf46fe9e64cc0b1b5fca33 : SUCCESS in 1h 24m 04s","accounts_in_message":[],"_revision_number":1},{"id":"989e862ffb19db2bc1faa71c958427112911dff6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 00:46:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ecc6056cfdbac74beab9021f4f7236def8192e1d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 02:14:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"24a13bfdcc1feaebce8a6cf9f04f314757ade9dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 02:14:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe129bcced50415b9d70bdd56376e0e6 : SUCCESS in 3m 48s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2de1397a11844b61a6e4eb43bfb3a894 : SUCCESS in 3m 03s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/86a37bd1eea6491e839b2baed910f718 : SUCCESS in 11m 17s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/73727504cc334f6ba285d05d267e26fe : SUCCESS in 8m 30s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/270362cd6e264f6b938995ec51c267b6 : SUCCESS in 1h 15m 20s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6b7a7fc797a14afaaa7472197847fba6 : SUCCESS in 1h 25m 31s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/38257c44e9654621a185decffc8adc64 : SUCCESS in 1h 25m 45s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/76f6f58f31d4435683ed0abb27218b5f : SUCCESS in 38m 32s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/4c90085d2a7943b1981018c219ac2b3b : SUCCESS in 18m 14s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e9e8fc53f66d47f5bf2c8ef17891637c : SUCCESS in 13m 53s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b7492ab7cd6d43d0810a8788bce0a82b : SUCCESS in 1h 17m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"58f3f8adc0e87a40e0a7f979a6d9bafd959e0aa8","revisions":{"58f3f8adc0e87a40e0a7f979a6d9bafd959e0aa8":{"kind":"REWORK","_number":1,"created":"2020-08-31 19:21:00.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/08/749108/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/08/749108/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/08/749108/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/08/749108/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/08/749108/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/08/749108/1"}}},"commit":{"parents":[{"commit":"d96daf92569551701155bc87a79ce314a48b48f2","subject":"Merge \"Do not validate database_connection format\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/d96daf92569551701155bc87a79ce314a48b48f2"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-08-31 19:20:51.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-08-31 19:20:51.000000000","tz":120},"subject":"Add Puppet Litmus","message":"Add Puppet Litmus\n\nDepends-On: https://review.opendev.org/#/c/740601/\nChange-Id: Ib09e32db27e71384f122b54a6d1d269aa1e1a375\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/58f3f8adc0e87a40e0a7f979a6d9bafd959e0aa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/58f3f8adc0e87a40e0a7f979a6d9bafd959e0aa8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
