)]}'
{"id":"openstack%2Fpuppet-glance~757527","triplet_id":"openstack%2Fpuppet-glance~master~Ia0f0fd0c6633e658c20351cf1e1d62a495361552","project":"openstack/puppet-glance","branch":"master","topic":"container_formats","hashtags":[],"change_id":"Ia0f0fd0c6633e658c20351cf1e1d62a495361552","subject":"Simplfy logic to generate list opts","status":"MERGED","created":"2020-10-12 10:21:21.000000000","updated":"2020-10-19 07:33:45.000000000","submitted":"2020-10-19 07:33:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"757527-1603092825000-dfe5b5ea","meta_rev_id":"e58fca9b1c46bfca7811697cafeb338214f18e5f","_number":757527,"virtual_id_number":757527,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-10-19 03:51:36.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-19 07:33:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-10-14 15:23:14.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-10-14 20:41:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2020-10-14 20:41:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-14 15:23:14.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-10-14 20:41:58.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-10-19 07:33:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4fe5b4325569f95358f3d98345f80da9108e995d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-12 10:21:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"13631cf1eaae1561949026956713d4e1cd113042","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 13:12:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f3f298e4a6ee49808ba1d28aeec40262 : SUCCESS in 3m 26s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d7728f85076249169f89a6d4e6146847 : SUCCESS in 3m 01s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/1c187bc2dd4a47d88e85795eabbfe0a2 : SUCCESS in 17m 43s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35a66dc430b54274a61036722b3ff829 : SUCCESS in 8m 51s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0e8e96c48c0148428ba5cb2e71f81d8e : SUCCESS in 10m 12s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/b0074a53eef546179c64adf025e5ae7a : FAILURE in 25m 53s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0061c0db31544b51b1b658a685433e51 : FAILURE in 27m 12s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b9bb41a8cf8c4560af9d6026e5fa904d : FAILURE in 29m 22s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/abc980a5fe0142408a687306e947bfcb : FAILURE in 24m 19s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/881d9c5c30fc4c2dbe626adcd639d306 : SUCCESS in 1h 29m 44s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6ab79c15dabb44969d0129075b637532 : SUCCESS in 1h 33m 43s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/eeba18617de94f0eb14d4b7288a46aba : SUCCESS in 1h 46m 00s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fcc4a587b5b7496996bfecad6b4d9262 : SUCCESS in 55m 02s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/f535047ba27d4efcbdafe038e481af2a : SUCCESS in 18m 18s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8ca8f43df1aa4ac797cd700fee9976e8 : SUCCESS in 13m 47s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/56c31c13df324d76b634625ff7816488 : FAILURE in 2h 42m 05s","accounts_in_message":[],"_revision_number":1},{"id":"14933e1b3ab3631f1dd2bdfb3c3cb7edc66b7f99","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-12 13:30:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"968eb4682182dbdca0982cadb2f9a91cd54d0029","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 15:41:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/135c0676f39c4b408f8b03a4a47fe3c6 : SUCCESS in 3m 35s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c5831a2f562648babd363d383a5564b9 : SUCCESS in 3m 59s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/32de7253c1a549528deb1c8b644146b2 : SUCCESS in 11m 24s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/959d98c9eab84d18ace5316e4fb27dc0 : SUCCESS in 8m 47s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a34fc0d880634109896518d8d31a5634 : SUCCESS in 13m 41s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/90d8d74aed894668bb25834e794ccd80 : FAILURE in 29m 29s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/324ce96d9bda496190cba183705d4b5c : FAILURE in 29m 59s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/327c025508fd4a9da2694845595d03f0 : FAILURE in 25m 10s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/541cf27167fa4928b70f31624604b078 : FAILURE in 20m 52s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/262a28c0e4404b99b1b7c4bd84e85367 : TIMED_OUT in 2h 01m 55s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b5f7bd06d96a4d39a8ccf93ce787b575 : SUCCESS in 1h 03m 10s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c975926965d841a79161b3318608dfe4 : SUCCESS in 1h 26m 19s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c7f5e6a4d7e14091899937266c731cb8 : SUCCESS in 44m 35s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/716dc4c2d65d4eb49363430c31a4ff3e : SUCCESS in 19m 51s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/caa9436dda98498a97795ae0a0665ce5 : SUCCESS in 14m 00s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0ca4afcf7dcb48c8bc080eb271beb53c : SUCCESS in 1h 22m 42s","accounts_in_message":[],"_revision_number":1},{"id":"9452f86097f97bbda7ed6422c402a72e6236d368","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-13 15:11:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"13a1f0d194e0945607a447c01b01f8faf8f023db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 16:47:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dbde5dd829e64275afc14ff5c1aea3a5 : SUCCESS in 3m 03s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a8f4c4ebd74941a0b53ba9835eb93507 : SUCCESS in 4m 00s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/83d1ac0a958344d797fdaf2261fda510 : SUCCESS in 13m 38s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a07d800194d49d199f7aafc1d2e1614 : SUCCESS in 15m 53s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f0ad44ac58f5449ebed08a658d0dcf55 : SUCCESS in 10m 17s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/e9a3c02c651c429caa201402c55030f2 : FAILURE in 37m 30s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2eb1cbeed5864bed8f41b28e1e23d6f6 : FAILURE in 36m 29s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/13d9cf057ce64784af2f931a1bbbddee : FAILURE in 32m 21s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/13716b2687894aa787d5aea5cc1aefe3 : FAILURE in 19m 15s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a84acb57f744437199b49df2e5a3c76d : SUCCESS in 1h 08m 47s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f1e611b5a64146babfb2bc85fc5e763e : SUCCESS in 1h 18m 51s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/915b4d9b7cac4e94873ab234e6e85e84 : SUCCESS in 1h 18m 20s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/75655c26cba64d799e64973d16c223bd : SUCCESS in 53m 43s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/04fe93bf8caa4f27b24562b60b615bd3 : SUCCESS in 18m 27s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f0607aa71417495a9cf548c254b98bea : SUCCESS in 14m 04s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/11de56af848a4ca687f5647a3b2cbeba : SUCCESS in 1h 31m 46s","accounts_in_message":[],"_revision_number":1},{"id":"7967c3772f9bda5a8c0195eabeba7b36f954368e","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-10-14 15:23:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d2756d8022a94799aec12cd1db02b606940ed0e8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-10-14 20:41:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d44472f76ef8d9de60b2d28985d03c379092f64e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 20:42:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"42bbf96ce9b76538e56c566a9b875ac0f4f112fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 22:20:16.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a86eadfac4bf40c4b6836b0355906bcd : SUCCESS in 3m 38s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a37ce3609824d718b3fbd48b3c27637 : SUCCESS in 3m 53s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/8fa97cf7af52456f914cfb4064f17352 : SUCCESS in 11m 46s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e9aa94ca242c495ba9544b8b4511cdb7 : SUCCESS in 11m 47s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/39148b587d3d48eaa8818185ee7ba795 : SUCCESS in 1h 08m 10s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a05e1c0391dc492fbc956a94b93efe48 : POST_FAILURE in 1h 24m 18s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a5c482e63a9f4bb1a1360172b0357fd4 : SUCCESS in 1h 29m 23s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/881a8ecf91ea4ba6b97866caa6963a53 : SUCCESS in 58m 05s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/bd2e54aef64448d9abdf7c00384b8e18 : SUCCESS in 23m 33s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/336b1d767b614f94bad7f806c7186fed : SUCCESS in 17m 16s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3027693ff31542c886c7df2818482a1b : SUCCESS in 1h 31m 49s","accounts_in_message":[],"_revision_number":1},{"id":"2ca8d8520de862add9875837aafa4d125f3663ae","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-18 07:39:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9cecf77bfa62a1ef5d33b38e4fc2016865037909","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-18 09:21:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f24c04a40e6c417cb380e3b4529baa5f : SUCCESS in 4m 49s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b27d220f19d2473789e7e2bb9f4813a6 : SUCCESS in 4m 56s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/0a842781d5d54dcbb6142461109a7c1e : SUCCESS in 11m 29s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8ddd98b886284154b83f08c80abf404b : SUCCESS in 9m 37s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4744678876bd479fb997872795499b26 : SUCCESS in 13m 46s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/a76d67b4df804734ac73ceaf922d3102 : FAILURE in 30m 09s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/76605bc80a7b499684bb760174b020e7 : FAILURE in 30m 59s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ce619337c7443e69fefa875c6537903 : FAILURE in 24m 37s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/cefe81d8458c484a9aabc1328015c45e : FAILURE in 19m 29s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8f2b7266f75a402ab96e0036673f8b1b : SUCCESS in 1h 25m 05s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d84757d791c3469d8c438fa39a768450 : SUCCESS in 1h 39m 10s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/286078c3060544c7aaf2763569796332 : SUCCESS in 1h 14m 28s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f67816cab17445e4aa348cb5f046fae2 : SUCCESS in 50m 20s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/05db270d0ec54af2ba9837a425073b15 : SUCCESS in 19m 21s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fcbf91963fca4445bda245404ec15d6d : SUCCESS in 22m 04s (non-voting)\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/f72ca45af9c64a358d7f9671ede2c09b : SUCCESS in 1h 15m 05s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/de7e102755144caf9262d9289696f223 : SUCCESS in 1h 14m 36s","accounts_in_message":[],"_revision_number":1},{"id":"00775fab8ba8cfd80cbc9c1966bc95b601d2711b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-18 09:21:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e6432ff0323fde28408e43a2bd5a93400adf7ad5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-18 10:56:02.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/32cbb55485254436ac0584644fe965c4 : SUCCESS in 6m 12s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f834fccd2fe442f794d792d29fd3b3f8 : SUCCESS in 6m 07s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/19fda011b1224f6eaa23105db776ed43 : SUCCESS in 13m 14s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9446aa1ffe2d4d67b0b056922148e328 : SUCCESS in 10m 36s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0854f9af7f624832ae8cbc78e373e8e5 : SUCCESS in 1h 19m 45s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6aee571db81e4bbcbcf62ad7205a8b82 : SUCCESS in 1h 33m 27s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fc89eabc986d495ab556b46fe9047133 : FAILURE in 1h 10m 33s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3f955f546cfd41a894ec3fbecefc7ecd : SUCCESS in 1h 02m 05s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/835b5c8d4f4d4aad83e4577d3dc28e33 : SUCCESS in 21m 44s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ee3e4aaf6b2c4d46a118dfc7fb5516f6 : SUCCESS in 18m 00s (non-voting)\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/48d0981b070c48a8bf03c3859816bb15 : SUCCESS in 1h 26m 02s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/be9f0c856255452ab320ffe3e5c6b287 : SUCCESS in 1h 26m 38s","accounts_in_message":[],"_revision_number":1},{"id":"00c1c9944205a746e6a7487c2f0113745e021d56","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-19 03:51:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"16babc257ed6ea92c2954a564ba8de9811ba0e0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 05:32:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b6e038d6c2a64b71ac1a4256b72576ac : SUCCESS in 4m 13s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/818d175936094279aa646e48f7be2034 : SUCCESS in 4m 18s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/3f21e5c49f824a209d43da4c1ac9b4e3 : SUCCESS in 10m 57s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/891cd1d1d236484f9d88515b8c696709 : SUCCESS in 9m 26s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a00d8d0312244cb18bcbcc247fb7d9ba : SUCCESS in 9m 51s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/ea201dd505c04b48a5f2c61c39edbef2 : FAILURE in 25m 48s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d4d2a5a8df434dd1a008fbbaf41908d5 : FAILURE in 28m 47s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d3deda7e724d469db8241fa821b8e57d : FAILURE in 25m 06s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/b847e3dcdb7343138f4c72be8b9e5c8c : FAILURE in 21m 57s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c9949b0faa1b486c9e4b4eb671eb85ee : SUCCESS in 1h 22m 18s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/267e6c9707604d53a61181df33b0033d : SUCCESS in 1h 34m 47s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/75058050771247c69997e7abae30c4e9 : SUCCESS in 1h 31m 49s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e56af73a6b004830ad607d3d887b3692 : SUCCESS in 56m 51s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/99bc43643cbe4ea89365e71aa1965e7b : SUCCESS in 19m 05s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/854ce174a8b34c87881830e6c80d52e1 : SUCCESS in 16m 04s (non-voting)\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/f106d0109e2143ea995d8a5612079369 : SUCCESS in 1h 27m 05s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/03be8704a4ea496b94561b3033c76bee : SUCCESS in 1h 25m 36s","accounts_in_message":[],"_revision_number":1},{"id":"671fb95e6e3105c22870bf8c2df8c55eccf32d89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 05:33:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"da9c31882124e2da53a5510f5af0066f7ecb0f54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 07:33:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/abda32deeed9447a8026d4e6ee4f618b : SUCCESS in 5m 53s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3fed7cc4b2474c6cb0bcac287ad8c4aa : SUCCESS in 5m 29s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/0df4df0d15334f72b9b56709fa302700 : SUCCESS in 12m 12s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce7d821ee99949da831c04c41fdc7557 : SUCCESS in 9m 40s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b995c7984b7b49938b815ccd428eab0f : SUCCESS in 1h 31m 50s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1d7f78cad87d47b48ffeacb023e66462 : SUCCESS in 1h 26m 02s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5c692e133b474e76927960a844683cfb : SUCCESS in 1h 28m 54s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/dff80cb473ad4614afdfe154e0174daf : SUCCESS in 56m 45s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/be975bdd40154e598443f266a38d8390 : SUCCESS in 19m 23s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d8b14bdfa08a429f960af5629bc94b02 : SUCCESS in 16m 59s (non-voting)\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/6871c7a205f44e94baad8f89ff7d0a21 : SUCCESS in 1h 34m 40s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/55e14ef11bcd44b5ab17dac3d8f680c8 : SUCCESS in 1h 33m 59s","accounts_in_message":[],"_revision_number":1},{"id":"04759315fb240d3ce6503bb48a365d49f462d062","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 07:33:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4af7ebd6bc9d96680f2b8eaf239d051c67bb30d9","revisions":{"4af7ebd6bc9d96680f2b8eaf239d051c67bb30d9":{"kind":"REWORK","_number":1,"created":"2020-10-12 10:21:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/757527/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/27/757527/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/27/757527/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/27/757527/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/27/757527/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/27/757527/1"}}},"commit":{"parents":[{"commit":"e49a829eca8d2ff0b5ba587d8b1eeadc7e0aab3e","subject":"Add support for image_format/container_formats","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e49a829eca8d2ff0b5ba587d8b1eeadc7e0aab3e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-12 10:21:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-12 10:21:00.000000000","tz":540},"subject":"Simplfy logic to generate list opts","message":"Simplfy logic to generate list opts\n\nChange-Id: Ia0f0fd0c6633e658c20351cf1e1d62a495361552\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/4af7ebd6bc9d96680f2b8eaf239d051c67bb30d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/4af7ebd6bc9d96680f2b8eaf239d051c67bb30d9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
