)]}'
{"id":"openstack%2Fpuppet-glance~763726","triplet_id":"openstack%2Fpuppet-glance~stable%2Fvictoria~I3fa8be17118960ddf194125760a155af099325d1","project":"openstack/puppet-glance","branch":"stable/victoria","topic":"dnm-stable/victoria","hashtags":[],"change_id":"I3fa8be17118960ddf194125760a155af099325d1","subject":"DNM: Check gate status","status":"ABANDONED","created":"2020-11-23 12:09:11.000000000","updated":"2020-11-24 14:06:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":763776,"cherry_pick_of_patch_set":1,"meta_rev_id":"1fbf63c4bd017af626c06a803648e9de7e475ac4","_number":763726,"virtual_id_number":763726,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-11-23 13:59:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-23 13:59:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"554682584a2f981ee771336ccf0c1fc8fb35bf5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-11-23 12:09:11.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5d170200b28a2b6b8ef81a53cec79f3fdf7ccc1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 13:59:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5b498476fde54a50be114b46f6198197 : SUCCESS in 3m 39s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a4913b78eb6447bfa1000068f604d8c9 : SUCCESS in 3m 23s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/786dc2691d5a48b39041f20adedace30 : SUCCESS in 11m 42s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1b0e840217e346078a036bdc4a447a2e : SUCCESS in 8m 42s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4e4382dfa1614d36be586c802a3f2f2d : SUCCESS in 11m 04s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/043e383684204a72abafb06800f24159 : SUCCESS in 1h 13m 31s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1689d1f2cdb0463ea7aa666afe1d4955 : SUCCESS in 56m 36s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff58c80661084bc2a9cf437f8860a67f : SUCCESS in 44m 56s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c8fa79873d124bb3a4e7bdd3b6a3dede : SUCCESS in 43m 24s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b618ca02321a4a2fb8931e5770f5329e : SUCCESS in 1h 15m 28s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/69a298c3fc7444a289f08be1d4184697 : SUCCESS in 1h 30m 46s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fd00d1b94ffc49db966b394e57dccbe5 : SUCCESS in 1h 00m 20s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8c9b18da418541c5a1c72a4f1e337a41 : SUCCESS in 1h 03m 47s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/6245a6cd0c1449b7aa8e3b93ec960b27 : SUCCESS in 24m 11s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e86dda738194229981a8f6d979b92ec : FAILURE in 26m 06s (non-voting)\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8d373f397f23480b8bca80260de053ba : SUCCESS in 1h 44m 59s","accounts_in_message":[],"_revision_number":1},{"id":"1fbf63c4bd017af626c06a803648e9de7e475ac4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-11-24 14:06:38.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3a7f2362fc1c8681f5b506763b7848156481a503","revisions":{"3a7f2362fc1c8681f5b506763b7848156481a503":{"kind":"REWORK","_number":1,"created":"2020-11-23 12:09:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/763726/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/26/763726/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/26/763726/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/26/763726/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/26/763726/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/26/763726/1"}}},"commit":{"parents":[{"commit":"22535f6787218dd91136389e7f0112fa3d3903f7","subject":"Build containers for single consumer job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/22535f6787218dd91136389e7f0112fa3d3903f7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-11-23 12:07:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-11-23 12:09:11.000000000","tz":0},"subject":"DNM: Check gate status","message":"DNM: Check gate status\n\nChange-Id: I3fa8be17118960ddf194125760a155af099325d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/3a7f2362fc1c8681f5b506763b7848156481a503"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/3a7f2362fc1c8681f5b506763b7848156481a503"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
