)]}'
{"id":"openstack%2Fpuppet-glance~768132","triplet_id":"openstack%2Fpuppet-glance~master~Ie4a8b933200e25c69fc1ac4985acab486d8a62c8","project":"openstack/puppet-glance","branch":"master","topic":"swift_store_config_file","hashtags":[],"change_id":"Ie4a8b933200e25c69fc1ac4985acab486d8a62c8","subject":"Deprecate the swift_store_config_file parameter","status":"MERGED","created":"2020-12-21 23:09:23.000000000","updated":"2020-12-28 04:33:21.000000000","submitted":"2020-12-28 04:31:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"768132-swift_store_config_file","meta_rev_id":"dd3b3914c060bf861c70912ee59ffa76f92b2d09","_number":768132,"virtual_id_number":768132,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-28 04:31:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-12-28 01:54:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-12-27 18:51:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-12-28 01:54:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-22 01:16:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-27 18:51:19.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-12-28 01:54:13.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"}],"messages":[{"id":"ab540f6dd780e0110fdf3ce322c7a2ed22a94415","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-12-21 23:09:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1028a04bab2b9c2ed88cb7b32a76e55cab166da5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-12-21 23:13:08.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b9a86ad85505f7b0e62865cd655064ade4b762dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 01:16:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/74f77116a63442d0a0ab8a769c7ac398 : SUCCESS in 4m 01s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/60369cbc604a402c89d8b7d2f6895d7f : SUCCESS in 3m 38s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/5aba1870d620423dae98e346ac6e0568 : FAILURE in 12m 20s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ebf0e99c0f3c42c7bba855fe8adab156 : FAILURE in 7m 52s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/39de1f11e7994f12b3f0a43fceaa03dc : FAILURE in 10m 27s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3463aa8557b84c628af46e2d9d742de0 : SUCCESS in 51m 30s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a666d3f3c1b41109eb57585838b0ac9 : FAILURE in 28m 59s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5f2c921c15f842a182c31250e2a9a2ff : SUCCESS in 47m 58s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0455ac68fbe9491486746fed43c9830c : FAILURE in 24m 08s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/93274a785188444994c58ac4cb2d4a00 : SUCCESS in 1h 14m 27s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a979a9de91e049798690327a063a18d9 : FAILURE in 47m 33s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ea8e57140023463e811b17f72f8f8801 : SUCCESS in 1h 17m 43s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/34d8731f9b5d4aa48e6bc5b15a90839b : FAILURE in 35m 42s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/60c412b9f63844e29b88294368e66647 : SUCCESS in 17m 34s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e010a94d91e420ba64094f894e95b23 : FAILURE in 14m 55s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/9f1e2d2921b34118a2559316f3986f7c : SUCCESS in 2h 00m 23s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/ca3b48f0231c4278a01d1cd7bd78f33e : SUCCESS in 1h 04m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7f8942feaf64ac5a7a1f532770d93f5 : SUCCESS in 3m 52s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5a6351b0984f4700bcff42d9eeaa9ab2 : SUCCESS in 1h 30m 13s","accounts_in_message":[],"_revision_number":2},{"id":"0e5c06d991340972c143997225d24f265a1d3db1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-12-22 01:39:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"acadc335a95957180a572b96ef12e630e45c2da6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 03:53:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba6ab0885e9b4ef7a2fdb1aeb5030de9 : SUCCESS in 3m 30s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f0d7d03cc4d43be9cd4d25b7111e160 : SUCCESS in 3m 24s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/76d3adb899ef4779aaa744c4b9b16019 : SUCCESS in 11m 48s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0502a1da2b8947ef8b8fc4a0655daa52 : SUCCESS in 9m 45s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/66cf64b7965c4f02b9acd9e89f33a86d : SUCCESS in 8m 57s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d91c43a1f7c4854a7d93325601983e8 : SUCCESS in 56m 08s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/16aff7371fe84d1d9b09074080df12a1 : SUCCESS in 1h 02m 44s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/133ee1ab7ba448fe93b8f0c7404600b7 : SUCCESS in 58m 44s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/12fc66375ea943fcae15741f402de2ff : SUCCESS in 50m 53s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3497b736834e484ca9902077d26f7a94 : SUCCESS in 1h 06m 12s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/676d38ba97e14866bca3202864ec6a24 : SUCCESS in 1h 22m 08s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6285b243041c4f24b7a4512813841d8b : SUCCESS in 1h 09m 40s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8516301d19d14ddc8654242ecc9e5209 : SUCCESS in 1h 00m 46s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/4923884ecaab4407a57eb00ca1081f1e : SUCCESS in 23m 42s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25e66195d4134704ada904df7dbdb9ac : FAILURE in 19m 28s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/092ae0cb09384eeea26fe68e7ee44f9f : SUCCESS in 2h 11m 43s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/7fcfa603b93a4435b58f1bb327afea41 : SUCCESS in 1h 05m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/955855ff86a04e67806503ac1e34943a : SUCCESS in 3m 14s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8ab7fd984f4f4dfc9ced4f58cf67b9f5 : SUCCESS in 1h 45m 52s","accounts_in_message":[],"_revision_number":3},{"id":"21ea2688ab40f6d73f1fbdc214cd05658d3b904e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-12-27 18:51:19.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"edb72add72e0a1759c884ec07e2728d179fb9ef4","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-12-28 01:54:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d99fac4d349514f86dea6f1bc685c4b0a106d068","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-28 01:55:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"72c79977f869d8846d9678d57f393384cc4c86a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-28 04:31:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce9b73a9093a452b83d99c271c4e300f : SUCCESS in 3m 05s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/df9c87d0124748299bf50756b8924666 : SUCCESS in 4m 12s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/ddd3c002a74649689408a23d203f7a7d : SUCCESS in 10m 53s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a397dc1145f445b880a8c421bce3fa40 : SUCCESS in 10m 53s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4f6dbf8ce21b4f419ea7d978e31463f1 : SUCCESS in 1h 27m 56s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6b1e6b673132441fb7899f16a128cad4 : SUCCESS in 1h 02m 19s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/50ecb6defb404646b91318739594d80a : SUCCESS in 1h 28m 58s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c25fdfe38b664aa5a364c89881139772 : SUCCESS in 1h 02m 17s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/95edc5bf88864e8eb477e5b8ba46d2c9 : SUCCESS in 18m 22s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/04ee582d57694f4eab8df0740a6332ee : FAILURE in 12m 34s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/ec2321a89bcf473bae1cecc29ab39b6d : SUCCESS in 2h 18m 31s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/65d7f640bbef47ddae6cc89c3d411f78 : SUCCESS in 1h 07m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a80e6964f71426082176b992cd53ed2 : SUCCESS in 3m 49s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7b309b84c7e4486ebe59794cbb1d9d79 : SUCCESS in 1h 28m 08s","accounts_in_message":[],"_revision_number":3},{"id":"3bc58690efd31ecf9fda5ca5f011890413f71908","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-28 04:31:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"dd3b3914c060bf861c70912ee59ffa76f92b2d09","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-28 04:33:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fffa8ceec21349ef9f5cb537f5ad82f1 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7f1729e9aebc20d892baaf4678221bcc24135853","revisions":{"036f0c461ee0c876fd2a53e4428ad062c3ae0a91":{"kind":"REWORK","_number":1,"created":"2020-12-21 23:09:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/768132/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/32/768132/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/32/768132/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/32/768132/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/32/768132/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/32/768132/1"}}},"commit":{"parents":[{"commit":"26e26d115d04f6208f316932f18a9b57bde4867d","subject":"Use glance-swift.conf by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/26e26d115d04f6208f316932f18a9b57bde4867d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-21 23:06:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-21 23:06:34.000000000","tz":540},"subject":"Deprecate the swift_store_config_file parameter","message":"Deprecate the swift_store_config_file parameter\n\nThe swift_store_config_file parameter is not very useful since we\nalways use the hardcoded path in the glance_swift_config resource,\nwhich manages the configuration file for swift account(s).\nThe parameter is deprecated by this change, and will be removed\nand hardcoded when we remove the interface from puppet-glance.\n\nChange-Id: Ie4a8b933200e25c69fc1ac4985acab486d8a62c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/036f0c461ee0c876fd2a53e4428ad062c3ae0a91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/036f0c461ee0c876fd2a53e4428ad062c3ae0a91"}]},"branch":"refs/heads/master"},"89000f415c94eec8251073f39f9f8582a22bd08d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-12-21 23:13:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/768132/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/32/768132/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/32/768132/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/32/768132/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/32/768132/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/32/768132/2"}}},"commit":{"parents":[{"commit":"26e26d115d04f6208f316932f18a9b57bde4867d","subject":"Use glance-swift.conf by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/26e26d115d04f6208f316932f18a9b57bde4867d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-21 23:06:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-21 23:10:06.000000000","tz":540},"subject":"Deprecate the swift_store_config_file parameter","message":"Deprecate the swift_store_config_file parameter\n\nBecause we hardcoded the path in the glance_swift_config resource, We\ndon\u0027t expect the swift_store_config_file parameter is set to various\nvalues, but set to the hardcode path, \u0027/etc/glance/glance-swift.conf\u0027\nin almost all cases.\nThe parameter is deprecated by this change, and will be removed and\nhardcoded in a future release.\n\nChange-Id: Ie4a8b933200e25c69fc1ac4985acab486d8a62c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/89000f415c94eec8251073f39f9f8582a22bd08d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/89000f415c94eec8251073f39f9f8582a22bd08d"}]},"branch":"refs/heads/master"},"7f1729e9aebc20d892baaf4678221bcc24135853":{"kind":"REWORK","_number":3,"created":"2020-12-22 01:39:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/768132/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/32/768132/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/32/768132/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/32/768132/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/32/768132/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/32/768132/3"}}},"commit":{"parents":[{"commit":"26e26d115d04f6208f316932f18a9b57bde4867d","subject":"Use glance-swift.conf by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/26e26d115d04f6208f316932f18a9b57bde4867d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-21 23:06:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-22 01:39:49.000000000","tz":540},"subject":"Deprecate the swift_store_config_file parameter","message":"Deprecate the swift_store_config_file parameter\n\nBecause we hardcoded the path in the glance_swift_config resource, We\ndon\u0027t expect the swift_store_config_file parameter is set to various\nvalues, but set to the hardcode path, \u0027/etc/glance/glance-swift.conf\u0027\nin almost all cases.\nThe parameter is deprecated by this change, and will be removed and\nhardcoded in a future release.\n\nChange-Id: Ie4a8b933200e25c69fc1ac4985acab486d8a62c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/7f1729e9aebc20d892baaf4678221bcc24135853"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/7f1729e9aebc20d892baaf4678221bcc24135853"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
