)]}'
{"id":"openstack%2Fpuppet-glance~772141","triplet_id":"openstack%2Fpuppet-glance~master~Id54b87472ae67a9005276e52de9dd68172e22eff","project":"openstack/puppet-glance","branch":"master","topic":"keymgr-opts","hashtags":[],"change_id":"Id54b87472ae67a9005276e52de9dd68172e22eff","subject":"Create independent classes for key managers","status":"MERGED","created":"2021-01-23 02:29:41.000000000","updated":"2021-05-12 05:18:33.000000000","submitted":"2021-05-12 05:17:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772141-keymgr-opts","meta_rev_id":"3d4a51fd84c3cdf417dfe7828933fc7abe171c7d","_number":772141,"virtual_id_number":772141,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-12 05:17:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-05-10 09:34:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-05-11 18:51:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-05-11 18:51:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-23 06:48:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-10 09:34:07.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-05-11 18:51:50.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"9f1ae0fdfb83caeb2f29830dd445fc299a11c078","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-23 02:29:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ccce60294e43cda4c20484a8fef52829aa6aa9c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-23 02:31:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8511379e8f80fc3cea87d3ecbf6499eb6c94ed34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-23 03:30:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"79d79875b1eba179ee695bd2c7b6938acb5297ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-23 03:32:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5fc3e70bfda2a1245c89bc2a144986087ffb1796","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-23 03:39:53.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"08630373537312681a1ddda93ae2f1fdb6e7fe7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-23 04:02:18.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c28f1d8af69a27c3bb4553da8be9909070d1bac6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-23 04:04:01.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"110d57deec92bdd0373ff4637ad58a564fcc86b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-23 04:11:43.000000000","message":"Uploaded patch set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"3efe50cb42a1e065708d521ee2a00af95f3bf843","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-23 06:48:50.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95bff0c0df974ff4bce4e44c544810c7 : SUCCESS in 3m 36s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa86bb9b27b441a4b1db2b31f08734f3 : SUCCESS in 3m 36s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/799838c15d71479386989236556a3ed1 : SUCCESS in 8m 03s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c84eb19e86044685a219650d8ae03ab0 : SUCCESS in 8m 45s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aeb058c064434da08be524c21e5b8ea5 : SUCCESS in 6m 44s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf67d26c851943e4b4514d7cac7ad65d : SUCCESS in 48m 29s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/04ec0a512a8b49818bf1b79a6844bf57 : SUCCESS in 1h 07m 36s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c063ef31b9543369ada1e09bf49b4ad : FAILURE in 52m 35s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f3392aaab5b441edbb2a464739f26ffe : SUCCESS in 48m 26s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7e90460255a04459aa44f350c1940005 : FAILURE in 6m 11s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8f2a89a62ef6473b8ae31927fb81a6c8 : FAILURE in 7m 21s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/33566292210f42449a5dba2fc9e1f750 : FAILURE in 5m 43s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/89a50184564442af93cc238dd2f94bb0 : FAILURE in 7m 14s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/55d585e5bf1f4d84b7027d64058c0392 : SUCCESS in 19m 08s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ab39d62fcf44538a3cfe1396eedb894 : FAILURE in 12m 28s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e9c16b64b24b40128f2fd38c8ce87cbf : SUCCESS in 2h 36m 14s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/eafd3ed5087340a096407d5fcd129730 : SUCCESS in 1h 07m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d50ec45c8f249479f137c6b1d7977c0 : SUCCESS in 2m 52s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ceac03047d4b4b6ea4bd61c2c62c79f3 : SUCCESS in 2h 02m 22s","accounts_in_message":[],"_revision_number":8},{"id":"a2b763787d77d7c11c283550cc3602a30d2814a4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-27 10:44:47.000000000","message":"Patch Set 8: Workflow-1\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"ff4f3d118718062ad1477f114468d6e8a81cce13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 13:10:35.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e204e714b0574cd6811c17a17bafa931 : SUCCESS in 3m 21s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/595527a084f54501b1669e2096cdbe64 : SUCCESS in 3m 06s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/bced054b47a64dc2b2b02d25053de436 : SUCCESS in 10m 15s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4228c82df5c648ca8de214ae4e91f75f : SUCCESS in 6m 16s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/272577242970473295776f9a46e5d645 : SUCCESS in 8m 20s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/917a3e429d30411db5d5c0cb547be2e6 : FAILURE in 27m 22s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/27a46e6c5ff8484d86052456f42d01fe : SUCCESS in 53m 44s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/828fc06ed1a94b57ab18250e6342c3a9 : SUCCESS in 49m 31s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fdbc598cfbf14deeb62d97f1eef61e26 : SUCCESS in 40m 54s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9098a885c7814ef1a73cfd1c07062353 : SUCCESS in 1h 35m 49s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6ee70c84ccef4236b306e673318a9087 : SUCCESS in 1h 49m 59s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0f6d21f582ce4c7eab9a08743fe9486c : SUCCESS in 59m 02s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b87d918703ff4cc6af4ef03ca80cfd05 : SUCCESS in 1h 01m 42s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/3fdc059fb9a84c3f81105aa0f3c804b7 : SUCCESS in 20m 04s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8981865bc6cc4d0fb21f1a0e95d7f78c : FAILURE in 14m 35s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/1c2b7f2258b84a1c8ab051991f6ac307 : SUCCESS in 2h 21m 47s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/bf6ac97c5a794683a31ab3e5ddb04fcf : SUCCESS in 1h 12m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b114277da55d4aee930d5c1b60048604 : SUCCESS in 4m 17s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/addffbafab284837aeb3013df76f1a04 : SUCCESS in 1h 47m 48s","accounts_in_message":[],"_revision_number":8},{"id":"38342f196126bc15a6b83a661f78f51d528ff507","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-08 04:43:21.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"525d3aba48166b4bc96c6db0251a73b433a80f38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-08 05:51:52.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3fd7d0a492244b3bb7802b0801701887 : FAILURE in 4m 12s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92a927d68a454cb7aeb4ac0742970bfc : SUCCESS in 3m 26s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4aeb9d2ad5de44b2af6ebaac53c2ff6f : SUCCESS in 3m 20s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/a9e07e3373be46e28e7b8d2b0b997ff0 : FAILURE in 8m 55s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/7dc13d1ed0c949688df96fb9d853fa4a : FAILURE in 7m 42s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/310fb32f757f47ed8c5de552a4312d6e : FAILURE in 6m 35s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/550a86d536e445788db6defc7907b95b : FAILURE in 8m 02s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35bab8668cfd436ca16b1480c1b25d2a : FAILURE in 7m 35s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23bc7e352cea40509725bd9ea046d70c : FAILURE in 10m 58s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f565bdfd46d941689068e02a9a52460d : FAILURE in 7m 31s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37dfe11e5e0f4dcab00b91ec5f605f2f : FAILURE in 7m 06s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fcdf989059e4e4084e97b32e29bc7c3 : FAILURE in 7m 31s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9c88269e7de941d78043e64fb40d3f13 : SUCCESS in 17m 41s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bf6e4fc59bd14713a0ed508b55a3627d : FAILURE in 10m 10s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6dd3697e1bd6445283c4a4a2f47977e8 : FAILURE in 11m 27s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/72d89f03a14742ea8ec284ff13045b55 : FAILURE in 11m 08s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/59f1b04173d7421d9173a3d8f1f10332 : FAILURE in 9m 24s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ef8c52752b75456ba0b186f5402aecc0 : SUCCESS in 21m 56s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/db47b3f2396e4004836715098dfacc05 : FAILURE in 11m 07s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/46b65289840e4dd3bc82608eb587d9a4 : FAILURE in 10m 21s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/013a92cfd55f433483db645d1b9423c8 : FAILURE in 12m 22s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/577cb733a7fc432e8360eb8ef8a912b4 : FAILURE in 11m 35s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6cb40e1fc71e4a16a92a07fb1e4fcdc1 : FAILURE in 11m 06s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/457f89c65a3c4b969236dfa8f8d14c8c : FAILURE in 6m 41s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/41bc7bf3a8c245a0a5fab2804442c91c : FAILURE in 10m 31s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fee4e3e923104f6e8be346b1709e9319 : FAILURE in 6m 34s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4438c4d5e3354fd5a679f21fb008d9fa : FAILURE in 8m 27s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23b969f0e97046509b3eb30da5665e36 : SUCCESS in 3m 46s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ed21a668f2bd4c7c9186e30b1b1f54bc : FAILURE in 1h 06m 33s","accounts_in_message":[],"_revision_number":9},{"id":"e2efbe2b5f1b7cd6805d00f66339231061d768ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-08 06:23:47.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"6eef36fb318a23ce1f9858fe48a49676f3808d6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-08 06:25:00.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"efaaede41488129eef6d3279f072892710eac7c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-08 06:26:41.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"4fb7b6ff8ca5e0da56600eb751c5b97e905cb486","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-08 07:36:39.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5792cbd4e5c846a7907c6ee09d139c90 : SUCCESS in 5m 41s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83e3577390c84ff8a08fe4f3e7535118 : SUCCESS in 5m 44s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3a7d791aa1b411b95521e0c5b839049 : SUCCESS in 4m 37s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/dcaf4166d2a44c3e9bd521436eb9e83a : FAILURE in 13m 02s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/6985a88addf74fa199c21ddcf6658f68 : FAILURE in 13m 54s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/874593fc3e8f4332b34703199379e5e0 : FAILURE in 11m 34s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b38d3f3f47f44322b8cb52d5528e0ad0 : FAILURE in 11m 35s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/63c676c09c41432fb73f4a470111ab5f : FAILURE in 13m 00s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b35df4cd495e41efa55879ef8547db5a : FAILURE in 13m 58s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b8ccf9da37e41648f25a5a99c9a3672 : FAILURE in 13m 40s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd44a2471f0446c7a7534d514bae6139 : FAILURE in 14m 02s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1ee206286004108860159e401238d5a : FAILURE in 13m 12s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4c883cec110f4fe6af81d3cf2d87e6ad : SUCCESS in 25m 39s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3b92dfdf21a0489dbabbd98f2d022a37 : FAILURE in 16m 28s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/801200dc6abb474db447cdb8ffc302cb : FAILURE in 16m 34s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ee156b5b5e814181be45dbd9d703296d : FAILURE in 16m 39s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/01abd1d716dc432c854286b358528cbe : FAILURE in 14m 11s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e772662f08ac45e99e10f089e81b0fce : SUCCESS in 25m 57s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/101a9ca7a691467fb93c187f9f497f22 : FAILURE in 16m 14s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/49df47a643774f06bb8c3ebfbae1223f : FAILURE in 13m 57s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7795445ac27c4f989f373587f82c0dbe : FAILURE in 14m 29s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b2b84a9b3b79474abec2f73b310d6e90 : FAILURE in 14m 49s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/16105a14da8344ad9cb8122df1916ea1 : FAILURE in 12m 24s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d32d163ea8149c6b7b162e58ea6eb25 : FAILURE in 12m 11s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/fe2f7591d4054e2ebe6d8daa2fd27163 : FAILURE in 14m 32s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ba9ccdd0927459a928816c2e6efb3f4 : FAILURE in 12m 07s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f8e83fbe3a6498e99c315bc14964fbd : FAILURE in 13m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17fa03f32a7c4a6893db192fff6668e9 : SUCCESS in 5m 47s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0094b528f0a447a2bf7b6d0115dd645e : FAILURE in 1h 01m 44s","accounts_in_message":[],"_revision_number":12},{"id":"876cc3fa5dbb03a3d78414d7b057a8114aca2eaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-08 07:58:58.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"02cae0c3139a9a44702420b09775df1f9f623b0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-08 09:13:12.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5cf1db6e8544e16a8a35c9fa7c30516 : SUCCESS in 5m 08s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a79274b8c647484ab29003fa4ad98029 : SUCCESS in 3m 22s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70dd58d8421a4422b2298b4aa7aac9a5 : SUCCESS in 3m 24s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/fdd31522e99f40d8b4f9d09d855afb79 : FAILURE in 13m 55s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/40132a86234e4a17a3ad260c2534dec9 : FAILURE in 9m 00s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4539fa5b3f3e4290b80dbf40b6fad42d : FAILURE in 10m 15s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1718db8443ba4bda850cff19cae02fbf : FAILURE in 7m 59s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5cb2a629cf43427384773f46de4cc4bf : FAILURE in 6m 45s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70f5af265aaa4b5aa69e5e3a22378052 : FAILURE in 10m 41s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1e23459dee54cc681a0b22443a95a37 : FAILURE in 8m 11s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c4faadf2d9d44e1a9d3b05babf56fbc : FAILURE in 10m 21s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56dd1c2c34a64a4a8376eb06e441081b : FAILURE in 9m 00s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bb5425d57bb5477ba6ac21c17207fc42 : SUCCESS in 22m 00s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3225a7cdbcad470fbe01ba660a828e26 : FAILURE in 10m 54s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2b619c9ab6114d61ac326d9e1cdcd635 : FAILURE in 14m 39s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/840005ff5e1542cbb2598c8cc4eedef6 : FAILURE in 11m 27s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/abef88509f5349178ff0ef4a4978ed4c : FAILURE in 10m 48s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cd2175bf1c4445a58b72d0f9888bd581 : SUCCESS in 19m 58s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9abfb74fbb984d7da03d57bab2d9cb4f : FAILURE in 15m 28s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9310d5c2e94546bca6832e4b95a25428 : FAILURE in 14m 08s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4e9c4a84452b4fef867f3e04dfc586e9 : FAILURE in 11m 49s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/729338d9526b4e34a9db9f9a6c3f219f : FAILURE in 11m 40s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3280767cae4843208e3cf4a2c94224ad : FAILURE in 12m 05s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e127cce98d0f41d9a684d81b35b6d39d : FAILURE in 7m 37s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/d47bef1216a84e1cb3d55d473fad5380 : FAILURE in 10m 08s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dcac500b541b41f0a46ca846ce677368 : FAILURE in 7m 13s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4caa4434a0204d3d89273cb639e3a1ba : FAILURE in 11m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb85d26bc94d4e7c89bf1ca4aee27721 : SUCCESS in 3m 57s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f6c88c3e90b64970bf6bfc6fc0ae27a4 : FAILURE in 1h 09m 51s","accounts_in_message":[],"_revision_number":13},{"id":"f2b624bc774a792f9538f2614a8751424b39657a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-08 10:35:16.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"b12b4e3351cf8df5b2db7da2647e908ac974680c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-08 12:19:10.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9c769747b304404ab67eb89dcc31caa : SUCCESS in 3m 47s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53ecec0b591146fb86dbbb84112d4210 : SUCCESS in 3m 43s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4133f4c4865445e0ab984fac50174a49 : SUCCESS in 3m 39s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/2ff5b0197b0c48d3920bcd555ccb00de : FAILURE in 12m 56s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/9dbcfba6e8964f498086700d935a1475 : FAILURE in 12m 41s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55beca9157144192b4310834cd1f0d40 : FAILURE in 9m 17s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5398dd73cab64c9aacd55284b921f402 : FAILURE in 9m 04s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d7d24291a704fe99eb88c5bb7dac62c : FAILURE in 9m 51s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ddd304d7a6947b9834f93f4cacfd0f6 : FAILURE in 7m 29s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72c5515e3b014edd8cd2d8388827aa4e : FAILURE in 7m 49s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc518f6c298b4f96b00d8ac27b357906 : FAILURE in 7m 36s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b5b035d7e854189984cfabbbf53ca7d : FAILURE in 7m 26s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f51ebc16066148fb9f4797a9a68a6a2b : SUCCESS in 22m 32s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/05922f1e2bf14919a25d09281ea482cd : FAILURE in 12m 43s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/61d0a7e68f5341948ea9d599808131c2 : FAILURE in 12m 20s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/19b649d3ea874586a2c79b429d3dfd65 : FAILURE in 11m 40s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3521d55f3e49476da9a932330e89df11 : FAILURE in 12m 33s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/eda5379017e745f8bb1d99b6f612ff1e : SUCCESS in 21m 53s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/48743c5107f84a47a6a1eed006ee6055 : FAILURE in 11m 31s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9127dae62c894974af99dbe67b125cdc : FAILURE in 11m 56s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cda4e57e823d45cda5b1c46bee235cdf : FAILURE in 11m 48s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c5b7cd8c96e84a6b8cd4084dc7de887f : FAILURE in 11m 36s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/583e260890b546f3b2b6b7ac902dd9a9 : FAILURE in 9m 28s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc962f861f83441696b33a7a7831afdc : FAILURE in 7m 57s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/44d94b92eece4ff4aba38c5ec7cee114 : FAILURE in 9m 47s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf1f131698a54a0ebe2bbd1893891bd7 : FAILURE in 7m 26s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7106778bccae466d8a89728e07f5e864 : FAILURE in 7m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0fd6ba518560499e82eadf12d7e9a811 : SUCCESS in 3m 47s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0fcfc03a0ecb4314a47143557acd516f : FAILURE in 1h 37m 21s","accounts_in_message":[],"_revision_number":14},{"id":"ba9a507f74049284edbcf31a396737050535d411","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-08 13:02:10.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"c11b6ac1be9eac3005c9c1a45026cfaa25d59529","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-08 14:33:27.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"132835eb4c9fd4c7397e0868bc11c93d9fe03161","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-08 14:35:06.000000000","message":"Patch Set 15: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":15},{"id":"0ae82ef2a98355dc1f469754649d6fe501235eef","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-08 14:35:31.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"59efbbdaa4c0280845e5556f2f1e4f7e265929d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-08 16:33:21.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0930c5205b3a4ab4a7f8e6d9dbeb304e : SUCCESS in 4m 28s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df26d8744a5c41059f2de21db543ce5c : SUCCESS in 3m 57s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b68b036253b64553963e981f04ae0c44 : SUCCESS in 4m 28s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6b0b9bb0ef334b69a017adbddd6f8e97 : SUCCESS in 13m 22s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/26e7fdc29d244c9891834bd42a5da8d3 : SUCCESS in 13m 13s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a92f908ade4439484c5ad4094c2773a : SUCCESS in 10m 05s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f4cce55135d41d6814bd784c7c72eb1 : SUCCESS in 10m 20s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8e0a874ab294f8d901c47f5d76dc852 : SUCCESS in 10m 41s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8bd3578be8f4cefabd9a4c1112f92b5 : SUCCESS in 52m 28s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a93d75270f57400696eb5643821b5936 : SUCCESS in 49m 41s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a11057f3d2a443c6acf82ec6c3526df7 : FAILURE in 25m 23s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41c4d958c1b0406bad6ccc4b6707f31d : SUCCESS in 43m 37s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/350508029cde4503a33cbf22bc90ef53 : SUCCESS in 21m 49s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b4a0455299b143b69bdf11c3b0c18591 : SUCCESS in 59m 08s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1d7381b2feda4b4ea2bd4a638eea0c73 : SUCCESS in 1h 28m 25s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/808c2c55eaf34415b6093184bb2994aa : SUCCESS in 1h 23m 16s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/249f2fd64c5b432daf0dfa4446c00bb6 : SUCCESS in 57m 20s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f4633a36fd334c04a86faf394403df17 : SUCCESS in 24m 19s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2902a24380db4d07ada3621219355838 : SUCCESS in 58m 50s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/dd8a7b4a6a10454fbf356f743ed8f198 : SUCCESS in 1h 35m 20s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e72ef66b8e1042109c6f70de0ba118bd : SUCCESS in 1h 06m 12s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8eb553945ce64296bd02438ebee0cdc1 : SUCCESS in 1h 05m 04s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d0a0f728ad6f4894ba2b31f07b7f4433 : SUCCESS in 19m 51s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39c172cea556487da11cb6b92a8c1675 : SUCCESS in 17m 57s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/32be0c1b9b684284ba2d5b78c3d62090 : SUCCESS in 21m 16s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e32d392e4134b0f88da32681404c6d9 : SUCCESS in 13m 08s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5be2ce7a1ce8439e8a7d8d679934561c : SUCCESS in 22m 00s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5607900b9e954c7187cb2133cffb1d09 : SUCCESS in 6m 01s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/baafdf71b49a4ea79a690ccdcb264dd2 : SUCCESS in 1h 55m 42s","accounts_in_message":[],"_revision_number":15},{"id":"cbe8ac7cdaf6f6c8aceed834b428bf4da016ced5","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-05-10 09:34:07.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"2f659fa1bd7a57c2d72950521a873fd31864a693","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-05-11 18:51:50.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"1403b3d9ea52c51cffad02cdb863bce2d58b0f61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 18:53:29.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"63774bec7f3259f3625f4ed10751e7676ddb2072","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 20:55:55.000000000","message":"Patch Set 15: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16cd8095363e437085b914459cbe71dc : SUCCESS in 4m 41s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05ea573947594f8aa28751e347914f56 : SUCCESS in 4m 26s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c89ceb79bff48e7badf8d5805ddd338 : SUCCESS in 4m 18s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f521798bd29c4766a1118b39a9c4aeef : SUCCESS in 15m 21s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/dcedae7505ef42e4a88568130233d89e : SUCCESS in 13m 32s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0320dbb721fe47058f353845dda82781 : SUCCESS in 11m 27s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35af1b7b474a43db917fc9db321be157 : SUCCESS in 12m 40s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/638fc1f29ea14333939eb5c1ff24abd6 : SUCCESS in 1h 28m 35s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1982b804d727401ba59d3ad9ecaae543 : FAILURE in 1h 43m 27s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/faafad2df52e4b949588ef448e934c27 : SUCCESS in 1h 32m 30s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f93ca7ccb7844887bd478ff5ee8116a2 : SUCCESS in 1h 12m 49s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e504e20cda8d48a2a8ca37bce93dbf8c : SUCCESS in 21m 58s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f17f11bf37a949f299a5b6c3569eac51 : SUCCESS in 18m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82bca4be9ba7461796cf932994a833cd : SUCCESS in 4m 51s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b061b5eeb92445d6a25a2abfcb5d254f : SUCCESS in 1h 58m 26s","accounts_in_message":[],"_revision_number":15},{"id":"e14e3043ea55c14ff971afb08876925be310939e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-12 01:56:59.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"b6ce1b41e4806df70c8ed98bba43b429a3b6e1c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 03:39:27.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6fa817c6188e4357abfc4ce636900235 : SUCCESS in 4m 11s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/629ff74f56e54acb9a64ed53450f49ae : SUCCESS in 4m 06s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd418db3cf3843b7974e3dd3744f1a78 : SUCCESS in 3m 30s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/2856db77c69b4461b9614b7247ebc8f2 : SUCCESS in 11m 38s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/b7cead8541864a4bb95ec5495a7af16c : SUCCESS in 12m 34s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bbec87ba9584473cbd150e9eb3136b54 : SUCCESS in 9m 32s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9155ea36edc040c384a67db0ac81031a : SUCCESS in 9m 13s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/abdaef4fadd544f28e7da40e765db964 : SUCCESS in 11m 18s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4daef0f1c0cd45c291f3d189d99b15be : SUCCESS in 56m 49s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d974ad8bd8c4a52b727df954f9fa6a7 : SUCCESS in 48m 00s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37438ee395154ceb91f3a3e3c7fb757c : FAILURE in 40m 13s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33e154bcbc444756868c9a3c9422957c : SUCCESS in 38m 59s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1d483b5a69cf438d8c590f68ed08e0bf : SUCCESS in 21m 22s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ad3f8acec4b8465f8e5b28336eeb3b43 : SUCCESS in 1h 21m 08s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9a711335808b4122bf032a3a84c3adfb : SUCCESS in 1h 21m 02s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/860aa703a6c643b49868796fc3a3fd7e : SUCCESS in 1h 21m 08s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/03ba9b00d06c4d17a6847665f1f15d2b : SUCCESS in 1h 03m 16s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ad6aafe21e044e30bb467001db5f8bd7 : SUCCESS in 22m 35s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/378b3fc1010f4797bf9bba3d3d442eec : SUCCESS in 1h 16m 26s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c99bc730e2eb4457b6caa567c193ffe7 : SUCCESS in 1h 10m 24s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2bb2ef043eaa41fdb663023c4da92bc8 : SUCCESS in 1h 23m 22s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d95d32cb356c4578ae6203ae2a89edf2 : SUCCESS in 1h 07m 50s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b0fe45f3ddf74a8789353c873c23edec : SUCCESS in 20m 23s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d26534e107fd4c65acc5d5b40d8b83dd : SUCCESS in 17m 22s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/dfc986b7b45d4afb8210a6ad25244355 : SUCCESS in 19m 22s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b6835f5ca3044599b38aeb8545d9498 : SUCCESS in 16m 40s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76193c3e9dfd4c45bc0529897547a561 : SUCCESS in 15m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60b40feee2e248aeb3d1fcccd8b824d3 : SUCCESS in 4m 37s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1c03d719dbaf4fc4aa06648bddab74bf : SUCCESS in 1h 40m 35s","accounts_in_message":[],"_revision_number":15},{"id":"3e076e07fc7004c1fe0affdc8f4b119ec3bdac74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 03:40:38.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"5526ca9b67367dc1342c755cb74c27b6eaa074e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 05:17:19.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80c9f36188fb42b9bdb35e7f92097ccb : SUCCESS in 3m 12s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/332b93dfcc9748e0b01b26f3f46c10b0 : SUCCESS in 3m 15s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac70bdaa6e3442158dbcbca5d7fc2bab : SUCCESS in 3m 18s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3368e21442fe441fbfe8818a0af7fd7e : SUCCESS in 10m 01s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/d884f1aa763b41b48bb6cb38617160a1 : SUCCESS in 9m 47s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/334d79dfab6f4fc58ca50c15fd7ef2fe : SUCCESS in 7m 36s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7168917a6a7c4cef9d14cb7b4f314e19 : SUCCESS in 8m 03s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2815a84cc4534694bb31cee0ae2a5cb6 : SUCCESS in 1h 09m 29s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d1b5ec0456a146ff94e75398c9964495 : SUCCESS in 1h 04m 28s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e391946d9fc746da94058940b45952d6 : SUCCESS in 1h 00m 23s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f12510cae65a488cba69f2766975b8d7 : SUCCESS in 55m 12s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/5a9550fbed6e44c7b23637094fcc8519 : SUCCESS in 21m 14s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4667888a2669487c980f6d6dba0775f3 : SUCCESS in 12m 54s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ca8ac31d1fa45d888d76b8bd795603c : SUCCESS in 3m 52s\n- puppet-glance-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ace254c7122f4bed952f617e7d58ebb0 : SUCCESS in 1h 35m 24s","accounts_in_message":[],"_revision_number":15},{"id":"e074e93c1d585f3b075ffaa543822c11f3c1f98d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 05:17:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":15},{"id":"3d4a51fd84c3cdf417dfe7828933fc7abe171c7d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 05:18:33.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30dcff38019e46f3844949eaccb4fc3a : SUCCESS in 55s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"bc7651fa5f7d622616c8634a55c9339306583c23","revisions":{"0355a13ae9049cab474d042320d82693d24b49a9":{"kind":"REWORK","_number":1,"created":"2021-01-23 02:29:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/772141/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/41/772141/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/1"}}},"commit":{"parents":[{"commit":"9a1acfc99b57ed1f7f2bb2fcf1c9d632f12e6399","subject":"Merge \"cinder backend: password should be secret\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9a1acfc99b57ed1f7f2bb2fcf1c9d632f12e6399"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-23 02:29:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-23 02:29:22.000000000","tz":540},"subject":"Create an independent class for Barbican options","message":"Create an independent class for Barbican options\n\nThis change introduces the new nova::barbican class. This class can\nbe used to maintain options under [barbican] section, which are\nimplemented in the barbican key manager of the castellan library.\n\nChange-Id: Id54b87472ae67a9005276e52de9dd68172e22eff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/0355a13ae9049cab474d042320d82693d24b49a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/0355a13ae9049cab474d042320d82693d24b49a9"}]},"branch":"refs/heads/master"},"45c021c6212d560326c3979672ca02c520aa9c0c":{"kind":"REWORK","_number":2,"created":"2021-01-23 02:31:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/772141/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/41/772141/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/2"}}},"commit":{"parents":[{"commit":"9a1acfc99b57ed1f7f2bb2fcf1c9d632f12e6399","subject":"Merge \"cinder backend: password should be secret\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9a1acfc99b57ed1f7f2bb2fcf1c9d632f12e6399"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-23 02:29:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-23 02:31:05.000000000","tz":540},"subject":"Create an independent class for Barbican options","message":"Create an independent class for Barbican options\n\nThis change introduces the new nova::barbican class. This class can\nbe used to maintain options under [barbican] section, which are\nimplemented in the barbican key manager of the castellan library.\n\nChange-Id: Id54b87472ae67a9005276e52de9dd68172e22eff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/45c021c6212d560326c3979672ca02c520aa9c0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/45c021c6212d560326c3979672ca02c520aa9c0c"}]},"branch":"refs/heads/master"},"cd41e10771f5c8abfc0c965ecabd6bbdc14b23aa":{"kind":"REWORK","_number":3,"created":"2021-01-23 03:30:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/772141/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/41/772141/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/3"}}},"commit":{"parents":[{"commit":"9a1acfc99b57ed1f7f2bb2fcf1c9d632f12e6399","subject":"Merge \"cinder backend: password should be secret\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9a1acfc99b57ed1f7f2bb2fcf1c9d632f12e6399"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-23 02:29:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-23 03:30:20.000000000","tz":540},"subject":"Create an independent class for Barbican options","message":"Create an independent class for Barbican options\n\nThis change introduces the new nova::barbican class. This class can\nbe used to maintain options under [barbican] section, which are\nimplemented in the barbican key manager of the castellan library.\n\nChange-Id: Id54b87472ae67a9005276e52de9dd68172e22eff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/cd41e10771f5c8abfc0c965ecabd6bbdc14b23aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/cd41e10771f5c8abfc0c965ecabd6bbdc14b23aa"}]},"branch":"refs/heads/master"},"1867487fd37d188dc4d2737f5506585b468cc198":{"kind":"REWORK","_number":4,"created":"2021-01-23 03:32:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/772141/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/41/772141/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/4"}}},"commit":{"parents":[{"commit":"9a1acfc99b57ed1f7f2bb2fcf1c9d632f12e6399","subject":"Merge \"cinder backend: password should be secret\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9a1acfc99b57ed1f7f2bb2fcf1c9d632f12e6399"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-23 02:29:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-23 03:32:42.000000000","tz":540},"subject":"Create an independent class for Barbican options","message":"Create an independent class for Barbican options\n\nThis change introduces the new nova::barbican class. This class can\nbe used to maintain options under [barbican] section, which are\nimplemented in the barbican key manager of the castellan library.\n\nChange-Id: Id54b87472ae67a9005276e52de9dd68172e22eff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/1867487fd37d188dc4d2737f5506585b468cc198"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/1867487fd37d188dc4d2737f5506585b468cc198"}]},"branch":"refs/heads/master"},"17e3ae76126760e7893bbdb77e73c5e09355820f":{"kind":"REWORK","_number":5,"created":"2021-01-23 03:39:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/772141/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/41/772141/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/5"}}},"commit":{"parents":[{"commit":"9a1acfc99b57ed1f7f2bb2fcf1c9d632f12e6399","subject":"Merge \"cinder backend: password should be secret\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9a1acfc99b57ed1f7f2bb2fcf1c9d632f12e6399"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-23 02:29:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-23 03:39:45.000000000","tz":540},"subject":"Create an independent class for Barbican options","message":"Create an independent class for Barbican options\n\nThis change introduces the new nova::barbican class. This class can\nbe used to maintain options under [barbican] section, which are\nimplemented in the barbican key manager of the castellan library.\n\nChange-Id: Id54b87472ae67a9005276e52de9dd68172e22eff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/17e3ae76126760e7893bbdb77e73c5e09355820f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/17e3ae76126760e7893bbdb77e73c5e09355820f"}]},"branch":"refs/heads/master"},"61c005d28ba2ef810e1aa03767f2a9be19d67895":{"kind":"REWORK","_number":6,"created":"2021-01-23 04:02:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/772141/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/41/772141/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/6"}}},"commit":{"parents":[{"commit":"9a1acfc99b57ed1f7f2bb2fcf1c9d632f12e6399","subject":"Merge \"cinder backend: password should be secret\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9a1acfc99b57ed1f7f2bb2fcf1c9d632f12e6399"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-23 02:29:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-23 04:01:20.000000000","tz":540},"subject":"Create an independent class for key manager","message":"Create an independent class for key manager\n\nThis change introduces the new independent classes to manage options\nfor key managers, which are implemented in the castellan library.\n\nChange-Id: Id54b87472ae67a9005276e52de9dd68172e22eff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/61c005d28ba2ef810e1aa03767f2a9be19d67895"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/61c005d28ba2ef810e1aa03767f2a9be19d67895"}]},"branch":"refs/heads/master"},"699dbb7faabf9d06060ed7f1257cdc5a221133b2":{"kind":"REWORK","_number":7,"created":"2021-01-23 04:04:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/772141/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/41/772141/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/7"}}},"commit":{"parents":[{"commit":"9a1acfc99b57ed1f7f2bb2fcf1c9d632f12e6399","subject":"Merge \"cinder backend: password should be secret\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9a1acfc99b57ed1f7f2bb2fcf1c9d632f12e6399"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-23 02:29:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-23 04:03:53.000000000","tz":540},"subject":"Create an independent class for key manager","message":"Create an independent class for key manager\n\nThis change introduces the new independent classes to manage options\nfor key managers, which are implemented in the castellan library.\n\nChange-Id: Id54b87472ae67a9005276e52de9dd68172e22eff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/699dbb7faabf9d06060ed7f1257cdc5a221133b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/699dbb7faabf9d06060ed7f1257cdc5a221133b2"}]},"branch":"refs/heads/master"},"3d03dd6dbae837a787e653de8789958bb3e37e1b":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2021-01-23 04:11:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/772141/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/41/772141/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/8"}}},"commit":{"parents":[{"commit":"9a1acfc99b57ed1f7f2bb2fcf1c9d632f12e6399","subject":"Merge \"cinder backend: password should be secret\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9a1acfc99b57ed1f7f2bb2fcf1c9d632f12e6399"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-23 02:29:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-23 04:11:19.000000000","tz":540},"subject":"Create independent classes for key managers","message":"Create independent classes for key managers\n\nThis change introduces the new independent classes to manage options\nfor key managers, which are implemented in the castellan library.\n\nChange-Id: Id54b87472ae67a9005276e52de9dd68172e22eff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/3d03dd6dbae837a787e653de8789958bb3e37e1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/3d03dd6dbae837a787e653de8789958bb3e37e1b"}]},"branch":"refs/heads/master"},"48781cbd7a85817948568f82456c19ae2055b7f2":{"kind":"REWORK","_number":9,"created":"2021-05-08 04:43:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/772141/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/41/772141/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/9"}}},"commit":{"parents":[{"commit":"43ebdd5a9a0b8c5f0b58816523eb94b5caf5eab7","subject":"Bump version for the start of Xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/43ebdd5a9a0b8c5f0b58816523eb94b5caf5eab7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-23 02:29:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-08 04:43:13.000000000","tz":540},"subject":"Create independent classes for key managers","message":"Create independent classes for key managers\n\nThis change introduces the new independent classes to manage options\nfor key managers, which are implemented in the castellan library.\n\nDepends-on: https://review.opendev.org/772676\nChange-Id: Id54b87472ae67a9005276e52de9dd68172e22eff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/48781cbd7a85817948568f82456c19ae2055b7f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/48781cbd7a85817948568f82456c19ae2055b7f2"}]},"branch":"refs/heads/master"},"82b4bb333979055828806d863feee748752e6ddb":{"kind":"REWORK","_number":10,"created":"2021-05-08 06:23:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/772141/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/41/772141/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/10"}}},"commit":{"parents":[{"commit":"43ebdd5a9a0b8c5f0b58816523eb94b5caf5eab7","subject":"Bump version for the start of Xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/43ebdd5a9a0b8c5f0b58816523eb94b5caf5eab7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-23 02:29:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-08 06:23:39.000000000","tz":540},"subject":"Create independent classes for key managers","message":"Create independent classes for key managers\n\nThis change introduces the new independent classes to manage options\nfor key managers, which are implemented in the castellan library.\n\nDepends-on: https://review.opendev.org/772676\nChange-Id: Id54b87472ae67a9005276e52de9dd68172e22eff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/82b4bb333979055828806d863feee748752e6ddb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/82b4bb333979055828806d863feee748752e6ddb"}]},"branch":"refs/heads/master"},"78cc9318aa13a79842ed0345d15465a8d33d0132":{"kind":"REWORK","_number":11,"created":"2021-05-08 06:25:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/772141/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/41/772141/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/11"}}},"commit":{"parents":[{"commit":"43ebdd5a9a0b8c5f0b58816523eb94b5caf5eab7","subject":"Bump version for the start of Xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/43ebdd5a9a0b8c5f0b58816523eb94b5caf5eab7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-23 02:29:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-08 06:24:51.000000000","tz":540},"subject":"Create independent classes for key managers","message":"Create independent classes for key managers\n\nThis change introduces the new independent classes to manage options\nfor key managers, which are implemented in the castellan library.\n\nDepends-on: https://review.opendev.org/772676\nChange-Id: Id54b87472ae67a9005276e52de9dd68172e22eff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/78cc9318aa13a79842ed0345d15465a8d33d0132"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/78cc9318aa13a79842ed0345d15465a8d33d0132"}]},"branch":"refs/heads/master"},"ab3caf4c4787590da4caf4556e7d9002c8b4863e":{"kind":"REWORK","_number":12,"created":"2021-05-08 06:26:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/772141/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/41/772141/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/12"}}},"commit":{"parents":[{"commit":"43ebdd5a9a0b8c5f0b58816523eb94b5caf5eab7","subject":"Bump version for the start of Xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/43ebdd5a9a0b8c5f0b58816523eb94b5caf5eab7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-23 02:29:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-08 06:26:33.000000000","tz":540},"subject":"Create independent classes for key managers","message":"Create independent classes for key managers\n\nThis change introduces the new independent classes to manage options\nfor key managers, which are implemented in the castellan library.\n\nDepends-on: https://review.opendev.org/772676\nChange-Id: Id54b87472ae67a9005276e52de9dd68172e22eff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/ab3caf4c4787590da4caf4556e7d9002c8b4863e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/ab3caf4c4787590da4caf4556e7d9002c8b4863e"}]},"branch":"refs/heads/master"},"d97c6516c340f09b72d6a6b2c8410cc6d6b858df":{"kind":"REWORK","_number":13,"created":"2021-05-08 07:58:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/772141/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/41/772141/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/13"}}},"commit":{"parents":[{"commit":"43ebdd5a9a0b8c5f0b58816523eb94b5caf5eab7","subject":"Bump version for the start of Xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/43ebdd5a9a0b8c5f0b58816523eb94b5caf5eab7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-23 02:29:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-08 07:55:57.000000000","tz":540},"subject":"Create independent classes for key managers","message":"Create independent classes for key managers\n\nThis change introduces the new independent classes to manage options\nfor key managers, which are implemented in the castellan library.\n\nDepends-on: https://review.opendev.org/772676\nChange-Id: Id54b87472ae67a9005276e52de9dd68172e22eff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/d97c6516c340f09b72d6a6b2c8410cc6d6b858df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/d97c6516c340f09b72d6a6b2c8410cc6d6b858df"}]},"branch":"refs/heads/master"},"47d1732b8df69bf4e6dbafcc1b9c002085d19a63":{"kind":"REWORK","_number":14,"created":"2021-05-08 10:35:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/772141/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/41/772141/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/14"}}},"commit":{"parents":[{"commit":"43ebdd5a9a0b8c5f0b58816523eb94b5caf5eab7","subject":"Bump version for the start of Xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/43ebdd5a9a0b8c5f0b58816523eb94b5caf5eab7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-23 02:29:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-08 10:35:07.000000000","tz":540},"subject":"Create independent classes for key managers","message":"Create independent classes for key managers\n\nThis change introduces the new independent classes to manage options\nfor key managers, which are implemented in the castellan library.\n\nDepends-on: https://review.opendev.org/772676\nChange-Id: Id54b87472ae67a9005276e52de9dd68172e22eff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/47d1732b8df69bf4e6dbafcc1b9c002085d19a63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/47d1732b8df69bf4e6dbafcc1b9c002085d19a63"}]},"branch":"refs/heads/master"},"bc7651fa5f7d622616c8634a55c9339306583c23":{"kind":"REWORK","_number":15,"created":"2021-05-08 14:33:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/772141/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/41/772141/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/41/772141/15"}}},"commit":{"parents":[{"commit":"43ebdd5a9a0b8c5f0b58816523eb94b5caf5eab7","subject":"Bump version for the start of Xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/43ebdd5a9a0b8c5f0b58816523eb94b5caf5eab7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-23 02:29:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-08 14:33:19.000000000","tz":540},"subject":"Create independent classes for key managers","message":"Create independent classes for key managers\n\nThis change introduces the new independent classes to manage options\nfor key managers, which are implemented in the castellan library.\n\nDepends-on: https://review.opendev.org/772676\nChange-Id: Id54b87472ae67a9005276e52de9dd68172e22eff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/bc7651fa5f7d622616c8634a55c9339306583c23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/bc7651fa5f7d622616c8634a55c9339306583c23"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
